)]}'
{"id":"openstack%2Fkolla~770090","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I2592a736206eaec811290e9fbdbf2540c0518ffe","project":"openstack/kolla","branch":"stable/train","topic":"bug/1902101-stable/train","hashtags":[],"change_id":"I2592a736206eaec811290e9fbdbf2540c0518ffe","subject":"CI: revert to public package mirrors after build","status":"MERGED","created":"2021-01-11 12:38:12.000000000","updated":"2021-01-26 20:37:56.000000000","submitted":"2021-01-26 20:35:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"770090-bug/1902101-stable/train","cherry_pick_of_change":761928,"cherry_pick_of_patch_set":17,"meta_rev_id":"ec07a83e518a2138268d6d66499f384e44b459f5","_number":770090,"virtual_id_number":770090,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 20:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-01-26 18:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-26 18:35:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-01-26 18:39:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 14:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 08:57:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-13 09:05:19.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-01-19 11:38:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-01-19 11:39:41.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"01ed9229372e07210847decff9068ffe3cc7c97a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 12:38:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7b7d15183acc575dd463bb872f001f071d1bfe06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 14:49:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/33ad32aacf454f96a0dd6a6ef9f01d73 : FAILURE in 40m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b632735f18ee4856b39b60c2e3aca93c : FAILURE in 48m 57s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/df7d40cd6f744c1f82bf59dc2df3e60c : FAILURE in 25m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4a18fa0530fd4657bf000aabf44b47be : FAILURE in 27m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/54f34173f18242e797adc72fbb98bf4f : FAILURE in 35m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ea3e1a9af807458097d09b09ae3d3155 : FAILURE in 18m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/422a7ee1ff6c4765b514280151bd02be : FAILURE in 25m 19s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/99710c154c6d4bc7ac55ebf7395e05f1 : FAILURE in 44m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/dd0eea5182644288a9c89cce3da04247 : FAILURE in 55m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/312f624198a141faa6889a320c3129bd : SUCCESS in 21m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/df3c25fb176d44d98919eeca93fb9227 : SUCCESS in 30m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7915515eb2dc4e06a8cc87c95350490f : SUCCESS in 58m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adf382f34cb84fdfa458ad8664d11026 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/041c6733d9584941814dc7c253c5ccd1 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c3ad1853691424a935df0976db49492 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34726c03b7744fcaae6bf50a299bdb02 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e6260b04a0459aa05d16b6d2190af0 : SUCCESS in 5m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a0a1c171f783444480d619b6ac529571 : SUCCESS in 26m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/92affa5c31294dc4b94cd4cc7c90e0f8 : SUCCESS in 24m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e6ce7963504451da76ed2d4b36cb931 : SUCCESS in 1h 19m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fcaf4d9913124a01bcc5936734325904 : SUCCESS in 1h 06m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f7f4e9c968d744f7bd589225a5dc4aef : SUCCESS in 1h 37m 32s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"724e2a33e3b43104409de224acb3a24c74f3d85e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-12 11:15:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e6eda97f7fe5c42ec3d93c30d0af4bb0f5ed0c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 15:20:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/51fd11610cd744cf857e15b4dbed7c6e : FAILURE in 1h 27m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/09a88f98b30a47c3be9e5856fbe9eac4 : FAILURE in 46m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d193b9115d7141a69ed6eb0319937d73 : SUCCESS in 25m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/78d6ae0ee4be44e285db7013c12c1711 : SUCCESS in 1h 16m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/93990e6cfc5e4ea597720b1368f41261 : SUCCESS in 1h 11m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4eb3f44d421745ce95bbef984171bfc9 : SUCCESS in 54m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/66075a2a1ae941519a912c2516e048b9 : FAILURE in 1h 04m 29s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ce6193727a934975a758709861ce1540 : SUCCESS in 59m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e56fee227b2c425588097a8f2f620353 : SUCCESS in 1h 33m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5c1505a223f54b4ebcb0bec8f4774c6c : SUCCESS in 21m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ca955d2f92894b44a76f7ed1bc8b4dd3 : SUCCESS in 17m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f57f4eb3b7894ae099b9008bea0404e6 : SUCCESS in 54m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b40ddb126c420091acc612e22b40e2 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02163be1332a4336bad84abd6289cdd0 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cca57fa0656b4a44bce57046d6f1878f : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0e332266c104c68838b513308e5b311 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632a5daf90584b24b78a7799c9994938 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e164e659ae0435fab0956f4686b384d : SUCCESS in 27m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f30dc111c5c4b498bd0d19fba0dfdb2 : SUCCESS in 27m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d63215287cd4b9aa9c27bd211a16852 : SUCCESS in 1h 00m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e26fb89ee01c4dd499ae72a261605993 : SUCCESS in 55m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ce575e24d424e98b94f527d89b9af62 : SUCCESS in 1h 06m 29s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"032433c929c4913f786e596f8c3195822ffb6236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 08:56:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ffdeecd33f132d55e186bff073b689221fc44e45","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 08:57:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c33f2338db7fab851f15fe1be86a6f2547800155","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 08:58:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c057139fa68c34526cc05fdea821447125220290","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-13 09:05:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3296117ac8fa6605e7072738114c839ad20754fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 09:10:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e2c1dd63a5e84403b29f97dafae584ff339b7663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 09:12:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a07727b89497719dd883534dbf5c1e1eec680e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:39:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5e74d04aa341487a8b795cf16d77339a : FAILURE in 37m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/875d8305005c4414a06a1f8d14a961a9 : FAILURE in 53m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/51ea8809ae944fe0a338c071182844b8 : SUCCESS in 35m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2de38133c16a46cd863e4d806b309d86 : SUCCESS in 36m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bdbd363532b9432ebf01ab51fa0da4d0 : SUCCESS in 1h 22m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a82c05fe7e4a443aa73b273ea3baafbb : SUCCESS in 58m 11s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e34131dd04ec45189aac4cafb1e194fa : FAILURE in 27m 51s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bbb55a1b3fc04630b407b3a2c375739f : SUCCESS in 1h 09m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6354f3692d6543528dda96d6d51a0cff : TIMED_OUT in 2h 31m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f641b2da980f413bacc0285e0ed899b2 : SUCCESS in 49m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e1d8d16e9a6c469a8c6c99b5b3988096 : SUCCESS in 24m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bdd753f1fa8642e4bd3825ba33dfaf11 : SUCCESS in 21m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8c0d635757c0400db13ab6d76659a6cf : SUCCESS in 54m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603a0afe89894a0b978e5fec508138c9 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/60004c1b4db541e191b42320330d8208 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d300d2dc8bc49849dd50d4782ddfcb3 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4960c432867d47399cb5f3980144bc60 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f5c075da2f430ead3ecdc8bc725088 : SUCCESS in 10m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7eee2b14b584e92891bc2f97ee99bd1 : FAILURE in 35m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/55b718f603ce47daa7a5e57fc51bc81e : FAILURE in 39m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3be4f4e9e1264f708953ef71de44743a : FAILURE in 21m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/82466093c05c45cebf1ddf5415ec5a06 : FAILURE in 18m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eef6eeae92dd4a3b9890bb8e8d591f84 : FAILURE in 22m 28s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"f42ebdd3ed3465c15db9418c0084c8c8e2ed47ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 13:55:32.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"bbf44b43e0df5df0111f6c81fd52e648c576fdcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 14:01:51.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"583bca24a1e5b080d35b31d63d8dc68da385a921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 16:09:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c523434aeb764a50a202e0b8b5b4e523 : SUCCESS in 48m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/76a14df1b80d4c80b6b2b929dfad7308 : SUCCESS in 44m 09s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/daf31cb9cfb74f0f9b7e4e1285024760 : SUCCESS in 26m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1d08ed3f2aef4d11913ca19f18be1648 : SUCCESS in 38m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/15fb2150739b4c7797879612be88e1c5 : SUCCESS in 1h 05m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/40f4e201fa8a4e9ebb027afa17cc1db6 : SUCCESS in 57m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/56f4a6afa6de40b9a69cb5d80994ecc2 : SUCCESS in 1h 08m 11s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/846a090f18e14b52a77dd32910a19f50 : SUCCESS in 55m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c88a6f59d6aa43e586d4aa2a260a80db : SUCCESS in 1h 15m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/17be0502c8cc4dd193495732a3a5aa71 : SUCCESS in 1h 00m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ef370b5003384ae3859930787bc74095 : SUCCESS in 22m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f465440377f64d6eaab8f1a7513b55d5 : SUCCESS in 20m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18e970ea9a6c43c4a07cc8e6a80bf972 : SUCCESS in 56m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff22522f0d74c6fbd17881842eefa70 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa3d744e2f9d4155923581b8118d9744 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37d51e6039374f7f8dbdc656e3086f01 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0899e18256494deaa20264d74feb705c : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804c92b50a48421fb5b1b51501b1d7e0 : SUCCESS in 6m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e54526b905d2459caafeb093338693da : SUCCESS in 36m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/13bc6d360c174c26a79dff638a08742e : SUCCESS in 35m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/29a9e531bb164e5f8102d8c80b35d65c : SUCCESS in 56m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee559d61510a43bdb24b56993ed43250 : SUCCESS in 55m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b7fd889a9ce54072926e6e86929e3459 : SUCCESS in 1h 04m 47s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"befd72e6b5fe050c19545b8cb0b6d56ff08f68b8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-19 11:34:17.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"ab1885fd0b75fe27d43a51fd70a76691e9860787","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-19 11:38:27.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f2ba5699145d5a8d3d1533e0fcbd7d797e3f689e","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-19 11:39:41.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"589a54a517a638800c36c9c9c0ea01b28f84039d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:22:51.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"b7b71676658d2cf63f4104610b0cabe95ac41fb4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:23:09.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"8c37315fb646444c4f378633ef5e6b06e7e6ee78","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:24:13.000000000","message":"Patch Set 8: Code-Review-1 Workflow-1\n\nlol, it has a FIXME","accounts_in_message":[],"_revision_number":8},{"id":"2a506067085d93949b3887c3843853f85a87bfc4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:24:37.000000000","message":"Patch Set 8:\n\nanyone reading this?","accounts_in_message":[],"_revision_number":8},{"id":"1136fd97eecd3987f165d028c97a96ed0543d6af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:24:49.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"5fd38760153d37e0529aca7fa011a23164f91a95","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:25:04.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"0bdb155e6bc092a2ef6b3b4a2d5b10511b150b47","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:29:50.000000000","message":"Patch Set 9: -Code-Review\n\n(1 comment)\n\nhmm, I believe it was WIP","accounts_in_message":[],"_revision_number":9},{"id":"c141b68d0ed9ed2c976932492068c2bd20dd7dbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 11:58:23.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"7bcafa1d54a82a7af920fcc6a72c4d634cf9c83e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 13:32:45.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/07f89eaf7c7f43a1b4b7a6b6d4422eaa : SUCCESS in 34m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4d8eae8c94374526b8784dd84fbfe7b0 : SUCCESS in 48m 27s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ef89d20e702144679085b5731dbfb0d1 : SUCCESS in 27m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f212c867addb48caa54f840bcbced609 : SUCCESS in 26m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/db4550e8133f49a6b4ff4da3069556ca : FAILURE in 6m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fc27df3171834dab9374e4bc1eb1858c : SUCCESS in 52m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/99eefd5f816249558b5027f5f1c4db44 : FAILURE in 5m 30s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bd0c628ccbc84b0f902aa07227c1696b : SUCCESS in 56m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aaaeb37613ca4a3ebb9d55ae025cb642 : FAILURE in 7m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe5f8a0490a74fa1a7097c16f4ead7df : FAILURE in 20m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f3892424e1874292abdb52d2f5d935d2 : SUCCESS in 21m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/886c89a57afe4ee4bb2335b98ac95034 : FAILURE in 5m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700ea47df9b3411c9b67a5f477350209 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35d9905342574ccb9c5a05943a1d47d4 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f80d369dd8be46baae9d0488cdbef33b : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45ac63374bdb4042b72e0e34b8c8f159 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a845580eb6e54f279ee0acb4c97afe85 : SUCCESS in 6m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87cebdfd4e9b49bdb32b7d638d8dd1db : FAILURE in 26m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e96e550a1b4545afab9ab20ea1dcf6a0 : FAILURE in 45m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2627cfb1dc104322ba3c82c4477e5b81 : FAILURE in 17m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e2a351d03373404091faae06ea1c1bd0 : FAILURE in 20m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e82dd1eec6664d8abb50a4d237dc8289 : FAILURE in 18m 57s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":10},{"id":"c70c583b194eac85bc6841a44819767a4fcf0449","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-23 08:52:13.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"e61c4dbebccc30cca9e1e902501c1493e218a630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 12:21:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3370943565be4edba69152267929e938 : TIMED_OUT in 3h 02m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1c7d13ec578f444db11cb1fa4fe0b02f : SUCCESS in 1h 58m 17s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c86adca6a02f49a991afb0b6c8eb347d : SUCCESS in 42m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c7dbe54f7175420595893616e16ed8c2 : SUCCESS in 36m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c086c5ce9a274ddd9bb388d85c07a2f8 : SUCCESS in 1h 08m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/83429258c48e4b6d9aae491e4e0004d6 : SUCCESS in 1h 04m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/64109506718c4912815f97ed05d8896a : SUCCESS in 1h 27m 29s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d6fa6e84865d4cd49b282e000693b1fc : SUCCESS in 52m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/808da0f2b6d2442b926b32e290e3f0f9 : SUCCESS in 1h 41m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2fc5cc339aaa45f6a07cce1c246bf8f6 : SUCCESS in 18m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/45cccbfd52564ceca94ee7fbc8722a8e : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/38f9cf1af78d4caa9cee3149ff9bda35 : SUCCESS in 59m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b07d5496bd524a25a0b341f4992a7703 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f21ff717d88408a8e4b0212f2f8835f : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e60bd8faca4c41fc9e41ecafd8149454 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21533dd454664868b7dede2d47e653c5 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c5b22e4636470cafb98d12e8a125ca : SUCCESS in 5m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05d900e0f41f4913b3d2dac1ed7dd781 : SUCCESS in 28m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/357be9cb7e8b4b0fafba6109ba007077 : SUCCESS in 44m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8b2a42981c64886a7fd79a13ff68551 : SUCCESS in 51m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0253e5f5f94c4982b6167313a8f76035 : SUCCESS in 51m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62dbaa8c0ea2496b9ac61cdab42c7c09 : SUCCESS in 1h 22m 24s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":11},{"id":"412a300409ca73b094217d215034a69c8fcff890","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-25 12:26:39.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"3c28dfe25442fae9f63666275653b0540e3c8492","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-25 14:02:39.000000000","message":"Patch Set 11: Code-Review-1\n\nhey, Mark, was it really finished? as it runs an apt-get update in all the images which looks wrong","accounts_in_message":[],"_revision_number":11},{"id":"be12e57bd987de06dd4c18c755d3cd5964af2f1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-26 10:44:26.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"39cf7e4df72a362f14903dc79e98444e676d204b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 12:00:23.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ddc3e5c828814df1ba80428d2402606c : FAILURE in 41m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2f7ee30ee41d41cba2e8b3d32a2668bb : FAILURE in 28m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ba491610b9744e9ab77c91bfbc6ad359 : SUCCESS in 28m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b312af1f7cc34a018b04f886fcd5c061 : SUCCESS in 28m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9e84a1df2b984c26b6356b610ee195dc : FAILURE in 19m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0621c8d13788427d99ebe7f6d5a1e404 : SUCCESS in 52m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d0ab260580aa4715905c1afa84a87f7c : FAILURE in 29m 40s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/72681271852449729ed2c0261120d446 : SUCCESS in 52m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6780499853e74c9b834f8f484d695ecc : FAILURE in 20m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0cb786c2e6954dd5a146e27984460e5c : SUCCESS in 23m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bc4309f330d64c0e90aff2304b342a7b : SUCCESS in 18m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e08038a8328a4d49b2e138ad598f6cd2 : SUCCESS in 1h 08m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda84f0c73bc4e0785670f02eb0fa9e0 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cdb715591b3b47c1afccd77e7c5a8aaa : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea6aa9ab9be4d4a9ab3b26eba414df1 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4112a8e2d6974b0380cfe10e566bfb7d : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e847340746d842df9ee20b29070bb3c4 : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a7d8665db1245c28d63be79d49eea9b : SUCCESS in 25m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e4b7a9f2e8346c890fd4571e0ebb9d2 : SUCCESS in 1h 12m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d27899edf0d45d08fc79b1175a46765 : SUCCESS in 48m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3a59d9984ee444a9b0b8556d6a83fe94 : SUCCESS in 54m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0c82f0fa60a1476fb4bf05a0dd38b86f : SUCCESS in 1h 00m 58s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":12},{"id":"3f2b1eb8ea54d0c7d1042825a1d09e89cfbbf919","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-26 12:20:01.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"a5e081443c872c42840c53dbea5e8897dc5654c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 13:50:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5e0a1ed27b40466cbba3dbfb4cf3054b : SUCCESS in 43m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e811fc4e32ae4f518f38705fd8a20d84 : SUCCESS in 48m 24s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e731ecc5d7e94802a6ff81e20f4b7c3d : SUCCESS in 28m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/074b7e0806de4c1d994ebf609fccd61e : SUCCESS in 28m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/552e0b3903ed4253a5dc78829faf9e5b : SUCCESS in 1h 20m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1812d94653d24db7abcc801a466da3f7 : SUCCESS in 53m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7402fe7632e0400bb123be6ca7024b7f : SUCCESS in 1h 25m 45s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3cafd2a82f3a4603b09796b916a31c09 : SUCCESS in 1h 01m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/52488d3e2656443bb69b9261a0f67ad9 : SUCCESS in 1h 21m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0ec6f5c5f8e742489d22058f9e73b95f : SUCCESS in 18m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0b98648befc845e8b902833b6269d8e5 : SUCCESS in 35m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9571f1e51a643d6b8e0de5938c95c07 : SUCCESS in 54m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c9fd3cf763492893d9e29c0ced2972 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18d9d0eb66ce4abb97db8a9c1193b635 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9383638fee4455a955dda664864bbb6 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/823b32c8ce44440c8adabe04e0c27b6e : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/125967b2c9fa4470a26a502d9801673c : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a3f0af6d3a8b47829ab1aa3baa0ffb22 : SUCCESS in 37m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef680f14bbe448b6b0da75c4292bfed2 : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/981dff909b6748749598335ff6537a12 : SUCCESS in 1h 12m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/53eb7f2e7cab479fbd6d666de030fb7a : SUCCESS in 1h 10m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0e08e266fa45440e83f4612b8dbf8a90 : SUCCESS in 1h 00m 29s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":12},{"id":"6cb29e4f515d83754dee28e0434548e09d7eff2a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-26 18:35:26.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"1f0808b0e51a0098144aa209bfda32129951386b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-26 18:39:25.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"992f608de4cec304c8319feadc0974a449f8120f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 18:51:35.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"72c28b855b85a333a6cd3584c7d238dec253b2ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:35:39.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a324f8c093d048ca9a7ae2adc911550a : SUCCESS in 47m 54s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8e06bc4c48d34773a52895e46cbca096 : SUCCESS in 31m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a0b1e4aa72c24cbd932793a93e31387e : SUCCESS in 1h 12m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/116b4c94be4e4761b186bac6f5e12b81 : SUCCESS in 57m 29s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/233290c653cb496cae3c2044317d67bd : SUCCESS in 1h 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f469fb9880944b08f4cc1fc6bb31764 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1594266f849a427f8435c0675db801d8 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e131f4bcd2b46d8a38479643d6f78d2 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6226f3167b4f43668611ea25dd4512db : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a9772c178649beb00ee308f455916e : SUCCESS in 5m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4fe299c7ae174b92a30879d6ca284fc3 : SUCCESS in 43m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/99a35053ab274720a459318105c0ef3f : SUCCESS in 1h 32m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/940de593be904bfe82750c288543f0ef : SUCCESS in 1h 26m 50s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":12},{"id":"e506af7cc4cd601ef6083394109b6422ca050649","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:35:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"ec07a83e518a2138268d6d66499f384e44b459f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:37:56.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0f5b34ff9e4849bb4a9aab23c69e68 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"c207f92c90108a1375b9a23e6b7a53bc959ffe34","revisions":{"ac9190bee2cf6661b69ce4cfdc1c3cd75aec8da1":{"kind":"REWORK","_number":1,"created":"2021-01-11 12:38:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/770090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/1"}}},"commit":{"parents":[{"commit":"128e4d32258d29a60820fcfb85a00bc91851e367","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/128e4d32258d29a60820fcfb85a00bc91851e367"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-11 12:38:12.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac9190bee2cf6661b69ce4cfdc1c3cd75aec8da1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac9190bee2cf6661b69ce4cfdc1c3cd75aec8da1"}]},"branch":"refs/heads/stable/train"},"60b60a66b22ed026c3a5849517b68aa030a8a05a":{"kind":"REWORK","_number":2,"created":"2021-01-12 11:15:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/770090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/2"}}},"commit":{"parents":[{"commit":"128e4d32258d29a60820fcfb85a00bc91851e367","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/128e4d32258d29a60820fcfb85a00bc91851e367"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-12 11:15:15.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60b60a66b22ed026c3a5849517b68aa030a8a05a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60b60a66b22ed026c3a5849517b68aa030a8a05a"}]},"branch":"refs/heads/stable/train"},"fc914f25295d25b61ea005eca71cf7afd9e85fca":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-13 08:56:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/770090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/3"}}},"commit":{"parents":[{"commit":"ea55ccbf1622822e78197f77a06a36b62bbe9dba","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea55ccbf1622822e78197f77a06a36b62bbe9dba"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 08:56:31.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc914f25295d25b61ea005eca71cf7afd9e85fca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc914f25295d25b61ea005eca71cf7afd9e85fca"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"84451e0d61157c2aa5e43aa29dd639e61a3b6782":{"kind":"REWORK","_number":4,"created":"2021-01-13 09:10:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/770090/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/4"}}},"commit":{"parents":[{"commit":"177cb550626f633b483641dfa82f6703c9f04772","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/177cb550626f633b483641dfa82f6703c9f04772"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 09:09:55.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84451e0d61157c2aa5e43aa29dd639e61a3b6782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84451e0d61157c2aa5e43aa29dd639e61a3b6782"}]},"branch":"refs/heads/stable/train"},"2c5798b2a1808f9473d70c96214aaa2afa8d2dda":{"kind":"REWORK","_number":5,"created":"2021-01-13 09:12:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/770090/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/5"}}},"commit":{"parents":[{"commit":"177cb550626f633b483641dfa82f6703c9f04772","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/177cb550626f633b483641dfa82f6703c9f04772"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 09:11:56.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c5798b2a1808f9473d70c96214aaa2afa8d2dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c5798b2a1808f9473d70c96214aaa2afa8d2dda"}]},"branch":"refs/heads/stable/train"},"02fd8e95e1f55368d999e44d6ba97fce5f643552":{"kind":"REWORK","_number":6,"created":"2021-01-13 13:55:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/770090/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/6"}}},"commit":{"parents":[{"commit":"177cb550626f633b483641dfa82f6703c9f04772","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/177cb550626f633b483641dfa82f6703c9f04772"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 13:55:24.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02fd8e95e1f55368d999e44d6ba97fce5f643552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02fd8e95e1f55368d999e44d6ba97fce5f643552"}]},"branch":"refs/heads/stable/train"},"9b0c3de139258b09312e3a10bc0b7a1cbbdd3403":{"kind":"REWORK","_number":7,"created":"2021-01-13 14:01:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/770090/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/7"}}},"commit":{"parents":[{"commit":"177cb550626f633b483641dfa82f6703c9f04772","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/177cb550626f633b483641dfa82f6703c9f04772"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 14:01:42.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b0c3de139258b09312e3a10bc0b7a1cbbdd3403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b0c3de139258b09312e3a10bc0b7a1cbbdd3403"}]},"branch":"refs/heads/stable/train"},"719845cba091b3e581d86770aa208355c82da973":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-01-20 10:22:51.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/90/770090/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/8"}}},"commit":{"parents":[{"commit":"02860bf9eb7bde215606d7c7b54d0eaaf64cfa61","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02860bf9eb7bde215606d7c7b54d0eaaf64cfa61"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-20 10:22:51.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/719845cba091b3e581d86770aa208355c82da973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/719845cba091b3e581d86770aa208355c82da973"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"d34ad6e6526fd56095a2ee5d9d95d0af663e6655":{"kind":"REWORK","_number":9,"created":"2021-01-20 10:24:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/90/770090/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/9"}}},"commit":{"parents":[{"commit":"02860bf9eb7bde215606d7c7b54d0eaaf64cfa61","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02860bf9eb7bde215606d7c7b54d0eaaf64cfa61"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-20 10:24:46.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d34ad6e6526fd56095a2ee5d9d95d0af663e6655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d34ad6e6526fd56095a2ee5d9d95d0af663e6655"}]},"branch":"refs/heads/stable/train"},"1a4851d5828a7c0f33a0b918ca3fc7cb189e1e06":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-01-20 11:58:23.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/90/770090/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/10"}}},"commit":{"parents":[{"commit":"4e6b60ea8ddaeb9cf82f491bc4254a57986b84d4","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e6b60ea8ddaeb9cf82f491bc4254a57986b84d4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-20 11:58:23.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a4851d5828a7c0f33a0b918ca3fc7cb189e1e06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a4851d5828a7c0f33a0b918ca3fc7cb189e1e06"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"dd8d1827d6f0ec28edd68b52c21ebf78d3d44c74":{"kind":"NO_CHANGE","_number":11,"created":"2021-01-23 08:52:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/90/770090/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/11"}}},"commit":{"parents":[{"commit":"297c15de4b92425286cba2eb3977ec5d877a7e9a","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/297c15de4b92425286cba2eb3977ec5d877a7e9a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-23 08:52:13.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd8d1827d6f0ec28edd68b52c21ebf78d3d44c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd8d1827d6f0ec28edd68b52c21ebf78d3d44c74"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"c207f92c90108a1375b9a23e6b7a53bc959ffe34":{"kind":"REWORK","_number":12,"created":"2021-01-26 10:44:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/770090/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/770090/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/770090/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/770090/12"}}},"commit":{"parents":[{"commit":"297c15de4b92425286cba2eb3977ec5d877a7e9a","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/297c15de4b92425286cba2eb3977ec5d877a7e9a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-26 10:44:22.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c207f92c90108a1375b9a23e6b7a53bc959ffe34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c207f92c90108a1375b9a23e6b7a53bc959ffe34"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
