)]}'
{"id":"openstack%2Fkolla~770091","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I47f8611ab74b6f33f74907b5afcacdefd65a4adb","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"I47f8611ab74b6f33f74907b5afcacdefd65a4adb","subject":"DNM: testing mirror revert","status":"ABANDONED","created":"2021-01-11 12:42:07.000000000","updated":"2021-01-12 12:11:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":763351,"cherry_pick_of_patch_set":11,"meta_rev_id":"b40c4c3204ece995d1fbe6751ecec4e3b93eb126","_number":770091,"virtual_id_number":770091,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-12 11:05:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 14:42:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a7cedc9b8ceb594c9423cf33511ab93f3f79b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 12:42:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"35f883114b14820936dcffe08e5c93a527ffe01e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 14:42:36.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ae7c943b3c1246e1b1b7ab919ae5029f : FAILURE in 53m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b9aa79e748034546adcdea0e20e94433 : FAILURE in 49m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ba1304a480ee4d798fa44f20275a3288 : FAILURE in 17m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a652db54a4084708b43b9b4980818784 : FAILURE in 22m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/28037978013640799490986747552c06 : FAILURE in 17m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/71b676a27ae54e259c260ae27951bd82 : SUCCESS in 21m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6c75b58cc576491ca1542756f28596f7 : SUCCESS in 22m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/56020499422540559276c520c9e5e22e : SUCCESS in 1h 28m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a67f87b1226492db73faf9e6ef7606f : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17f52c12f34e4d949f0e656dda1568f2 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f532b071e84b465aae86182b0c391901 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83017f46f3d47d08a5fd9b900fe9337 : SUCCESS in 6m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d24c4688bafe4cb2bbb5fd41bd55e80d : SUCCESS in 19m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/720c070b292d4bd78ddecd034aa51bee : SUCCESS in 49m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f7332eec239454f8335f16ddcc38367 : SUCCESS in 1h 18m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a52f8597590d488eb25710e4a2fcecc4 : SUCCESS in 56m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a6266b873bd44f5c8676e4ca699a61ea : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bb3b991fed1240258e819849a602a437f97c62cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:20:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/45ef5a151d604f8b9397dfa40eceec7b : FAILURE in 2h 48m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aaeef7efd3284449a2c949763e610a3f : FAILURE in 2h 50m 56s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/45878b089d1548f48c745d167d5f920a : TIMED_OUT in 3h 04m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/60f88e51f6a447af95d12a201ccfc587 : FAILURE in 2h 55m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d83135d7279fe086a76c4a3ccbb20ef182bd352e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-12 09:33:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d3728fdd54316b50dd583219e35f22dc6ac4321a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:23:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3525c86421d34ed1be5d3206f8b0006c : RETRY_LIMIT in 3m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e275cb8db42c40a3a24a79da14d38752 : RETRY_LIMIT in 2m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6a0ed9bc731490b93153cfd73e21c2d : RETRY_LIMIT in 2m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3eb098088857406dac3835e33e6ac155 : RETRY_LIMIT in 2m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2143c9d9759b7f72140be5ce3549e0d8c57b3dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 11:05:52.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/03b75d391d694659818e4a34625c30ab : SUCCESS in 31m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/69567b370491430885f2b07402ce3a83 : SUCCESS in 29m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4f2965c0a84449e9ae9c50163d967fa6 : SUCCESS in 59m 35s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/852ef3b5bfd347eaae978c097f575afd : SUCCESS in 1h 21m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9b35889fb6c143c29146f46c72aa80f9 : SUCCESS in 1h 07m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/feb7e98efd714ab2af9b9f6604ec3721 : SUCCESS in 18m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e1d5f24e392b46a09323d587711490e9 : SUCCESS in 17m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ab1561979634c7e80459914dc9c6039 : SUCCESS in 55m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2e8f5fb0d2f4b01a36aca7ec030d108 : FAILURE in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4c62871a9ea4c6f9b66bbe94bab2f1c : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cd296dd6bdf4e4cb066cf359f77350f : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2942498a871b442f998019bfd4c205c3 : SUCCESS in 5m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/98816575bdc64eb1bbff3aff874d2572 : SUCCESS in 19m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/78b04cca61994522a58cd194032de188 : SUCCESS in 22m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6efd836eaa8e49f2916417727147f8d2 : SUCCESS in 56m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/998b040c1e9647b594f01c648b21745f : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c19a1af8adf347169c267a0147fe5cfb : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b40c4c3204ece995d1fbe6751ecec4e3b93eb126","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-12 12:11:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d34bbbd666d3ca19629b0c4f08d86d6e0463c95c","revisions":{"80a6f1c9f87ffa9b75701ff5938ba4fccbe0fe61":{"kind":"REWORK","_number":1,"created":"2021-01-11 12:42:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/91/770091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/770091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/770091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/770091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/770091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/770091/1"}}},"commit":{"parents":[{"commit":"adcf72abe8f0eeac8b69e28dbb78fd24f944116f","subject":"CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adcf72abe8f0eeac8b69e28dbb78fd24f944116f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-11 12:42:07.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80a6f1c9f87ffa9b75701ff5938ba4fccbe0fe61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80a6f1c9f87ffa9b75701ff5938ba4fccbe0fe61"}]},"branch":"refs/heads/stable/victoria"},"d34bbbd666d3ca19629b0c4f08d86d6e0463c95c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-12 09:33:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/91/770091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/770091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/770091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/770091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/770091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/770091/2"}}},"commit":{"parents":[{"commit":"556b9382a5f8b2e72c65dd3176877bd53ce3f7d7","subject":"CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/556b9382a5f8b2e72c65dd3176877bd53ce3f7d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-12 09:33:44.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d34bbbd666d3ca19629b0c4f08d86d6e0463c95c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d34bbbd666d3ca19629b0c4f08d86d6e0463c95c"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
