)]}'
{"id":"openstack%2Fkolla~770093","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I47f8611ab74b6f33f74907b5afcacdefd65a4adb","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I47f8611ab74b6f33f74907b5afcacdefd65a4adb","subject":"DNM: testing mirror revert","status":"ABANDONED","created":"2021-01-11 12:43:47.000000000","updated":"2021-02-03 12:07:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":763351,"cherry_pick_of_patch_set":11,"meta_rev_id":"31d60a43482b7f61abbf144c7d15de4ffd789597","_number":770093,"virtual_id_number":770093,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-18 14:22:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 15:19:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f04589e789acd7ce30a0d82b87b8fda4063a028","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 12:43:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8d91ea7b6c0bba446fbce76dab36da6fe680fac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 15:19:26.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a8ef9ff73b4e4b9abd55a9eb24a0cdd2 : FAILURE in 38m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c55322a732f44cb4a0b3c863eb957f85 : FAILURE in 47m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/92858b439f564654a90b2fb712a763a9 : FAILURE in 36m 30s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/64182407b760456eacf0f9819259d3e6 : FAILURE in 28m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8a9c8ba8255940b195b291730e4fecaa : FAILURE in 32m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7d9218e832ea45ffb7ba334c4cd40276 : FAILURE in 1h 55m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/27f8dfacdf5b41f5890e3053e2e881c5 : FAILURE in 29m 32s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d827919a0aeb4c868a86c93299ebea2c : FAILURE in 18m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8d506b4fa8a24b388b002bc19c614dca : FAILURE in 31m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/aebb2f16c5b24b9f980e8343230aed00 : SUCCESS in 53m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4eb3c9e46a3e40a5985bea9b1d011dab : FAILURE in 35m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0a42e24f5aff443e81b01216d48bc871 : FAILURE in 32m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/25171bd8c2b14ac8aa5b38c3e33445f7 : FAILURE in 31m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b74ae27628e4ce3aa7df3d5032dea8a : FAILURE in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ae11f49fa934eec81e32645bf2883fc : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2611792ee35b4daa99716e689e621359 : SUCCESS in 7m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df848f8942eb4038aed52a9d7e95ca0a : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8982a8ccd39f4f55ad4028904e0f2fe3 : SUCCESS in 6m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/93afa1f706b540b99a3565f29950e4aa : FAILURE in 27m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6390d3ac5e694acd895ebe113e99277c : FAILURE in 25m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a8c4beb026b4c388bcc33e767703c56 : FAILURE in 28m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ba825f09871840d6a4db9137ed2839b4 : FAILURE in 28m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/749324fee25c42819df64457076c3015 : SUCCESS in 1h 13m 39s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"69cc1a5cf25e6845b82692d732da08e9cdc117da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-12 11:15:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b53c36170cc0e918442727486a83dc4bbf52a25d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 16:08:37.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/931ae0afe8ae4462bf4c0a13f836f481 : FAILURE in 1h 25m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/24c838ed77574c219f48034245ddbc33 : FAILURE in 1h 09m 20s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a733df0447cc435fa76a79bc38cbe614 : SUCCESS in 28m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ed21256b64a84db2adec401ee975b564 : SUCCESS in 33m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/94e74eb8f11d4b88ac314052e02af0c4 : FAILURE in 34m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e0267c9f29e647d8818cfac8564106fa : SUCCESS in 55m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/30e52153713346839fbd071139673643 : FAILURE in 58m 19s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2b7ef935e5524327854e54ad0508fcdb : SUCCESS in 1h 16m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8421ee03a6d94345aba080d676ff7070 : FAILURE in 27m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/70be697eaef948079669017b904279d1 : SUCCESS in 39m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bcc15125c642423fa61ef949ab6c4313 : FAILURE in 23m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b1a7db6ddf51455495b3011a26e14ffc : FAILURE in 19m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ffa449edd78f4754b1f062ea014f1c1a : FAILURE in 34m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74d7a4bad1b4b749d1aecb35c42c814 : FAILURE in 9m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a83f337ac10a458cb3a73b497cc59b03 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebd81c76758f429294e380e262e4accb : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/481d18433bc641f8b9c8a56a0db568d4 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc15f6dd81b34154bf2911d937940e3c : SUCCESS in 23m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e57c6a6fb7ba494c9b0da4f39a7dd63f : FAILURE in 48m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc66686229b24fafafb66d6cdc3368fb : FAILURE in 30m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb09933df6914ab88585cae7f6be6d4a : FAILURE in 22m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c2c6b0a68454977bacecf157938adf1 : FAILURE in 21m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f6a561142261467da1e22f75ad6ca907 : FAILURE in 16m 34s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"a641ed5afe692965293574c5a942a6d83a31e22f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 16:14:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"eee9bee97b5ee160b4488263e0a315b60a7319c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:02:05.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f4b9676da559492597e6c3bc3793fc48 : FAILURE in 46m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cde8511ae04b42e2b56bf3a3f0e63e0c : FAILURE in 47m 22s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/afa93c02e9374afc951762fe5ab4a16f : SUCCESS in 27m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f64e5b208f6944b4a1c8f642ddea0f1e : SUCCESS in 34m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0af367cfcd474e51bf7b94ba5f967138 : FAILURE in 32m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9641112d4ff64ccbad2361227bc46622 : SUCCESS in 57m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7ef9fa7ec002467e800dadaaaf02aab8 : FAILURE in 58m 46s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7a2d223fe30941f588647f4f77d5793a : SUCCESS in 1h 31m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/26ab5e7cf14243278d7aa7fa33d3e89b : FAILURE in 1h 03m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/92356d00987b431797a8c873b27e63b0 : SUCCESS in 56m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f8a0144133cb4ff1ac5fe5e0b416bbfd : FAILURE in 21m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/612fd3ebdc994ba1ae5ea5177a036520 : FAILURE in 18m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36e5559477aa4ed99471445e04a27acd : FAILURE in 36m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd38956384449b7ad4f34538376c249 : FAILURE in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd393889c64b4d3b81ccf1af643a3439 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dc686411bae4b278fa836a85af4076a : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09f9a1b4e39d44519761f02f8c06fbff : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/490429b35d004b979ba31f1f84d3ec76 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8725f85dc2d4f57b6f37d75722ccbea : FAILURE in 2h 59m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/152086f5a2e042a6bdf5b5b818f49c13 : FAILURE in 3m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a21851b6b934779b16eece57d124de5 : SUCCESS in 1h 24m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c39611dc699245ec9f6baa48a9305567 : SUCCESS in 1h 06m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95043d8d6b184ca98ab2ad36ce98d3ac : SUCCESS in 1h 56m 00s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"ea4c92b01e99802374ec7841a0f929e86f57b81b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:44:40.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"fe5bb9b00bb4db12e240a85e6ad464d92b46a61d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 18:24:17.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7dd52a27065d4b03a5d7abe71df9ebc0 : TIMED_OUT in 3h 01m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cf80aceea19e4f8093e2babfd2cd11b5 : TIMED_OUT in 3h 01m 43s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/43792c764f094deea2fa5a90234ec915 : SUCCESS in 57m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9ecb00b2a9474741b784a5be5a2e4d3a : SUCCESS in 1h 02m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e826c468c9454a7faee7a7f250caef20 : SUCCESS in 1h 23m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a0a2cd7b77204688b253a642b017837b : SUCCESS in 59m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7389e57cfc9e4f64934b01a5eeec3c92 : SUCCESS in 1h 07m 45s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8e0d25e579094ed8894f7d55a1cc3ada : SUCCESS in 56m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b4009253f86c40ba824a6064d822c796 : SUCCESS in 1h 18m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3fff7ba8a7e0428ab4a2b9e4070aba77 : SUCCESS in 48m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/524396d2b68449d48ef600cb5d6e362a : FAILURE in 21m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/94cb4680e52a4c4bb1dc6ec4be983320 : FAILURE in 18m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e0575f71798c434588561664701ed379 : FAILURE in 16m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bff5edcdef84713bcf7070ad6620b70 : FAILURE in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff4037c308bc455da6c56477cceeeaf6 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/813a787942f9420295ad2442ed4270c5 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90639afd1f6c470a84961ee4f4a6e6c1 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b9ea75f7d3447cbb4103d752b18886 : POST_FAILURE in 12m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71fd5cb90f5e42da9a91516fea02750c : FAILURE in 28m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/44374f6f0b5f4d04bf59b09b0a14f07b : FAILURE in 35m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/921f3f03a5434a69867a148d331731f4 : SUCCESS in 50m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f1ee2d2d5c64b3f8a4e040493376033 : FAILURE in 23m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eccddabbfd804b0f9902bf6d644479d2 : FAILURE in 19m 08s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"7316687114774b58e74f0cc8f41354aa4c150cdc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-15 17:01:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fbbaac3ca481d53ca337d5ebe0dde2cb809eab2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 19:07:35.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9926e39bfe67461f8f9c5534ccae4def : FAILURE in 5m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b812df64e12942d593612e3507ee214d : FAILURE in 42m 15s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/565be1e57e274568b7fe5493cc4ac224 : SUCCESS in 28m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f92453e2eafb451088102b196ce75ab8 : SUCCESS in 35m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0ad73c4a0a71408a833a661f3cedaaf5 : TIMED_OUT in 2h 02m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ba1e96b9a5ef4d36a36de0bfcecb9e24 : SUCCESS in 1h 01m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/68dd09dd86e34d1cb2ae30e7e87ecb49 : SUCCESS in 1h 30m 57s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2081afc2e0be4c7790bf996eba68bbc8 : SUCCESS in 56m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/77370182432f47cb8efcb30e83ca70d7 : SUCCESS in 1h 15m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a245a2144665492784a005c705f32f40 : SUCCESS in 50m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ef321b3030fb4e979d10c96c1ee911f3 : FAILURE in 19m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/163cbac46ad148efbb58c3554ddc22d5 : FAILURE in 21m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6f0eb364851b4d038da464b65bcf1067 : SUCCESS in 51m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e842fb3e067458ab18b51cc26799ce6 : FAILURE in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f84be03a11a148799658931b4dfaf8c4 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3de3deb4f164911a0481924cc7479dc : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7d0ee5611044a4a87a1123087686182 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b72733a4a934ca7b14ffd1f249f3839 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85bd1dc1d06a4011b3a3d865bf669999 : FAILURE in 4m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c830c3d7aca74626b8d75bf9c1634969 : FAILURE in 1h 41m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f14a3c4c21d4df089754b9235a28886 : SUCCESS in 54m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/72c31971e8bc405a8a9172336ff35c98 : SUCCESS in 50m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/edaa0e5cad2141828fb452fcdb0db2b5 : FAILURE in 17m 56s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"9f8efe00460a7e033dbed805442070c4d4d96f87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-18 10:05:05.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"283b2da3a713bab20325276931b5ef052320b48a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:22:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/87b81fc15cc14af8b895ac107b98692f : RETRY_LIMIT in 1m 48s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3172a46d48394be2a39ede14f06b30b6 : TIMED_OUT in 3h 01m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/13094760cdab40a495e1b4c1c952772c : SUCCESS in 24m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6b2b5a47dd594c129e83423cc795ff5c : SUCCESS in 39m 46s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fb392ff0f11f437a964cab383e14a169 : SUCCESS in 1h 23m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2b572ae98a8e43d28db897ee9332a7ab : RETRY_LIMIT in 1m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d43af915cce34c0cbdcbd4d77b23c632 : SUCCESS in 1h 07m 18s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/69b4bbd311ea4b38b6933c9801a30f18 : SUCCESS in 58m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/07668b07985a41fda8b65b7e1775c082 : SUCCESS in 1h 19m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/943d8e54d67b4f7cae3743a3e42b6c5f : RETRY_LIMIT in 6m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ef7b01bcfc0e487fbca5ac86608c0b57 : FAILURE in 20m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/912c3e18a9244c03903e121cb3c9d24c : FAILURE in 21m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f9318f964dba4afa8cf4ae6d0ddc4509 : FAILURE in 15m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b7f8b4a2c0486e85c3313564f8e1c0 : RETRY_LIMIT in 2m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88a1e4daf4844871b9b8ec9fc827e03c : RETRY_LIMIT in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/473aef60837245d7b9707d07565a471b : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2be8332f7114db4958ee0a7ed7bef99 : RETRY_LIMIT in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/000885f007e9461ca94b47608db40443 : RETRY_LIMIT in 2m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8e30334447f4963b8b2f92615c2c37d : FAILURE in 23m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2dc8de7115c640b5831797eca71e1170 : FAILURE in 28m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/351f4fd4b8034aaaab5d9dbb999f73dd : FAILURE in 19m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/080900ceb3e7470599fada7610c18856 : FAILURE in 20m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4ec6f16ad3cf434389fbcdf6d2707ff4 : FAILURE in 19m 36s","accounts_in_message":[],"_revision_number":5},{"id":"31d60a43482b7f61abbf144c7d15de4ffd789597","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 12:07:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4aa478b224c2e3878b2d278181f4649186e28ab5","revisions":{"d3cb9bd4ee0a2eb8b570782782745dcc848249b3":{"kind":"REWORK","_number":1,"created":"2021-01-11 12:43:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/770093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/770093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/770093/1"}}},"commit":{"parents":[{"commit":"ac9190bee2cf6661b69ce4cfdc1c3cd75aec8da1","subject":"CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac9190bee2cf6661b69ce4cfdc1c3cd75aec8da1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-11 12:43:47.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3cb9bd4ee0a2eb8b570782782745dcc848249b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3cb9bd4ee0a2eb8b570782782745dcc848249b3"}]},"branch":"refs/heads/stable/train"},"20673e0869c301d6a06452b0620dab3a741628cd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-12 11:15:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/770093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/770093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/770093/2"}}},"commit":{"parents":[{"commit":"60b60a66b22ed026c3a5849517b68aa030a8a05a","subject":"CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60b60a66b22ed026c3a5849517b68aa030a8a05a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-12 11:15:38.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20673e0869c301d6a06452b0620dab3a741628cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20673e0869c301d6a06452b0620dab3a741628cd"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"51810d79178bae4854cabd8dc0493239c130c753":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-13 16:14:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/770093/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/770093/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/770093/3"}}},"commit":{"parents":[{"commit":"9b0c3de139258b09312e3a10bc0b7a1cbbdd3403","subject":"CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b0c3de139258b09312e3a10bc0b7a1cbbdd3403"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 16:14:52.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51810d79178bae4854cabd8dc0493239c130c753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51810d79178bae4854cabd8dc0493239c130c753"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"60fb7e4baa2202851698aaae5207850a96456c9b":{"kind":"REWORK","_number":4,"created":"2021-01-14 09:44:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/770093/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/770093/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/770093/4"}}},"commit":{"parents":[{"commit":"9b0c3de139258b09312e3a10bc0b7a1cbbdd3403","subject":"CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b0c3de139258b09312e3a10bc0b7a1cbbdd3403"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-14 09:44:27.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60fb7e4baa2202851698aaae5207850a96456c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60fb7e4baa2202851698aaae5207850a96456c9b"}]},"branch":"refs/heads/stable/train"},"4aa478b224c2e3878b2d278181f4649186e28ab5":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-01-18 10:05:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/770093/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/770093/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/770093/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/770093/5"}}},"commit":{"parents":[{"commit":"9b0c3de139258b09312e3a10bc0b7a1cbbdd3403","subject":"CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b0c3de139258b09312e3a10bc0b7a1cbbdd3403"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-18 10:05:01.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nDepends-On: https://review.opendev.org/771109\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4aa478b224c2e3878b2d278181f4649186e28ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4aa478b224c2e3878b2d278181f4649186e28ab5"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
