)]}'
{"id":"openstack%2Fkolla~770514","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I53e8ba8d4baa4434cf3c49ed94d1f7cb44099bee","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"I53e8ba8d4baa4434cf3c49ed94d1f7cb44099bee","subject":"Add a storm-base image","status":"MERGED","created":"2021-01-13 10:09:25.000000000","updated":"2021-01-13 21:23:21.000000000","submitted":"2021-01-13 21:21:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"770514","cherry_pick_of_change":770445,"cherry_pick_of_patch_set":1,"meta_rev_id":"b4639aa29134b14cc93434267c249d48b1aca145","_number":770514,"virtual_id_number":770514,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 21:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-01-13 11:01:04.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-01-13 11:55:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-01-13 11:55:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 10:10:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-01-13 10:16:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-01-13 10:16:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-13 11:53:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-13 14:07:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df89fce77b8c5ffd5dbf997395894c179d9f35f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 10:09:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/storm/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"99a62ae206f7c6892772339ea7d5c4a015ed77e8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-13 10:10:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af5ce91d240b2f1d332d24f533ffe2d0f4588a3b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 10:16:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c03a875abc6e0176bc2d86f48bb328fb390fd63b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 10:16:30.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"1c7d363c36244167efb2c07804a7a922f5d2a098","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 10:43:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e5cd0781db3512cb34bb2761712736840ce95186","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 10:44:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f98798bbbbf880240e90a5bf0ef4659517e75e8c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 10:45:13.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"a70ecdfe22a1f1fe3d32b17d7c3a71b5ecea7fa8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-13 11:01:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"84da7c16e354d772734fb57f6fc410ee389595ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 11:53:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/94653b2546b04fc3a226b256ccc50bc1 : RETRY_LIMIT in 3m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a078ffc21d0a414aa0282be94d985d33 : RETRY_LIMIT in 2m 56s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f8b0907ba1c4754aba1d650676a8b04 : RETRY_LIMIT in 2m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b0e151bb6d54b918144eae34abe2e32 : RETRY_LIMIT in 2m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ab1eb1f903429095c7e70d68f5d1e7d2ab35254e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 11:55:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5cfabbc38e530c91678ac36a2a59240c1abbf23c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 14:07:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3335b3c3e9c44db7b254c4e8d00081fe : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/01c12b6c80ee4ff184d09ab332274a97 : SUCCESS in 21m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d6bd95f8df534fb994644517c9fd962b : SUCCESS in 1h 06m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4ad9293b2ff94c81af7ffacc88ba450d : SUCCESS in 1h 25m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a06a2da2b49a48eca988a15a71a45bfb : SUCCESS in 1h 42m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3ee905e0f1f646b6adfd299bf25a57d0 : SUCCESS in 50m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2f5e2a7e486c437c83993104e14d5bb2 : SUCCESS in 50m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/424f840b5aa545e8b7fd61dd6596ad2d : SUCCESS in 59m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6492b8e6eff84cce8df7e84a4632d8b8 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f26f8c400cd245cd974da92e58932088 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9259b32f96f945688e8e500846cbef65 : RETRY_LIMIT in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb70d1f9eb240dd8b8563ee29b04b75 : SUCCESS in 19m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f48b24a1c8ba492788d253ec720fe2db : SUCCESS in 2h 56m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/56aa906913954ab29b9f0c7bf42f8b62 : SUCCESS in 29m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2fa7841c33b4fc8800b977af1361ceb : SUCCESS in 1h 10m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e3e5a204be04cbfa7d41fe39e048672 : SUCCESS in 1h 14m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d251226e609d45a7b2b2bfa59854a0d0 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":3},{"id":"deeba59a71e1b6d015665d5b3eca8efa7c86ae6b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 14:58:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"711dc7757e2437adae0f1ba77fd1841809b25319","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 16:09:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/71802bfd6dfa4cd2931bb833f6c5fd16 : RETRY_LIMIT in 2m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/36183750fbc4431e837902ffa085d6b2 : RETRY_LIMIT in 3m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2143a93f05804ebba33545b22a7f1d2f : RETRY_LIMIT in 3m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e49921b55b354db5bbf4e5032083f1ac : RETRY_LIMIT in 2m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f99f3d6df8dcbd4006f73f788a5cc05ece8b20ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 17:36:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/22fceccce560495d83db6d15f7795f93 : SUCCESS in 27m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7afe707f3d0f4ee6ab6c9ceb3e2657db : SUCCESS in 21m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6d44ef6bf6024a67b3df506bde6d93e9 : SUCCESS in 1h 06m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a69212e2941744a380e75b1787afaccc : SUCCESS in 1h 03m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d024f15a8b4048c09cfdf068d6ef8932 : SUCCESS in 1h 06m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1b502f6ad50145f0ae784ed53f2b758c : SUCCESS in 21m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0cbe87c80cf74e9faacc8f639478e990 : SUCCESS in 24m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f89a21200b1f41068b3822cef0c97e9d : SUCCESS in 1h 00m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b09d0a5cee8a4b1a92dfc9c738705bf1 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/765acf68d0fc4e2b959ff86f1434689e : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd19161076840229b704a87232dfd2f : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8b34af5b394746a2609e3c6c7c650c : SUCCESS in 9m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f0fd521a0b9c433f9505b8abeb716b46 : SUCCESS in 1h 32m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/700c544efb7a4712940d277c5488e6b4 : FAILURE in 30m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/69c54766f47c4fb9866a27315abf86f3 : SUCCESS in 58m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c6a935980e534218ab7df7ecaa629271 : SUCCESS in 54m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a0336a7a771547bdaab23b0ca9c7177e : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":3},{"id":"ccbb710ae8f7f3c1bee98a02096cd8da9aa388f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 18:24:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"39bf82fe68e935f74104151bcac5db994e2b2f3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:48:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2dfa7bff9379450ba9a2f27ad01ae8d4 : SUCCESS in 59m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a82bcb37dc2c4b5aa0939f57f4951a58 : SUCCESS in 55m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c02baf0e65543569143cf446d73d0c3 : FAILURE in 1h 15m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c487ac0e11784100b587e3d1e88aa0a6 : SUCCESS in 55m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4e6f7752dad1ced8f54fe32856030675e64e3792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:03:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7670eede81ac409897e1e984ea42426d : SUCCESS in 25m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f902707c198e445da34e02ee0e09336f : SUCCESS in 47m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f76b36cfbb324d819cf2a597efab5ae1 : SUCCESS in 1h 07m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bc24b572c9ea4ccb983f065102087ff8 : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ad659c5ccf19444ea3fdef3ee899de46 : SUCCESS in 1h 09m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/01144fe08bb84a768432c6e289507675 : SUCCESS in 16m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/959dab52cc9e435c9f85d55c95607ac8 : SUCCESS in 45m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0cb11ff570874f4382d303834d940650 : SUCCESS in 1h 22m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79d173dbd3b6416ba4e8caa33413657e : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f72651979504e17abb317d474787928 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1efd08d10bcb419c8a6ba15fb9f97f4f : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53d58598c3842c6acf0cc7c5f973c0d : SUCCESS in 12m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e256a3a8a5844f6a7a5e18434f0c2c7 : SUCCESS in 45m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8bfdb18b030d4363a56736aa7f4218a0 : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b020cc705264a49a9edefde80f263fe : SUCCESS in 54m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b744e1717d354f47af2dc683cd20552d : SUCCESS in 51m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ca594fb6548b4458951f2982e5733e1f : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":3},{"id":"6f88bd0834ade06d338e14b4eaa58d3841536848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:03:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2969f3595caea01f4e2d19c875e9d45674e2bfce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 21:21:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/dbc9d531b29440b6b5b624c19577a2e4 : SUCCESS in 22m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/42ee241a3ed44188bd617334547cbd47 : SUCCESS in 1h 03m 13s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1a500fd7c5ce469bb155f9fbc2b1a6d4 : SUCCESS in 57m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c50e657fc354ee1962f48d715c7d181 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7be1c9b8b5b4fd68242b587f3e5712c : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee987358026a471c964fa1de8b513a43 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a35af8b6e444b378fa6cb547a94f84d : SUCCESS in 5m 01s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b44cd6005444268b68e31433c025085 : SUCCESS in 19m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e2f9c7f201b4fb2bd83f44d33b3b65d : SUCCESS in 54m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8bd48a85810746b1bbaec510e643597a : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":3},{"id":"8b54a204340eeb6cb8839c7c623edf085b1d1969","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 21:21:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b4639aa29134b14cc93434267c249d48b1aca145","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 21:23:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83c4ef1b3b5441a09743e705a51c7e91 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"db7dfe14130ed1d3a9afe480283e5744fa82e7e5","revisions":{"91fed5b9d08952bf470e949be3f9a1056047d7e7":{"kind":"REWORK","_number":1,"created":"2021-01-13 10:09:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/770514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/770514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/770514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/770514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/770514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/770514/1"}}},"commit":{"parents":[{"commit":"cf795f3003066be53664a6b2709d20b3f1af07f7","subject":"Bump prometheus-openstack-exporter version to 1.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf795f3003066be53664a6b2709d20b3f1af07f7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-12 16:18:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 10:09:25.000000000","tz":0},"subject":"Add a storm-base image","message":"Add a storm-base image\n\nRefactor storm image dependencies. Previously we had:\n\nbase -\u003e storm -\u003e monasca-thresh\n\nBoth storm and monasca-thresh images are final, and each includes a\nfooter. This causes issues when the footer expects to be at the end of\nthe image, or there to be only a single footer. In particular, since\nI2592a736206eaec811290e9fbdbf2540c0518ffe, the footer used in CI jobs\nreverts to the public package mirrors.\n\nThis change introduces a storm-base image, and refactors the dependency\nchain:\n\nbase -\u003e storm-base -\u003e storm\nbase -\u003e storm-base -\u003e monasca-thresh\n\nRelated-Bug: #1902101\n\nChange-Id: I53e8ba8d4baa4434cf3c49ed94d1f7cb44099bee\n(cherry picked from commit 93204ecfeca7ba394bd4c3872147357aa708f46e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91fed5b9d08952bf470e949be3f9a1056047d7e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91fed5b9d08952bf470e949be3f9a1056047d7e7"}]},"branch":"refs/heads/stable/victoria"},"4dbf78ae136127e12c68a227879beab21c169348":{"kind":"REWORK","_number":2,"created":"2021-01-13 10:43:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/770514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/770514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/770514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/770514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/770514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/770514/2"}}},"commit":{"parents":[{"commit":"cf795f3003066be53664a6b2709d20b3f1af07f7","subject":"Bump prometheus-openstack-exporter version to 1.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf795f3003066be53664a6b2709d20b3f1af07f7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-12 16:18:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 10:10:14.000000000","tz":0},"subject":"Add a storm-base image","message":"Add a storm-base image\n\nRefactor storm image dependencies. Previously we had:\n\nbase -\u003e storm -\u003e monasca-thresh\n\nBoth storm and monasca-thresh images are final, and each includes a\nfooter. This causes issues when the footer expects to be at the end of\nthe image, or there to be only a single footer. In particular, since\nI2592a736206eaec811290e9fbdbf2540c0518ffe, the footer used in CI jobs\nreverts to the public package mirrors.\n\nThis change introduces a storm-base image, and refactors the dependency\nchain:\n\nbase -\u003e storm-base -\u003e storm\nbase -\u003e storm-base -\u003e monasca-thresh\n\nRelated-Bug: #1902101\n\nChange-Id: I53e8ba8d4baa4434cf3c49ed94d1f7cb44099bee\n(cherry picked from commit 93204ecfeca7ba394bd4c3872147357aa708f46e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dbf78ae136127e12c68a227879beab21c169348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dbf78ae136127e12c68a227879beab21c169348"}]},"branch":"refs/heads/stable/victoria"},"db7dfe14130ed1d3a9afe480283e5744fa82e7e5":{"kind":"REWORK","_number":3,"created":"2021-01-13 10:44:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/770514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/770514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/770514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/770514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/770514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/770514/3"}}},"commit":{"parents":[{"commit":"cf795f3003066be53664a6b2709d20b3f1af07f7","subject":"Bump prometheus-openstack-exporter version to 1.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf795f3003066be53664a6b2709d20b3f1af07f7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-12 16:18:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 10:44:42.000000000","tz":0},"subject":"Add a storm-base image","message":"Add a storm-base image\n\nRefactor storm image dependencies. Previously we had:\n\nbase -\u003e storm -\u003e monasca-thresh\n\nBoth storm and monasca-thresh images are final, and each includes a\nfooter. This causes issues when the footer expects to be at the end of\nthe image, or there to be only a single footer. In particular, since\nI2592a736206eaec811290e9fbdbf2540c0518ffe, the footer used in CI jobs\nreverts to the public package mirrors.\n\nThis change introduces a storm-base image, and refactors the dependency\nchain:\n\nbase -\u003e storm-base -\u003e storm\nbase -\u003e storm-base -\u003e monasca-thresh\n\nRelated-Bug: #1902101\n\nChange-Id: I53e8ba8d4baa4434cf3c49ed94d1f7cb44099bee\n(cherry picked from commit 93204ecfeca7ba394bd4c3872147357aa708f46e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db7dfe14130ed1d3a9afe480283e5744fa82e7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db7dfe14130ed1d3a9afe480283e5744fa82e7e5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
