)]}'
{"id":"openstack%2Fkolla~770727","triplet_id":"openstack%2Fkolla~master~Ia636747bc2d886f44abb2f4f816fcb621b90a298","project":"openstack/kolla","branch":"master","topic":"feat/remove-novajoin-image","hashtags":[],"change_id":"Ia636747bc2d886f44abb2f4f816fcb621b90a298","subject":"Remove novajoin image","status":"MERGED","created":"2021-01-14 08:05:49.000000000","updated":"2021-01-14 14:49:33.000000000","submitted":"2021-01-14 13:06:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770727-feat/remove-novajoin-image","meta_rev_id":"83a6a89a494250f23e619f5c105c009d60c7dead","_number":770727,"virtual_id_number":770727,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 13:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-14 09:23:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-14 10:37:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-01-14 10:37:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-14 09:23:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-14 10:37:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-14 11:19:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8a051ece926be26f4810b36e8656fef828f3730","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2021-01-14 08:05:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cf7a0134ee196ee056129638fb0e4c43cbc81d8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:23:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12ab6ff6a253fba439610eba0d1e7defb6799c77","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 10:37:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51b725a41c175054b9a255159e6339a747eac211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 11:19:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/07875902f6374cbbb45a586f38ed846c : SUCCESS in 24m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/bbc8fe8d715b41508c09b3b6cae631ac : SUCCESS in 1h 04m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0a6160984bd4429194668b43299df1d8 : SUCCESS in 1h 48m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4d064bf2c32940e1add4448d587dcc4d : SUCCESS in 1h 09m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3d5d269935714c4c9441885783a8c383 : SUCCESS in 1h 12m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cdced3870a2b41379a87705c7d62bd3a : SUCCESS in 31m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/70e68bc7ef764515bb5e1b820d31b26f : SUCCESS in 21m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c211ca755dc040dba801ddf3d0c6ae73 : SUCCESS in 55m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bbd6486c44243d09f44954f3512fd20 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea409e6e6440473ca84d29e593cfc324 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15bf39ee67fb4ef4b1b9f8b5ff78b227 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2acb5eebce7b4c7ebbd7a966e21dd87f : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203d091cf6fa4305b3f08f1c5e9a8e77 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84a5eea1eadf4860955a29e22b368c4f : SUCCESS in 4m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7405888c8ce14439844e12f62a2bee52 : TIMED_OUT in 3h 05m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb9205b11cc945f99e011f65853ea25a : SUCCESS in 1h 00m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/72e6083538004187b3baf36bf5688a2c : SUCCESS in 1h 12m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c7f5448bd154894b6dc58e40c475925 : SUCCESS in 1h 09m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/497449d5a7024f788959ef835bad88f4 : SUCCESS in 1h 28m 11s","accounts_in_message":[],"_revision_number":1},{"id":"24a096989dd41cd4c180945cbe4a8327a228c468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 11:19:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0423a6ab29e2cfe9114c8d327b78946e1546bad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:06:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9e934ac35b344e3dbbafe5aae3d20d32 : SUCCESS in 41m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9a74f78700fd459f9c75671a9221fea6 : SUCCESS in 1h 03m 50s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/42db9c57efa2496c9b6cb36f32889eac : SUCCESS in 1h 07m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e63c23dda714020bc47cf29eaf9524b : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34c4a291d0394893a933198a33c6280f : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1741d9f4e3d74a67acecbc464b46f04c : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef1312ec1ca040f8a56c78268a100ae8 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bc070e59d5c4d4290514be462f3340c : SUCCESS in 4m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e4efa5220bb43d2ab81bcadacffbacd : SUCCESS in 1h 05m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a9dd3e69c1f4b66b274e5c7423b5beb : SUCCESS in 55m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9e16d1b892b340ca8bd77a006f4d2466 : SUCCESS in 1h 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b4d59f7bc17013ae860bfeca4bd55dcf7ba86e2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:06:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4a542117b5c9b64773288c615d45cb20eb2bb77f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:17:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac9ef1de8ae4c8b8fb85e795e2195c4 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5277b4779d7a4cc8b30b648115a10008 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1},{"id":"83a6a89a494250f23e619f5c105c009d60c7dead","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 14:49:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a17ed618de147a6ae2860850dccbe21 : TIMED_OUT in 3h 16m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d1c8d0413bf4b369cc4c2286246dad5 : TIMED_OUT in 3h 16m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ac5ddc552c234d38bfb191049eb32b9a : TIMED_OUT in 3h 02m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4204ab0529dd42628fed18be81f7189c : TIMED_OUT in 3h 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc2a97f0f1e84aebf0a59e654d6793439dfc4d18","revisions":{"cc2a97f0f1e84aebf0a59e654d6793439dfc4d18":{"kind":"REWORK","_number":1,"created":"2021-01-14 08:05:49.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/27/770727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/770727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/770727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/770727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/770727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/770727/1"}}},"commit":{"parents":[{"commit":"93204ecfeca7ba394bd4c3872147357aa708f46e","subject":"Add a storm-base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93204ecfeca7ba394bd4c3872147357aa708f46e"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:05:35.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:05:35.000000000","tz":60},"subject":"Remove novajoin image","message":"Remove novajoin image\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: Ia636747bc2d886f44abb2f4f816fcb621b90a298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc2a97f0f1e84aebf0a59e654d6793439dfc4d18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc2a97f0f1e84aebf0a59e654d6793439dfc4d18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
