)]}'
{"id":"openstack%2Fkolla~770729","triplet_id":"openstack%2Fkolla~master~I3d101f171bd1cc6f3c192ddc98cea0d63bccb739","project":"openstack/kolla","branch":"master","topic":"feat/remove-nova-mksproxy-image","hashtags":[],"change_id":"I3d101f171bd1cc6f3c192ddc98cea0d63bccb739","subject":"Remove nova-mksproxy image","status":"MERGED","created":"2021-01-14 08:15:03.000000000","updated":"2021-01-25 15:17:55.000000000","submitted":"2021-01-19 13:31:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770729-feat/remove-nova-mksproxy-image","meta_rev_id":"8892958bd281c27e8b57eeb34edab9646013eb3a","_number":770729,"virtual_id_number":770729,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 13:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-01-19 11:45:03.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-01-15 16:13:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-25 15:17:55.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-01-19 11:45:03.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-14 09:31:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-14 10:18:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-14 10:38:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-01-19 11:45:03.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-01-25 15:17:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"f0738bb8dce8c894f9faa23c0a21455155873327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2021-01-14 08:15:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaf80f60575f14385f0e7228e3a8ac2d0c4a01b3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:31:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c72283add121af66753df44bb1d40376d27f3f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 10:18:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3e620d0640f2465db67519008ebd9487 : SUCCESS in 25m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b3bb4fe17bcf4257a06045f79af6dbfc : SUCCESS in 1h 19m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fe27af951b8049719f8108569acb2424 : SUCCESS in 1h 07m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cdf5fb79b29e41e6ad6a12e32119f105 : SUCCESS in 2h 01m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/97831dbd9c8e492db2515dfdfde3c451 : SUCCESS in 1h 33m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3219562942874ab8a3e086f4b58a74d7 : SUCCESS in 1h 08m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a4413d8e3ab1448f8bede025cd14644f : SUCCESS in 19m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3b3d170be43e4e85b851c5aa18bd7693 : SUCCESS in 52m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eafaa7ed888d4583843faf12dd30f8ce : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00151395be74495b88c38436c064631a : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97efb77fc306494bb2cb3a5a3bf9fe64 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45946e6bfb2046d385aa3481fb85d1ae : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b476c91930ef4ca0866d2512ba65c60d : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec849bcd34ea4cc4a598d5905e8c62bc : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d02b39b30b7e43498e8316b92c38b171 : SUCCESS in 23m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/399a648acf274685b2992da6dc37c7b9 : SUCCESS in 53m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ac7416503b74950824c855ef0028227 : SUCCESS in 1h 13m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d664a4627389418d9d50defac676b957 : SUCCESS in 52m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/11f36bc6d0c44640a7289e21bfcaaa61 : SUCCESS in 1h 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"77fddbb830b5f56172c5e9d440050f0223260001","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 10:38:21.000000000","message":"Patch Set 1:\n\nneed to rebase on \"Remove novajoin image\" and \"Remove zaqar image...\" as it conflicts greatly","accounts_in_message":[],"_revision_number":1},{"id":"5fdfbafbf971529d9fb77e5bc60af629eb46bc83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 16:20:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/56cd1c04578d4f039a678c6e9c15459a : TIMED_OUT in 3h 20m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d91239092a58468bab940f7f7d637fe9 : TIMED_OUT in 3h 12m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a92689dc51c448d875cbc709f6c1cc4 : TIMED_OUT in 3h 02m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/de4a0d78b15b4d43bbde637601d65d8a : TIMED_OUT in 3h 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2bf0fb1f3561bcc279b24fdd00af3a612bd63df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2021-01-15 13:09:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2445bbd8cde86831e7e0b54004b286b91239e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 14:36:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b53aa3e25140452cbd1be16fcd802255 : SUCCESS in 25m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5bb429dc2d744c1890b95c94fc9c6f45 : SUCCESS in 44m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7fe1ae6952c2435ca7dcdc6a2958f3dd : SUCCESS in 1h 03m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ed39ec1c206d4e488ffa5805caa90de5 : SUCCESS in 1h 01m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e7d7fbe16ad2446d831041a08beb2486 : SUCCESS in 1h 12m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/19dd3f47a7004cb78ed260679032e0ef : SUCCESS in 30m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/25dc67b737db47e1a0c5cea23983bd01 : SUCCESS in 27m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/df3737d7ae0d4652b512f6fff6c6331d : FAILURE in 16m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e9073e83c145babe0b142e5dd2686e : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26954033ac7c45c7809e6a81522a385f : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f744357060c4d9e9a840b134ffd2167 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c09347a59a74868b6b81f87b26b0a7f : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9914edc915bf4bd7b68abc89b67a457b : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b85b7cea33884a0f8c56c59a0ad8cf61 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c2d78d4095dd467eabe0a3af6038857b : SUCCESS in 34m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/edae888070454884b413023930db064a : SUCCESS in 28m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c33a8b92701c4a67ac3be6cfbd6cc38d : SUCCESS in 51m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/701b818bee06433ab05e35fb5ed6b81e : FAILURE in 17m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/32e45c7274f140f69f79d03a43d56f29 : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7ff88cfbdb77de4b94d4f778d9265e2ce598b923","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 14:46:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ffabb856a5a4c0cbbefcd41b8cfccde : SUCCESS in 1h 30m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/459d9bad1110405ca23c5cb829ab7779 : SUCCESS in 1h 28m 56s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c436e4f6034467e87e438d116aed1c1 : FAILURE in 1h 26m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/40a15609121b42cdb2919079da66900a : SUCCESS in 1h 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64df1851c9c62f7bdb3bcffca641998955437111","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-15 16:13:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"945b2a07064b743f15f24d33a45cc7f60ef33b3c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-19 11:45:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b90ec4cee6f5bc0bcbbc06b66b46098d255579f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:45:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1fb1cc0ca3b3c7a0cba7cd093199a2229c104cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 13:31:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/959edd3c14b44d85aa2cbf7f0e8f444c : SUCCESS in 42m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ec371047042a40b99c6dee4addde1cd5 : SUCCESS in 1h 16m 33s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e2b288dc90bd4a05bce112e2a58bcb72 : SUCCESS in 1h 25m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/063a34ecedd3432188a2f832888aded1 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acc8d61aeb054935850c50ac2db0540b : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3ea1ff5d4a941909309d2d99d4ba81d : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2311b75dad3448ddbfc78eb877bf15cc : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34f1af7a380f460d87895a55cdb988ad : SUCCESS in 4m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/81eb43d4fcdb42aea2b69fe66d3d096d : SUCCESS in 27m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b354a46a7fd54a80a2b141e89b9a0b5b : SUCCESS in 56m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac899e8139b946d28931cda29b504d69 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":2},{"id":"05ca64e0231bc6ccd180618c8b3967196d9e25d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 13:31:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3d2df5bb53ab1e9904f022b598f4a2ecb125b354","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 13:33:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d34b49bfa4964fa8b13c1d0b070ec332 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1f7c89bd16a4ebe93f9634b7b965609 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8892958bd281c27e8b57eeb34edab9646013eb3a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-25 15:17:55.000000000","message":"Patch Set 2: Code-Review+2\n\nah, this is what blocked zaqar","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36e5ef2d78dda256efc87242d22baffb7059802b","revisions":{"fbd2e1e84cea2ed5613860c2e52f8e529a3a8143":{"kind":"REWORK","_number":1,"created":"2021-01-14 08:15:03.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/29/770729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/770729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/770729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/770729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/770729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/770729/1"}}},"commit":{"parents":[{"commit":"93204ecfeca7ba394bd4c3872147357aa708f46e","subject":"Add a storm-base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93204ecfeca7ba394bd4c3872147357aa708f46e"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:13:51.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:13:51.000000000","tz":60},"subject":"Remove nova-mksproxy image","message":"Remove nova-mksproxy image\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: I3d101f171bd1cc6f3c192ddc98cea0d63bccb739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbd2e1e84cea2ed5613860c2e52f8e529a3a8143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbd2e1e84cea2ed5613860c2e52f8e529a3a8143"}]},"branch":"refs/heads/master"},"36e5ef2d78dda256efc87242d22baffb7059802b":{"kind":"REWORK","_number":2,"created":"2021-01-15 13:09:58.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/29/770729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/770729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/770729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/770729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/770729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/770729/2"}}},"commit":{"parents":[{"commit":"8bf9e8a90ce94880c351e96634c82ce04f86d5b1","subject":"Merge \"Remove ec2-api image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bf9e8a90ce94880c351e96634c82ce04f86d5b1"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:13:51.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-15 13:09:23.000000000","tz":60},"subject":"Remove nova-mksproxy image","message":"Remove nova-mksproxy image\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: I3d101f171bd1cc6f3c192ddc98cea0d63bccb739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e5ef2d78dda256efc87242d22baffb7059802b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e5ef2d78dda256efc87242d22baffb7059802b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
