)]}'
{"id":"openstack%2Fkolla~770730","triplet_id":"openstack%2Fkolla~master~If4f4e86619335bca6819f2cb6bc8e718e613cd5c","project":"openstack/kolla","branch":"master","topic":"feat/remove/ptp-image","hashtags":[],"change_id":"If4f4e86619335bca6819f2cb6bc8e718e613cd5c","subject":"Remove ptp image","status":"MERGED","created":"2021-01-14 08:17:17.000000000","updated":"2021-01-14 17:26:31.000000000","submitted":"2021-01-14 13:28:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770730-feat/remove/ptp-image","meta_rev_id":"072e4e56713662abcbcb1f22811c93b24885ba53","_number":770730,"virtual_id_number":770730,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 13:28:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-14 09:32:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-14 10:33:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-01-14 10:33:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-14 09:32:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-14 10:33:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-14 12:06:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2736ccc9a74e3a17c9d4a66fe02873b656638111","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2021-01-14 08:17:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cb6b292838267fb4575224fb3077210a3b9e3cb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:32:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b2e5fe14e08b39002fd1a6d421b80318bb5b8b5f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 10:33:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5191830b79af443db0bd58641b17b754215f3ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 12:06:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/18e3a721f5104559a4cfb1f325d95427 : SUCCESS in 3m 04s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d217570a3c014508bec55555ffd9add1 : SUCCESS in 48m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ad007ef74e814732a3fcf0d6f3f36ef5 : SUCCESS in 36m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f861c64cf7564a8d944abbb4a92e5280 : SUCCESS in 1h 30m 23s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/87153a090c384242b510016a218efb9b : SUCCESS in 51m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/21990d7a77ad4e489b0b4456924ecdab : SUCCESS in 1h 19m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d064cb2de24f49d9a336c7cbbca211f5 : SUCCESS in 52m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/782d87baeb0644c2abb7fdf021f34785 : FAILURE in 40m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/488f8ff342b04e4e831227f3b10359a6 : SUCCESS in 1h 00m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2298d2cd57cf4445acea907f4de49005 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1116dd9f5044cc78f34a1e025ee64a9 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1411712904e64057b7bbce4fc75780d2 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76576d26ed094e3fa0ce7a9e92cceb00 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07ae8b2944f4fcba25800ff8cc1a81a : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d3326b7c17b4ffeb7eb3a9a95747288 : SUCCESS in 4m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b6745a1648f94cfaa210a67973a00a0c : FAILURE in 19m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/73c8d9353de2483684083a57809b4225 : SUCCESS in 45m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f194b9c892c84178ad77dbd33e09e0d7 : SUCCESS in 57m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f6afff97790849ed9604e890b40d7fb7 : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/28d15999ae554d4490494dcfb1d26533 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a536c59c53ecdb3ad42b078f2baa8fcd23cea7f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 12:06:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c88603b780b85f879ae59d9236e20360d6925d98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:28:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c0f28e1448a340bd956aa5b634f0fed3 : SUCCESS in 3m 01s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a1efc3de4556484a9f1ea4dcd2bdbe24 : SUCCESS in 38m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1aa2416add4d4f2e87a685c862f18a6b : SUCCESS in 1h 02m 29s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5afe86172a7541bf94bad844542632ac : SUCCESS in 1h 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36193c1b0bfc4752884202b9c65e0cff : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b6626525eb94f5bba8ad76c0028fa80 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/003b7bd4d8ec45d7bfa1341f9f8855a1 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fbdf4a930144e4ab56f680d0016df59 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaad6fab65fa4847bca9ce32a768ff96 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/78a1c7e6b0a949f3a87ba25cac78f22b : SUCCESS in 42m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/780a175edf8249cd8600a2a32e8ee559 : SUCCESS in 1h 01m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/06d2c1d9fa424f019c5712fd6c97229c : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5c1f94e061510cf99473e4aca0bc2cc031de6318","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:28:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"37c2fa1e22a6387c9b3b891683a443977832bda0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:29:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969550058e0241e9879213da828c9d3a : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee3912197f9f486c8dab313e360feb94 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"072e4e56713662abcbcb1f22811c93b24885ba53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:26:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e693fdc04e64ca586a2530b11ae225f : TIMED_OUT in 3h 15m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5fb6df4eb16240fcaf99b939b30dfbd7 : TIMED_OUT in 3h 12m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6ac99fb11ba4673b87d551f68221dc2 : TIMED_OUT in 3h 01m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a0a35027bfc24163a6bf97720b33088a : SUCCESS in 58m 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e2b6f6d3c848042f0b13b0817da9694a91e935c","revisions":{"2e2b6f6d3c848042f0b13b0817da9694a91e935c":{"kind":"REWORK","_number":1,"created":"2021-01-14 08:17:17.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/30/770730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/770730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/770730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/770730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/770730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/770730/1"}}},"commit":{"parents":[{"commit":"93204ecfeca7ba394bd4c3872147357aa708f46e","subject":"Add a storm-base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93204ecfeca7ba394bd4c3872147357aa708f46e"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:14:38.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:17:02.000000000","tz":60},"subject":"Remove ptp image","message":"Remove ptp image\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: If4f4e86619335bca6819f2cb6bc8e718e613cd5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e2b6f6d3c848042f0b13b0817da9694a91e935c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e2b6f6d3c848042f0b13b0817da9694a91e935c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
