)]}'
{"id":"openstack%2Fkolla~770757","triplet_id":"openstack%2Fkolla~master~Id3187b2b95e9b9bd3d9233a498f7c45e66c604c7","project":"openstack/kolla","branch":"master","topic":"feat/remove-zaqar-image","hashtags":[],"change_id":"Id3187b2b95e9b9bd3d9233a498f7c45e66c604c7","subject":"Remove zaqar image and related plugins","status":"MERGED","created":"2021-01-14 08:28:55.000000000","updated":"2021-01-25 22:35:38.000000000","submitted":"2021-01-25 22:33:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770757-feat/remove-zaqar-image","meta_rev_id":"914ac2887b3a8a08e3ed3fc14cd61bdfc8538332","_number":770757,"virtual_id_number":770757,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-25 22:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-01-25 15:12:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-25 15:12:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-01-25 15:12:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-14 09:26:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-14 09:41:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-01-14 09:44:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-14 17:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-25 15:12:13.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"35fd7bdf0f4d7c3f17964f81cdbc4ee37868e3ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2021-01-14 08:28:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c334d32623bf083d9e5f78bc36b7f7dc5b0ebd0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:26:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"106efbc7c3dc9c0e88d63e6fc62c5cd043369f48","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 09:41:43.000000000","message":"Patch Set 1:\n\nI will rebase this on https://review.opendev.org/c/openstack/kolla/+/770566 to avoid the conflict and have easy victoria backport","accounts_in_message":[],"_revision_number":1},{"id":"698e6acd04ae0f312730d09fb6ac8d6aee260039","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 09:44:52.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ba19c2bab88665414994ae952096a8112940073b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 09:50:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dcdf4f34be732b3049a74d58fd767414349b2bc6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 09:54:05.000000000","message":"Patch Set 2: Code-Review+2\n\nI wonder if we can also remove the zaqar clients scattered around","accounts_in_message":[],"_revision_number":2},{"id":"d781988c150ca62b6694621030461e15b2627ac2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:54:26.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I will rebase this on https://review.opendev.org/c/openstack/kolla/+/770566 to avoid the conflict and have easy victoria backport\n\nI don\u0027t think we need to backport it, it was deprecated in Victoria for removal in Wallaby.","accounts_in_message":[],"_revision_number":2},{"id":"585ff0eb3b7dc476c74109b08d786d086557dba5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 09:55:24.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e I will rebase this on https://review.opendev.org/c/openstack/kolla/+/770566 to avoid the conflict and have easy victoria backport\n\u003e \n\u003e I don\u0027t think we need to backport it, it was deprecated in Victoria for removal in Wallaby.\n\nI meant the Horizon change - as Zaqar would block it.","accounts_in_message":[],"_revision_number":2},{"id":"1be41fc29cdc024e5c0a5eab9177c28fd730f857","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2021-01-14 09:55:51.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+2\n\u003e \n\u003e I wonder if we can also remove the zaqar clients scattered around\n\nI\u0027ve been thinking about this too. But I was not quite sure. I think it doesn\u0027t matter if the client itself stays in for now.","accounts_in_message":[],"_revision_number":2},{"id":"4c2837d721d51846d2bc9e59189ec0e0061727ee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:55:56.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e I will rebase this on https://review.opendev.org/c/openstack/kolla/+/770566 to avoid the conflict and have easy victoria backport\n\u003e \u003e \n\u003e \u003e I don\u0027t think we need to backport it, it was deprecated in Victoria for removal in Wallaby.\n\u003e \n\u003e I meant the Horizon change - as Zaqar would block it.\n\nOk.","accounts_in_message":[],"_revision_number":2},{"id":"38ddb27988d153e9847d1bcf81a8a331eab4ae8f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:56:33.000000000","message":"Patch Set 2:\n\nWe could do a client clean up separately.","accounts_in_message":[],"_revision_number":2},{"id":"20841156f00c0d933be54499a4e36755e63240d4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 10:23:07.000000000","message":"Patch Set 2:\n\nRE client: definitely, it could actually still be pulled in as a dep. Needs checking.","accounts_in_message":[],"_revision_number":2},{"id":"7c9d888c678ddfbf029c11578ed97ecc0c87941c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:30:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9aefac3d89424f13a70ce2d9856ab5d8 : SUCCESS in 3m 56s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/387c752abd5f49d494bc08a10bd07844 : SUCCESS in 22m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d2a5401195164b6098dc4f99002b9c44 : SUCCESS in 40m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/56d83301a9614db28193b6505bfe7ed3 : SUCCESS in 1h 00m 27s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/045092ccb00c45ae97b19c612ddaa5ba : SUCCESS in 1h 01m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0c7e6932fb3944aba02cf0b751b2f394 : SUCCESS in 1h 11m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/891e114bbf914669814d2f1b4816b82e : SUCCESS in 33m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/51f106a50e684d2ca355650421c980a4 : SUCCESS in 18m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/33a640d6ddfe40e6ae08af5ce9886617 : SUCCESS in 1h 02m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d1996498334779a92c1d4fda1c95b0 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9fbb61d2cf147a2ab6a29f73d5f0e24 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04e05cce79704d1ca6ef9eab38965ccd : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9ce4769f5c1436188a79e241756ce95 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21284a7439c34f3ab9f00715aba18129 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3196a206a74b424fb32419c32e318595 : SUCCESS in 3m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d696c5e7e82d462486c45b99916fb07d : SUCCESS in 19m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/99d125cb4f6b4c9b868d1682fc376453 : SUCCESS in 28m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/60587f35fa8849d88079c559eed362eb : SUCCESS in 56m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf73a7cb22444b05995aa0afbfb8ed10 : SUCCESS in 59m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5206fc4c637846949da355d85daec82a : SUCCESS in 1h 35m 35s","accounts_in_message":[],"_revision_number":2},{"id":"09c352dbf4da3035d8e5a6d5a7954423b8947dac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:31:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/97b9c766882b4bb091b73ff6533e2b94 : TIMED_OUT in 3h 12m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7bc5b309ef264e67a492df1569d2a357 : TIMED_OUT in 3h 06m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/87fbcefec2c846d79b7130ff88fdd88e : FAILURE in 2h 49m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ceb26f84cff4f8a7076977dbe4cee1b55bdd09e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-15 09:01:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"00e8dc092f6a65265193feb808b8cb27055f5310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-15 17:25:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"dce62e0e034f4d246077831c209ead2a7ec9aa5d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-15 17:25:49.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b411a6c688c8347b501db9bca8752f69efb31b15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 19:02:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/686e1bdf743b4e1caff22382038596d2 : SUCCESS in 1h 30m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/010d7c06ef2f4ac0a422b380742402bc : SUCCESS in 1h 29m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef22dcfb64cb422d98655d6f906faae8 : FAILURE in 1h 24m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b9a78f5e841e4896bc3de9b61d05211b : SUCCESS in 1h 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f3475a4933eccfdd51463492af072f3dc4dd8443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:53:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cc60dda447f14540abab588e44363781 : SUCCESS in 2m 54s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8e20d0ea7dc34fca9a91a0134091e099 : FAILURE in 5m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3817c47473054c80b3fa90b6de0b480d : TIMED_OUT in 3h 26m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7a6af9bef1364ddab081815783b48916 : SUCCESS in 49m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/77131929e2e746fca52a01480660cc47 : SUCCESS in 1h 03m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1efa0d50d20149a985bd1b0b89a67f1f : SUCCESS in 1h 13m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/17c7346387414f6da00f1a6ddb719bfa : SUCCESS in 30m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/11b780a9e4c547a480d7c6b49c051d4e : SUCCESS in 19m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/785c8e25ddde43bfa21c200270c9435f : SUCCESS in 58m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f47d54348474373a94a7a1550298a1f : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecba67a94e6f4f4089faf9514f67154d : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beee3ea70e49433bb22761647c82edcf : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d5f3497c8a34f24a9ec0ceb3715d65a : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/178dbf312ebe44f8ae71ca7951d3d9db : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5c18523960d4cba9af562724f60ec71 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b31de20c246e40cba5fdcf9bfc39e82c : SUCCESS in 19m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3051159416d84d59af8f7ae1833fbcc2 : SUCCESS in 33m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fd9348dddb24ac5803a8f0b54eb3b5e : SUCCESS in 44m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e202a6b5f464b579a8910454ad7f92c : SUCCESS in 55m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b277b677f2724fa9b9849587bf113361 : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":3},{"id":"c5d490b76b542d74b1394449dc334e6f7d27b0e5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-16 15:13:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"51e7a9fb47eeb4c7078d63fd25aedafd17510420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 16:26:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c26bfbb658994b71a0d718df7a9a17e5 : SUCCESS in 2m 10s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/266da1ff0cd2487d852cdd988fa8fb50 : SUCCESS in 26m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5593dc0f8a704018a0684f1dc813735c : SUCCESS in 43m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2205a404ec714c9190db4e13694fddb0 : SUCCESS in 1h 06m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fe0b62b4aea3458b9cb5ae1597d5f4b9 : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0e1f816ff9e84d2ba2ca6c5aae1a439a : SUCCESS in 1h 10m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f8863563cc5a45a8999052777d4af3ac : SUCCESS in 35m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3b4f4af275d141bea3603df53fc987ed : SUCCESS in 23m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0b9e84bc12434f90aa3592c7f1f99025 : SUCCESS in 50m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4450a0860c3e4c12aca99b5e0fc57902 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e87a182fa304563ac798aa5e77b67c4 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67756cf0683146dbae8f1d51caf0f788 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2db96c73136439b9ee795ca230b4583 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6012797ece4cba9c128c0963a546ec : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a6cc993a1ba4a55931034d3af098eae : SUCCESS in 4m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b81a7778cdd04204883ad110e5084ab2 : SUCCESS in 24m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a08ca8066d44ab2b2581a9cb34a8956 : SUCCESS in 50m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e97dc5c40656455993cbdb4389767a35 : SUCCESS in 1h 01m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2008666eb31744e399308961bc6454e2 : SUCCESS in 59m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fdeffe5a690b4f32813eaa9d6c581025 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":3},{"id":"60c8728b204a9c98b7c3af6bec86b223cb866aca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 18:31:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c8920e2f0ac49218c457355bd6ddf59 : SUCCESS in 3h 04m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/589a43e07a0f466bb7b86c9086f42630 : SUCCESS in 3h 00m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e2db89a57824438abfb44d8ed4c9ffc : TIMED_OUT in 3h 02m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/717a2a9e867b457fb2edf40464a07d5a : SUCCESS in 3h 07m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92651b41c18957aa35f1cc228167e0929269269e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-25 15:10:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bbfa656d7a5df0253b6b340fce689019f56b6489","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-25 15:12:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7b814de4c088936990ba5ce959162e3c85740c27","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-25 15:12:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1952e463f735177b6f1682df2254837df7eb67d6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-25 15:12:57.000000000","message":"Patch Set 4:\n\n(only rebasing in fact)","accounts_in_message":[],"_revision_number":4},{"id":"4b0afa0548b2dd723a44d06402a4148d05d50e15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 16:40:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e03a518464a47154b6d1984c2f07d3b15e350051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 16:41:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7de6dca95e764436a1b815cf65b075b9 : SUCCESS in 1m 54s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/92912611bff948bbac06bc81868249a1 : SUCCESS in 24m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b405b7875b8b43f18d29b45708b46757 : SUCCESS in 42m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5bbdd7679f0b4435be0b6794a64fd38c : SUCCESS in 59m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d62e995ff400483fbd3af5ef67476777 : SUCCESS in 58m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c3c25d1d4cae414185a97dcd3951a961 : SUCCESS in 1h 15m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c0f0974f835d400db2b3082f6c4a0538 : SUCCESS in 27m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5a4d418db2e846ff9ac88dbd5698c797 : SUCCESS in 26m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3e40cb7855f0400ba73b192b4d6fe90e : SUCCESS in 54m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/792ad236938847628bd4a2f4ea078b32 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb9a43e7691c41f19450b23dfc0b32b5 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0049e00553914386a24dbfaa67d2980f : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2a33dd9829341609784ba6593a93298 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e02724360d34291b3fdca1066d04263 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c63e4f95d2496384dfbd01f0596870 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3a3b7d4afcbe4418807b5cfed0e2cd11 : SUCCESS in 27m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9d8fa257a2cb441db210aed8f49fe83e : SUCCESS in 43m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ebf8b14494d43ef8627cc8bd69187cd : SUCCESS in 1h 06m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9c493e193494460ca11a6d930cd041bc : SUCCESS in 55m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb9768e9926448759f93c30ed99337ac : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":4},{"id":"c385ca81f5a10580489fb9dd1c7994bbed4bc9d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 20:46:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2da4aa80a0de392f9cfe74a08d5420841b41ba9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 22:33:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bcfee32f8aad4e0a8934dd9cec1ecf27 : SUCCESS in 1m 58s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ccd7930a6ee64a1596e40841a6b281af : SUCCESS in 50m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cb4f1097c45d4109a37913b8903ecb83 : SUCCESS in 1h 19m 06s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4649a4c0342d4ca2b5b705df34e6e2c2 : SUCCESS in 1h 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508449b7d1db4b78811fd126a3b1769d : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8bfdab4c0e847178b22e540155b2780 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/527ee060f6ed4aa6afbe17b3ef2960bf : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed254b1fbca48679165a04951ea7be9 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c3fada191a8430eb035a3dfd0330c26 : SUCCESS in 4m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/94aaa3a5dcb94b548c4acae303c80543 : SUCCESS in 45m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a9808add36f4e808c845f8f9c323b30 : SUCCESS in 1h 45m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/56145119ecfd4b07a0d109ffa22883ca : SUCCESS in 1h 10m 13s","accounts_in_message":[],"_revision_number":4},{"id":"0effbdc296b72ebb8b6e4128d783f79ebe130c2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 22:33:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"914ac2887b3a8a08e3ed3fc14cd61bdfc8538332","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 22:35:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae95b2ba964b4fa5ae618cc8c4f79db6 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b507c4e385354fdd81dc39efbcfcbac9 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0028073ebbe5335847ebaa720122962e494a478a","revisions":{"4aa25ee56ea75fcb1e68346446c2a017a349bbcc":{"kind":"REWORK","_number":1,"created":"2021-01-14 08:28:55.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/57/770757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/770757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/770757/1"}}},"commit":{"parents":[{"commit":"93204ecfeca7ba394bd4c3872147357aa708f46e","subject":"Add a storm-base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93204ecfeca7ba394bd4c3872147357aa708f46e"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:28:37.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:28:37.000000000","tz":60},"subject":"Remove zaqar image and related plugins","message":"Remove zaqar image and related plugins\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: Id3187b2b95e9b9bd3d9233a498f7c45e66c604c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4aa25ee56ea75fcb1e68346446c2a017a349bbcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4aa25ee56ea75fcb1e68346446c2a017a349bbcc"}]},"branch":"refs/heads/master"},"e08d8154eb265c157a40d14c5d96052bbc7aa893":{"kind":"REWORK","_number":2,"created":"2021-01-14 09:50:53.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/770757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/770757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/770757/2"}}},"commit":{"parents":[{"commit":"8371f097976b66bb78b776b4bb83d40a05490df3","subject":"Horizon: gentler `-o nounset` handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8371f097976b66bb78b776b4bb83d40a05490df3"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:28:37.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-14 09:49:48.000000000","tz":60},"subject":"Remove zaqar image and related plugins","message":"Remove zaqar image and related plugins\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: Id3187b2b95e9b9bd3d9233a498f7c45e66c604c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e08d8154eb265c157a40d14c5d96052bbc7aa893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e08d8154eb265c157a40d14c5d96052bbc7aa893"}]},"branch":"refs/heads/master"},"420fe520ce4fded55d899416c35d4b9b012ea672":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-15 17:25:38.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/770757/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/770757/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/770757/3"}}},"commit":{"parents":[{"commit":"38ccdf215f0ac949a204b231959762bf5a6c313c","subject":"Horizon: gentler `-o nounset` handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38ccdf215f0ac949a204b231959762bf5a6c313c"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:28:37.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-15 17:25:38.000000000","tz":0},"subject":"Remove zaqar image and related plugins","message":"Remove zaqar image and related plugins\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: Id3187b2b95e9b9bd3d9233a498f7c45e66c604c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/420fe520ce4fded55d899416c35d4b9b012ea672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/420fe520ce4fded55d899416c35d4b9b012ea672"}]},"branch":"refs/heads/master","description":"Rebase"},"0028073ebbe5335847ebaa720122962e494a478a":{"kind":"REWORK","_number":4,"created":"2021-01-25 15:10:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/770757/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/770757/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/770757/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/770757/4"}}},"commit":{"parents":[{"commit":"48e6309926f42b46369a7c06cc9f81e9188da802","subject":"Horizon: gentler `-o nounset` handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48e6309926f42b46369a7c06cc9f81e9188da802"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-01-14 08:28:37.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-25 15:10:21.000000000","tz":60},"subject":"Remove zaqar image and related plugins","message":"Remove zaqar image and related plugins\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: Id3187b2b95e9b9bd3d9233a498f7c45e66c604c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0028073ebbe5335847ebaa720122962e494a478a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0028073ebbe5335847ebaa720122962e494a478a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
