)]}'
{"id":"openstack%2Fkolla~771191","triplet_id":"openstack%2Fkolla~master~Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421","subject":"Add octavia-driver-agent","status":"MERGED","created":"2021-01-18 12:09:15.000000000","updated":"2021-02-13 08:02:31.000000000","submitted":"2021-02-13 08:00:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"771191","meta_rev_id":"f5cf58d7791598da9fac202532109eb60df892d8","_number":771191,"virtual_id_number":771191,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-13 08:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":1,"date":"2021-02-06 10:50:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":2,"date":"2021-02-08 11:16:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2021-01-27 14:03:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-02-01 09:41:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2021-02-09 16:43:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":1,"date":"2021-02-08 11:16:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 15:58:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-19 11:51:21.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2021-01-19 11:51:21.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-01-19 11:51:57.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2021-01-19 12:01:01.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-01-20 09:04:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-06 10:50:40.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2021-02-09 16:43:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"7e1897d5ce3fb606b567afd1b48a405e02c86037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-18 12:09:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5866d8c62847f7b7a9e5d26d70fb66298ce330b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-18 12:28:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3f67e058e25c611c3831a7b63f3fabafd92a96b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-18 12:40:25.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3bb40f2ff5955044dd696a641b78509e74e3ebeb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-18 12:41:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"66578cb406fc6f033f7d51b1571bdc2488e95cd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-18 14:02:07.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5808fcd9ae921b85a5e4c622b1a7023ca389deb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:58:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/83e6d928d91c415ab7516825c86c8253 : SUCCESS in 29m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/816680f61cbd4cb7b82ed7b60c401e15 : SUCCESS in 33m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c316e6ccd1544eec8ee6d17271556f6f : SUCCESS in 57m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/685bbfeb8865467383f8e77a304a3a2c : SUCCESS in 1h 03m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a8dffe09eb5e4a82944f94f2ff696238 : SUCCESS in 1h 15m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f2d87078d4d4426ca35d6c998b9d70e7 : SUCCESS in 29m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ed4faed71ecf4dc6955182ac6eb03b72 : SUCCESS in 18m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8fa62a909a094817a0e917b56960de57 : SUCCESS in 1h 03m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d733c303bfcd4828abc6e723b0c7fc87 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b83d0bcb1a87489386bc5ff3ce854e22 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1873d0eb5a64d87ae50d522b2bad925 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afe16b1f8e224be999b3e12b78996be2 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6efe1a4615834e4782c63a6f4b374e9a : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31e38514b7f04ab080d0817c8c0b96ca : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1ea491f61fe44c8bfb4e3f77e8e4ecb : SUCCESS in 21m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/42d178785efe4e2fa7c502411e00eefc : SUCCESS in 53m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d6cb8d6d71148b6b8f546d76b034561 : SUCCESS in 56m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85d14c7ebcfc46a8a128b70ecad696ca : SUCCESS in 51m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/89a3449a8adf44cf866486b4611f89d4 : SUCCESS in 1h 37m 42s","accounts_in_message":[],"_revision_number":4},{"id":"cf6430a1914f9ec0d01802c63e4de62c6c417943","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-18 17:25:14.000000000","message":"Patch Set 4:\n\nK-A change here: https://review.opendev.org/c/openstack/kolla-ansible/+/761872","accounts_in_message":[],"_revision_number":4},{"id":"745c5ff37fec66a2e6a60a298bdffba253f1c30f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:39:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/524de0f1eb6c4562b5b7c560bcaaead7 : SUCCESS in 3h 20m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a907eba6b1024e468f6228cb2010e654 : SUCCESS in 3h 10m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dd42c688f40a4e38bfc68a41122561c9 : TIMED_OUT in 3h 09m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/76f7d2418cc7457db0be10a96ba9a9a4 : SUCCESS in 3h 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ec04cde80283d6f95acb0644a1183d5f2ccb6b7a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-19 11:51:21.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70535abbb4c52aa36fb9104d0b2b622cbb5807ef","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-01-19 11:51:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"18586fab47da8a9ae078611a6de2d559e9650902","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-19 12:01:01.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"62112cac47729ab5f3e99b5702871586f8bcfa53","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-20 09:04:51.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"43c5f495b59e098403b13117b43ed8fecd1bc393","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-22 13:53:36.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"547d4ee19f9cd2bda31ac825b650f583feddeefd","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-22 13:57:04.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9744b463aafde57d16ef4e6e90e527f256e493b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 15:55:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/21c681d611044830b3655ee0852b1b71 : SUCCESS in 24m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/343c3f2405d64aa496549e1752280e33 : SUCCESS in 40m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4137a42bee52456380d4e8cffb1a1e7a : SUCCESS in 1h 05m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/707cf8d84f4f4d998d9f0c147ea93fbb : SUCCESS in 1h 09m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e71a4362b4a243a8a41e95682bb00f39 : SUCCESS in 1h 15m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f29e78a8777748ceb98cadaea7dc8744 : SUCCESS in 31m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dd21900cfe9a4cc89612c0aaea9be129 : SUCCESS in 20m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a6d7722b0e414127b139e7bca7adef09 : SUCCESS in 53m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77157fbde9d949c79276214c17116e09 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea313ea1617a41f5951d14e1a7ceec78 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/947ed4689f1f4fb390c6710d265624f9 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85a98becbdf84239bbe2998a87c7966c : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52fa46f240874523bcf7d507267ed1c6 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30b3e122074a4701bc84425c964d4c6e : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c83c79e35c0c4b5781d21937dbdcd997 : SUCCESS in 22m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/57e9ccc399f84e4683f13b793d68ceff : SUCCESS in 29m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6599f50442a941d090d8a342d4b7a12e : SUCCESS in 59m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1e51e27577e404898fc4682f04a2414 : SUCCESS in 53m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f86540d6ec174100b3626fc9342ba8b0 : SUCCESS in 1h 34m 24s","accounts_in_message":[],"_revision_number":5},{"id":"062c8e85c1108a50dd18f64e5250ab8f5a479982","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:00:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2911f9bcd9e307c50db8379851fe8c427bc5c2b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-26 12:22:34.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"a5f924a9c62805adbef43a077205d6e0ea785e29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:22:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d70e448fb10e477798a75d4c825d35c7 : SUCCESS in 1h 25m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/026d587b0e5a49e3b23bf7756a84b02b : SUCCESS in 1h 23m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/600675dea50d4046925c27e02f11226c : SUCCESS in 1h 47m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/26ee176efc334fcf98d192623829cb4a : SUCCESS in 1h 00m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"24187e45b4b1380bcaab74b504b5dce99daf322d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 15:25:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/38fcad8b066f4e6aad03ac3b71ceb95e : SUCCESS in 24m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1d5c410943de4aa184085732935aeec1 : SUCCESS in 41m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bf63cbfc02b24413910d427e2a751478 : SUCCESS in 1h 04m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/68670176599b48f8b74d74c22ff06f4a : SUCCESS in 2h 03m 10s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/756c7e2ce79b4a6c9b9a03552841e93d : SUCCESS in 1h 12m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/546fbe5d15df4be7811c9537f571450b : SUCCESS in 24m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4aac057d094a4ddfb32d318eccee4241 : SUCCESS in 26m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/640e511566584ddc9e80cbaa143d1142 : SUCCESS in 57m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e513e7e102e453d903d75c8c1cbf8b5 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ec467b2826347aca670996c11f1dcc6 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2109fc8e69464ad5944dde8dc918bfd6 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a67084d69eb647cabe11e5898131a687 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32dba1b099704a60964d3e089955e7ab : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74ef577698ef45d4858073615f8a7932 : SUCCESS in 7m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5550d7f7b36b4150a93646cc6ac5e549 : SUCCESS in 19m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/511f3cbd340c4faba99437d0ebe9230f : SUCCESS in 26m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef891b768ccd47b2a8b293ae89d4f0b7 : SUCCESS in 48m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07e52316678047f88d4354785024ea37 : SUCCESS in 55m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dd573b3689f84e5a93a1f3ad75c860bf : TIMED_OUT in 2h 51m 46s","accounts_in_message":[],"_revision_number":6},{"id":"e276593cf49b56352b1371a69cad9db198718178","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-26 17:09:40.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"55906adb4afcd7ee6d7f6e892ab9d31f5def2feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 18:57:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/47526f5995db420eb226c8e3b2dda5d5 : SUCCESS in 21m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4da4c9830cab485a83340924a8667e11 : SUCCESS in 41m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/44e313722f9c4441bf726e42be01dde6 : SUCCESS in 1h 10m 32s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/48845976e6c942469affb9c1c9e4c693 : SUCCESS in 1h 02m 10s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0edd3af6f02d4e388af5a19fc9af9588 : SUCCESS in 1h 21m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d7dd9f906e5f426f95ab02e01ed7faa4 : SUCCESS in 33m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4e97a837fe1c43fd8ccaf3d090c36b45 : SUCCESS in 17m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8f9a0a9a11b146329301448732bd262e : SUCCESS in 49m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6bbb04769c4ed087fe5b557d896123 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eda68ffc13eb421f9e78046507228ce2 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1973f67d8b444a49bcf9c8e20882d0ca : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c64cde63d2e461bbcf7a8aedc722849 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a105afba3804a018a7d0a2a1c5bc856 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78be2adf5e3643329c1a7ac98f862eb8 : SUCCESS in 5m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/52219dc03b624cca8c8ed0e809d527b1 : SUCCESS in 23m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac7f8b2f3ee6482eb12f6062d1abe7c8 : SUCCESS in 40m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4313e02b967f4f5494a8c8e5609cf7f5 : SUCCESS in 52m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/edfd96d3013349709c11b8db8ec4e530 : SUCCESS in 51m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/13942bac6e5b4e5b93ba5ee0f47fef33 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":6},{"id":"990d8258fcc7d30f80919b181d9a4128dfc95264","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 19:04:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/142fec8509bc464998820a3831d8ec0d : SUCCESS in 1h 19m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7d3999db5bb489e8d0a19bcf03fea2a : SUCCESS in 1h 15m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6f10a646296c4623a4b55c1ad10f228c : SUCCESS in 1h 37m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/932c2854717448fb9d359325bdad285b : SUCCESS in 1h 05m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1fad8d3f4a1ed1023ed6a0fc0298a3b6876efd0a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-27 09:45:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f6728700f440a3db37e858cce0538d8242853efc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-27 10:33:40.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"e43aab86ddd16a08befc8ac950ae8848f1f64692","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:22:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0017f8e4db1f4f56832868fe0039d2d8 : SUCCESS in 1h 08m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e89289bc21a149a09db5f75b49d5f134 : SUCCESS in 1h 00m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a854666e94f4b8490d45b067f569a85 : SUCCESS in 1h 47m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/36da028a3d5b46ceaaa662418ec63595 : SUCCESS in 1h 02m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"969faf2e14812603e34a104198b5594217f1ac7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 13:44:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/edab9ea1be4b41afa5e7cf929ca1499d : TIMED_OUT in 3h 05m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/127082fded954940b24ed2b093e5c063 : SUCCESS in 36m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/00a58bc1fb0a440c9bbbed82694cb2a9 : SUCCESS in 1h 10m 20s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3f475e5c50664e66a115901e10ced9f3 : SUCCESS in 1h 08m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fa112f40cfee4967babc245b8efc93e5 : SUCCESS in 1h 12m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/be7bd15e427d4a0392620d6201562518 : SUCCESS in 38m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/29e42f7aa7b94cfba8456803142b18f6 : SUCCESS in 18m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b859cef79c7a4e8da0b4218b4cce7ecd : SUCCESS in 50m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1a1e1639a044bc4889b663117779601 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebdcfd0e7b9343189354b626cbcfb51a : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b980e5b0a51d43ce867352c3f02d88c2 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12bf51c85abc4a8c850147d2d7939052 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea99a67bcd674e4f8fd6f4694cc61bfd : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ee37204e9524a88a420293111a563f5 : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ca803610da1481b9fbc1296c70e85c0 : SUCCESS in 21m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7c03e53eeeb4c949e9b3d1a058d403d : SUCCESS in 38m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6501eedfa43f4463b4efce64c2b645d9 : SUCCESS in 50m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1381ef6bea84f7e854be5507116f5dc : SUCCESS in 48m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a00e2b236586465699340c1f363cf5e5 : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":7},{"id":"07862cad97d6e175e7981954110f8c9f615eaa8f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-27 14:03:54.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"d3c4291a6ba7a9d7ce6d6c79796a3b55112abde2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-01 09:41:36.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"3aa3307fbbd423adb2ac35835a7bda2737bbf35f","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2021-02-06 10:50:40.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"2977c7078d17dc965aa9bf22f116e26692f2d11e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-08 11:16:21.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"539fc30ca3b32ecbf376e6c1278287fdeba5783c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 11:16:46.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f70f41a211cbe2d8526bee1946265eb84a6cee64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 14:23:31.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/61cc4c0fcd654b4aa11e288d682cd3a3 : SUCCESS in 33m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7a46f71aa0f346e39a5afeac0e031e65 : FAILURE in 4m 08s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c22404ad7ff74354bc5ebe28997e214b : FAILURE in 51m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e2851ff82d43d09cb45eb027796d88 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67989a9d5bf742798a0cd7e626b12234 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c963f8cecf484392b3efb3430936ff9c : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3679ec4ea9264773999307a5e53df3f5 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3eeb449f70af48e6ad3561c0b13e4159 : SUCCESS in 3m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1692fc68db164aa7bdef5a1cd144386b : SUCCESS in 29m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eaf404f16e9041d2ba7581c041272085 : SUCCESS in 53m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a9b789647de94b34b7fb78cb81468a6c : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":7},{"id":"5c74f7f539ff1f1e65c1eeaa1f5c9b7f30d01c66","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-09 16:43:04.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"64e257b75b1015a724b8a867a5ca1ca7b306c066","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-02-12 07:18:10.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"affbdfa7836335f71ddbf2a90c1ba81307bcfc22","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2021-02-13 05:05:50.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"fa3d400fbe2038398a1538f2bac787d35e29d0f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 06:38:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e1e7636fefca4598bd7a1c1b4358d4af : SUCCESS in 21m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0aecb4d8b41449588176f5a84089f19d : SUCCESS in 36m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0460594d50ac4fbfa20fbc75810ec69c : SUCCESS in 1h 05m 24s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8fe21fc5cde34cef87da0fb583edde87 : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/393428c56f6e4ee3aa6c276d3f79827b : SUCCESS in 1h 27m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/99de1d156b6a442bb615aaaa24382d77 : SUCCESS in 32m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/887e55c78c814ca1b20887a0b37228fd : SUCCESS in 22m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f11cf8d57be34f57aa6c93a557567e55 : FAILURE in 36m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bece8957cc9d42a3a799e11665b11cd5 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76aaca2851054850aef5339edca46fa1 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a74d42fa7674430a486917b04257138 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72098f5a8817438ea342d197fa0fc643 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151cfb05151e4f00aa1647a504d38580 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/573ec48260604c9c9689f8bbfa41e93e : SUCCESS in 3m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/10965ed231374de8845d26ed06555066 : SUCCESS in 21m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f90186fda88407dacb8a0e9316a75c3 : SUCCESS in 35m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf700f38cb1a4d3b901127dc98d5bc99 : SUCCESS in 52m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2022d7612ced453498604d52c3f17e9b : SUCCESS in 47m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b9efc44d25a4494a295083459787713 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":7},{"id":"0460873bd2317496a7d0db5d99cb1684062540df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 06:39:09.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2b716a0ed9ae67a15ffae10377a7670941e534c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 07:04:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/fdb00793752b4cf2ab97d4e04b55ff57 : SUCCESS in 1h 23m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0f1adfb317f4815b96c5ce41aea0dc2 : SUCCESS in 1h 27m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7dfd56aeb484a33bb047c71545745da : SUCCESS in 1h 57m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3efa0c0ae014676ad0f8acaa4efa8dd : SUCCESS in 1h 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2a5907341f4ac45672233b2e4bbad2e7f55781cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 08:00:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6eb83e93d854498dae2d0656b863ea08 : SUCCESS in 48m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/052180b55fdf4faaa151be5d87d3c3fc : SUCCESS in 52m 58s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5fd56ac9170c4cee871c1c384271e234 : SUCCESS in 1h 02m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a47222a2a664e1b8ed22882b8947f4c : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/579de0ea1de84ec2af120193afaf8135 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef81ccd05fcd4bbb951b0039e7d40afd : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a70853054fb422baa3fd484223bf603 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d114a9ba6e6d411bb6442832ae664437 : SUCCESS in 4m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6488742160e44cfa9a7fbef0e1c019e1 : SUCCESS in 48m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0b6bf014bd54969b953e6f286c179dd : SUCCESS in 48m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be897a6fa316476ab476b0c2e9780fd8 : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":7},{"id":"d4d79a5c75c3b2f0829e1be54904f3a3f5186d57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 08:00:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f5cf58d7791598da9fac202532109eb60df892d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 08:02:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080a8f618e174d67be1f32b5692c2cbb : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8c03db1b2a143cb9f0a221f78fed732 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"785fe5641de3e46b17e52d16e0c8a9e02b8a5633","revisions":{"97d3f0b7fece1b88cd494b17d915185143566b75":{"kind":"REWORK","_number":1,"created":"2021-01-18 12:09:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/771191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/771191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/771191/1"}}},"commit":{"parents":[{"commit":"62da72f5dbf93071b8997ccd5a93f4f0788721e0","subject":"Merge \"horizon: set FORCE_GENERATE to no if not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62da72f5dbf93071b8997ccd5a93f4f0788721e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:08:54.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:09:11.000000000","tz":60},"subject":"Add octavia-driver-agent","message":"Add octavia-driver-agent\n\nChange-Id: Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97d3f0b7fece1b88cd494b17d915185143566b75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97d3f0b7fece1b88cd494b17d915185143566b75"}]},"branch":"refs/heads/master"},"a36d490380149b04968ef9dcdceb37a5572c7676":{"kind":"REWORK","_number":2,"created":"2021-01-18 12:28:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/771191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/771191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/771191/2"}}},"commit":{"parents":[{"commit":"62da72f5dbf93071b8997ccd5a93f4f0788721e0","subject":"Merge \"horizon: set FORCE_GENERATE to no if not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62da72f5dbf93071b8997ccd5a93f4f0788721e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:08:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:28:35.000000000","tz":0},"subject":"Add octavia-driver-agent","message":"Add octavia-driver-agent\n\nChange-Id: Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a36d490380149b04968ef9dcdceb37a5572c7676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a36d490380149b04968ef9dcdceb37a5572c7676"}]},"branch":"refs/heads/master"},"83418b42659a579d388c7ff4dfeb47807f91be4b":{"kind":"REWORK","_number":3,"created":"2021-01-18 12:40:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/771191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/771191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/771191/3"}}},"commit":{"parents":[{"commit":"62da72f5dbf93071b8997ccd5a93f4f0788721e0","subject":"Merge \"horizon: set FORCE_GENERATE to no if not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62da72f5dbf93071b8997ccd5a93f4f0788721e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:08:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:40:21.000000000","tz":0},"subject":"Add octavia-driver-agent","message":"Add octavia-driver-agent\n\nChange-Id: Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83418b42659a579d388c7ff4dfeb47807f91be4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83418b42659a579d388c7ff4dfeb47807f91be4b"}]},"branch":"refs/heads/master"},"4660f988a0262019b823cdd65a00f200c3f60ff1":{"kind":"REWORK","_number":4,"created":"2021-01-18 14:02:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/771191/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/771191/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/771191/4"}}},"commit":{"parents":[{"commit":"62da72f5dbf93071b8997ccd5a93f4f0788721e0","subject":"Merge \"horizon: set FORCE_GENERATE to no if not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62da72f5dbf93071b8997ccd5a93f4f0788721e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:08:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 14:00:01.000000000","tz":0},"subject":"Add octavia-driver-agent","message":"Add octavia-driver-agent\n\nChange-Id: Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4660f988a0262019b823cdd65a00f200c3f60ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4660f988a0262019b823cdd65a00f200c3f60ff1"}]},"branch":"refs/heads/master"},"5f367ef42d9d68e42743f10d57c6b28ece982d22":{"kind":"REWORK","_number":5,"created":"2021-01-22 13:53:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/771191/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/771191/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/771191/5"}}},"commit":{"parents":[{"commit":"62da72f5dbf93071b8997ccd5a93f4f0788721e0","subject":"Merge \"horizon: set FORCE_GENERATE to no if not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62da72f5dbf93071b8997ccd5a93f4f0788721e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:08:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-20 10:05:29.000000000","tz":0},"subject":"Add octavia-driver-agent","message":"Add octavia-driver-agent\n\nChange-Id: Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f367ef42d9d68e42743f10d57c6b28ece982d22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f367ef42d9d68e42743f10d57c6b28ece982d22"}]},"branch":"refs/heads/master"},"3675c7147dc62f814fc4ed551adffd4c203b916a":{"kind":"REWORK","_number":6,"created":"2021-01-26 12:22:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/771191/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/771191/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/771191/6"}}},"commit":{"parents":[{"commit":"62da72f5dbf93071b8997ccd5a93f4f0788721e0","subject":"Merge \"horizon: set FORCE_GENERATE to no if not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62da72f5dbf93071b8997ccd5a93f4f0788721e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:08:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-26 12:22:25.000000000","tz":0},"subject":"Add octavia-driver-agent","message":"Add octavia-driver-agent\n\nChange-Id: Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3675c7147dc62f814fc4ed551adffd4c203b916a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3675c7147dc62f814fc4ed551adffd4c203b916a"}]},"branch":"refs/heads/master"},"785fe5641de3e46b17e52d16e0c8a9e02b8a5633":{"kind":"REWORK","_number":7,"created":"2021-01-27 10:33:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/771191/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/771191/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/771191/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/771191/7"}}},"commit":{"parents":[{"commit":"62da72f5dbf93071b8997ccd5a93f4f0788721e0","subject":"Merge \"horizon: set FORCE_GENERATE to no if not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62da72f5dbf93071b8997ccd5a93f4f0788721e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:08:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-27 10:33:36.000000000","tz":0},"subject":"Add octavia-driver-agent","message":"Add octavia-driver-agent\n\nChange-Id: Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/785fe5641de3e46b17e52d16e0c8a9e02b8a5633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/785fe5641de3e46b17e52d16e0c8a9e02b8a5633"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
