)]}'
{"id":"openstack%2Fkolla~771825","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I3abda8f972ce854d7fee22b1fb2fc43595668a8d","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I3abda8f972ce854d7fee22b1fb2fc43595668a8d","subject":"base: install setuptools to fix build on aarch64","status":"MERGED","created":"2021-01-21 15:00:36.000000000","updated":"2021-01-26 20:50:08.000000000","submitted":"2021-01-26 20:48:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771825","meta_rev_id":"c9216ab342f3b3ddece2a73b6826b21fe783854f","_number":771825,"virtual_id_number":771825,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 20:48:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-01-26 18:53:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-01-22 10:46:13.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-01-26 18:53:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 16:55:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 10:46:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-26 18:53:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"be7c9356b8b9a704bc0934d3d8d319db95c7cf3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-21 15:00:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1133db03fcd3a611fad589ed297081a8960830c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 16:55:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/319d13f886f548458b0f0e556a98f13a : SUCCESS in 27m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/81b4bee8f6a7485eae9d295d44387fea : SUCCESS in 1h 15m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7a3c414d519547e1b3471d922ebac77e : SUCCESS in 1h 51m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e37b0db9b08a4d4aaf6f7448f6a8095f : SUCCESS in 1h 03m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2db08548edab47f6a0b3dcf18b8a15ed : SUCCESS in 1h 06m 14s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/bd154af8b4bf46aeaa9536d609325c15 : SUCCESS in 16m 18s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1cbea0fe0a164fde9c99e0f48226da2c : SUCCESS in 20m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/64e5c6e1837d44f89e9b0f5be5b9f014 : SUCCESS in 18m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/13f0c4f6a0c64129a0aba40a6e14b782 : SUCCESS in 52m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c990c9b5e7422ba66fd3af2a6cb559 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2612f240b6e49c187dd38eda2c98ccf : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bd259f9a82f4d56bf08add0842cd101 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31b52e52bb94440a8314587ae390fb8 : SUCCESS in 6m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/63611fe88f0f4eed936790cae3740549 : SUCCESS in 23m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/55213a56afa54753a6654f64ee45dc72 : SUCCESS in 26m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a951191c83894c3f8149346845983b46 : SUCCESS in 58m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8af791def7d84255bb8879fad2747d4d : SUCCESS in 50m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/57f85ccd07bd472b8b04d2fe9be9b72f : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b996b2b3f8d2475af5076b5f2a06332132eced35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 18:22:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d5a06043bfb649009e07f8d3d4e4cdc1 : SUCCESS in 2h 38m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e19f36a9b92e46469e1537d5dc073d44 : FAILURE in 51m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/077b008f2ebd4fef81dccbf192995390 : FAILURE in 2h 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9916bd5b587bdd9fae8fc6f5ba898521feaa3353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-21 22:05:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c5ca7ddea1f2e416e1169437d83c565c66daba8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 22:55:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a699b76a0132c9e655163ab9a7386e07581ff31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 00:05:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/971da26f57334a3b9910733ebbe6fd0b : SUCCESS in 21m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/40e5f3a5fc1d4ab2b57bc9dc0fe8ff41 : SUCCESS in 38m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8058e423f20949c5921522a9ee4f17bd : SUCCESS in 1h 05m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5f3b013d74964dab8814dd0bb6007ba9 : SUCCESS in 1h 03m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5bec0e6b58a5405f97de2201b04e7d19 : SUCCESS in 1h 09m 43s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/a932d39c54ae47efbf032b0f64310b7c : SUCCESS in 13m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b29279de90e342d0ae57cfe87ad41e1a : SUCCESS in 20m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/237a9f2b94694c1c882599571fd31589 : SUCCESS in 22m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3c11890df6224ce08aa72c1cdc7c083d : SUCCESS in 1h 28m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3118d412c0442909cfe64ec3e3bd70b : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6957828dcfc043c382a63bd9a85a7680 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7bfe4ae26c984c66b374b30d5ba6f8ba : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bceec0b7130f4f6eb67374998ea78209 : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f2c8a063ceb04be58333cf026400a732 : SUCCESS in 23m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60d7f0cc37a64bb39ead3175e653d3e4 : SUCCESS in 35m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e7c0452834949449b463fd7e7265668 : SUCCESS in 53m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d0d2f661d7e4144818fa28386386e67 : SUCCESS in 1h 03m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/df48c1998de44ea1a6a988a267d3021d : SUCCESS in 1h 54m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d278204a18f3c5d5ba8887ba75738136b4198eab","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-22 10:38:53.000000000","message":"Patch Set 2:\n\nrecheck arm64","accounts_in_message":[],"_revision_number":2},{"id":"7e81280ff6f7611c8df5f70b927246f64e71dfb8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-22 10:46:13.000000000","message":"Patch Set 2: Code-Review+2\n\nI would also install wheel for sanity but whatever","accounts_in_message":[],"_revision_number":2},{"id":"022d0256d6197e68087e4a17a2b71c5aa44c8f6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 11:35:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92eefa99e02a28df34c38170dd590b3376b25a79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 12:44:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a50181c50b20475ab08897aa514d0bee : SUCCESS in 28m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0c3566d4628f4cb0b67b769cfd465dbc : SUCCESS in 30m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/23b5c609f1a64e63a96dbcdeafa2b8d0 : SUCCESS in 1h 10m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f0b0d4a5ae62419ab9bcdfa5459957db : SUCCESS in 1h 01m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5698419f613b48ee9d5f4aae61a70c1c : SUCCESS in 1h 10m 55s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7ef56581999340759a583337c6faf02d : SUCCESS in 18m 49s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/16611233fb63403790d18e31b091a654 : SUCCESS in 31m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3797965ef2774f6ebef51a816858d65e : SUCCESS in 21m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/896c672ef8f1427bbde55f8e97c2b11e : SUCCESS in 1h 02m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7263574877cb44559ba5265f12da7a8d : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e6aefd664174b7ead58135faf59e5fe : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efc69e0ba32d409e9fa731d5a6069035 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f782574ee2c742f08ea0dded0d8529ca : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/62206618d6624a7788abacad59a2dff0 : SUCCESS in 22m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32658bf2c4f5422e86e8cdf2ee5ae266 : SUCCESS in 24m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a3d2e21819a4fb88b97d1a0b00685a2 : SUCCESS in 57m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4edfc63535814f4ba6c2d035003c0f33 : FAILURE in 6m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/46875f95ff0d421fb8885987cd1a0d9a : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ff7c93f1bd414dad32c960bd6fc0470d419220b0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-26 18:53:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"93820e0c6d056852c3465bddcbd48e9bd47bbd5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 19:02:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e230f27846976f09aeb33fe10ce6c310c3f5cd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:48:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/253b0070b477484a84f6b9a04f76b981 : SUCCESS in 28m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bce204fa647f46fd94c44fdf460198fb : SUCCESS in 1h 25m 02s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1e8612f2c9a844c9a7b8977067580bab : SUCCESS in 1h 38m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/213d9324548a45d9ae2126a57f7f9f4a : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7939b89cb07b4f0ea2056b7064ece267 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a92010f0bce845ca9d409489a84c9d9c : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6f3a6887754156a57bb1361f034b60 : SUCCESS in 7m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc04fbf505914eb2987f700a11690b7f : SUCCESS in 45m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ecc98d2ec8ad481aa709113ac6f03201 : SUCCESS in 1h 19m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/88b17192c8a647f8bf80d32da724cd5b : SUCCESS in 1h 26m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e8238b08aa6989e50ab04bf00b663e3ed3c24db9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:48:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c9216ab342f3b3ddece2a73b6826b21fe783854f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:50:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11309620e7c64829bc72841864d4cbe2 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57d05bc7e0c7ca7f045afa38f23fc74f3844d289","revisions":{"5b8566d4eb729a51046848570a209acade71ae12":{"kind":"REWORK","_number":1,"created":"2021-01-21 15:00:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/25/771825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/771825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/771825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/771825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/771825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/771825/1"}}},"commit":{"parents":[{"commit":"213452b4ba6f015a863e7bbcf28ae9f44094e2c7","subject":"Merge \"CI: Use Docker proxy for pulls\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/213452b4ba6f015a863e7bbcf28ae9f44094e2c7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-21 15:00:13.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-21 15:00:13.000000000","tz":60},"subject":"base: install setuptools to fix build on aarch64","message":"base: install setuptools to fix build on aarch64\n\nVirtualenv used in ussuri has problem installing packages on aarch64:\n\nINFO:kolla.common.utils.openstack-base:    Complete output from command python setup.py egg_info:\nINFO:kolla.common.utils.openstack-base:    Traceback (most recent call last):\nINFO:kolla.common.utils.openstack-base:      File \"\u003cstring\u003e\", line 1, in \u003cmodule\u003e\nINFO:kolla.common.utils.openstack-base:    ModuleNotFoundError: No module named \u0027setuptools\u0027\n\nVictoria/Wallaby do not have that issue.\n\n\u0027python3-pip\u0027 package in CentOS 8 and Ubuntu \u0027focal\u0027 (victoria+) depend on\n\u0027python3-setuptools\u0027 package.\n\nChange-Id: I3abda8f972ce854d7fee22b1fb2fc43595668a8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b8566d4eb729a51046848570a209acade71ae12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b8566d4eb729a51046848570a209acade71ae12"}]},"branch":"refs/heads/stable/ussuri"},"57d05bc7e0c7ca7f045afa38f23fc74f3844d289":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-21 22:05:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/25/771825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/771825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/771825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/771825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/771825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/771825/2"}}},"commit":{"parents":[{"commit":"a6e528dfbbae13e91099af08b645fca65c03be34","subject":"nova-compute: \u0027pmdk-tools\u0027 is x86-64 only in Ubuntu \u0027bionic\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6e528dfbbae13e91099af08b645fca65c03be34"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-21 15:00:13.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-21 22:05:13.000000000","tz":0},"subject":"base: install setuptools to fix build on aarch64","message":"base: install setuptools to fix build on aarch64\n\nVirtualenv used in ussuri has problem installing packages on aarch64:\n\nINFO:kolla.common.utils.openstack-base:    Complete output from command python setup.py egg_info:\nINFO:kolla.common.utils.openstack-base:    Traceback (most recent call last):\nINFO:kolla.common.utils.openstack-base:      File \"\u003cstring\u003e\", line 1, in \u003cmodule\u003e\nINFO:kolla.common.utils.openstack-base:    ModuleNotFoundError: No module named \u0027setuptools\u0027\n\nVictoria/Wallaby do not have that issue.\n\n\u0027python3-pip\u0027 package in CentOS 8 and Ubuntu \u0027focal\u0027 (victoria+) depend on\n\u0027python3-setuptools\u0027 package.\n\nChange-Id: I3abda8f972ce854d7fee22b1fb2fc43595668a8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57d05bc7e0c7ca7f045afa38f23fc74f3844d289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57d05bc7e0c7ca7f045afa38f23fc74f3844d289"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
