)]}'
{"id":"openstack%2Fkolla~771846","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I2076ab4a634418df5863174588a72e6c2ce8f05e","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I2076ab4a634418df5863174588a72e6c2ce8f05e","subject":"nova-compute: \u0027pmdk-tools\u0027 is x86-64 only in Ubuntu \u0027bionic\u0027","status":"MERGED","created":"2021-01-21 16:34:58.000000000","updated":"2021-01-26 10:53:48.000000000","submitted":"2021-01-26 10:51:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771846","meta_rev_id":"96afd7d008f61070d952206334ef198b88ca700d","_number":771846,"virtual_id_number":771846,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 10:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-26 09:21:53.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-22 10:43:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-01-26 09:21:53.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 18:08:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 10:43:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-26 09:21:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"9d5fd473617dac3cf8cffa671e8f7240182880a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-21 16:34:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c397db27e8a1dfe4317608516a166c726f53ab43","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-21 16:39:01.000000000","message":"Patch Set 1:\n\nMaybe backport to Train will be needed.","accounts_in_message":[],"_revision_number":1},{"id":"4dda1ee511ef09ecf0fca0653a77db4e1dea7dac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 18:08:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5a3f6201abc04659af90d3d7c22b7c71 : SUCCESS in 30m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d00069be95884cb7b29704839e978d18 : SUCCESS in 1h 00m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3d79ade6209a4f9c93ce426121f78bf9 : SUCCESS in 1h 27m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3817619688bf4b679e2b5d2a8e9f44d2 : SUCCESS in 1h 21m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a00398c56c8745648e01c1c7a41c9c8a : FAILURE in 9m 39s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/136a4b90e452424db5e39fbd750b2bf7 : SUCCESS in 30m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bac6265fdc384880a7e0aa649c084ab3 : SUCCESS in 1h 01m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/aa0190e97527451f91a7ea1d8ceefea9 : SUCCESS in 57m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c25e49302ac641a38357fd69121438b7 : SUCCESS in 1h 00m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ea97bba562048359aa18a94b115de16 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5263b1508ff4a68b520426a27d93631 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/203c93f0c4fd44a6927023e069530044 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41dd26604d42449694045e54421b9949 : SUCCESS in 7m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/74c511d4e1e94d4ba68a9c04e4ae49c5 : SUCCESS in 22m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b11c65db12024a4ab304c12823863e92 : SUCCESS in 34m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8d9894b0d94477aac00e1e102a78552 : SUCCESS in 1h 19m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/97c82fa41ce54b6c96cbf5b54247d78f : SUCCESS in 50m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f185527d5cba4397a9713142c43ffdd2 : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a1957a05a1fc234da79f93aec153ff18649b4c91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 19:45:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/33d1156a46ac4e49b9e0a2ec324dabf7 : SUCCESS in 1h 38m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f98e10c55ca6dab5dd9d99b84733fc5f2ee69d0c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-21 22:05:37.000000000","message":"Patch Set 1:\n\nrecheck arm64","accounts_in_message":[],"_revision_number":1},{"id":"f1387c2f19a6debd2d62aea8a267241d5cc7e114","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-21 22:07:12.000000000","message":"Patch Set 1:\n\nI hope that the \u0027setuptools\u0027 change and this one would not be required to be squashed into one.","accounts_in_message":[],"_revision_number":1},{"id":"d01e484f63489a808b14c2526f272bcba266366c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 23:00:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"126664a472f9e096091e0d13af1c73e34c90b361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 23:50:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0cb685e1a74d44c59321c7207fff0e30 : SUCCESS in 21m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4489e796877e4b36b5c61678ad78481b : SUCCESS in 30m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c76ca76dbaf946618fd552b691a6f68f : SUCCESS in 1h 06m 36s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5ece1bf89b374e208dd65dc132558238 : SUCCESS in 1h 00m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/45369cc812b440e4b34c7f163ceca842 : SUCCESS in 1h 06m 17s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e1d00a58479c463e845ea6fe06a44a17 : SUCCESS in 18m 05s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/46ab0062bace41a6ad92514729d1c31b : SUCCESS in 29m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/001cfa0e1bb942fbabab6522b60d1ad1 : SUCCESS in 30m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/29b721237971438abad7fdecb050ff33 : SUCCESS in 57m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5971cd32f7407b905734a6ac1e840f : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e56a1d2c8d3e4ae180efcdc7737cd85d : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed3e677fc0584d02bc1545111873f70f : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb3b1889ef4419f88a658b062272c5e : SUCCESS in 6m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2d5b9e078c414eab9f87473d16a649ec : SUCCESS in 1h 01m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3dde28af5d3f46099720bd12aab66a1b : SUCCESS in 36m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/992500900909414ca569d2cd8ff536e5 : SUCCESS in 1h 02m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/97b82158b319437b8b445f9ca64be80a : SUCCESS in 1h 14m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/188942681ce94fc5a0490e7c091af029 : SUCCESS in 1h 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9bbc90320a4bec2c0245934800df3671dc821210","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-22 09:09:50.000000000","message":"Patch Set 1:\n\nrecheck arm64","accounts_in_message":[],"_revision_number":1},{"id":"49359a55a30e0002fc8f3cd1beb3bf37d519f841","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 10:24:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9fb6f6eefdd6450c836a552083ec6970 : SUCCESS in 1h 11m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/01a470571c6246979aa6fbda6694f192 : FAILURE in 1h 03m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a41178a5eded4dc2b0cba818f081a9a8 : FAILURE in 46m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/26ec8a3fffc246b0b0c5a28a986655cd : SUCCESS in 1h 08m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"898a8978ae91e29390a3da5326f158f89c483927","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-22 10:43:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4fce6cbc3369bf2725c0ee1b9fe7684dd4250d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 12:30:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0f7f981a93ef469cadd80eb869754ae7 : SUCCESS in 18m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b587b528752f4adc8f2dbfed9a782861 : SUCCESS in 27m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8d44f8ce0fbb418eabe769b50ee36090 : SUCCESS in 1h 00m 18s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e43580e6b2e246d2b85e714c3f0b43e6 : SUCCESS in 54m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d68d07b0d9a840c2a19312d914c26c48 : SUCCESS in 1h 08m 56s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7aa446ffbc384afca995c8cfa5e36a05 : POST_FAILURE in 21m 15s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c14723a9aa404ffe983c22dd2d789108 : SUCCESS in 25m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/774031417768485a9fffa40aa2c15072 : SUCCESS in 21m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d8be6ab8e0724267bf238f565c144ca8 : SUCCESS in 58m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a852d8ad014c15903fc95e20a0cd05 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/985e59fea16b43888616a0fd454ffa68 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c9c002d96284e72a2aa4bda12711de9 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d922416d876a4649ab56d42c464914a0 : SUCCESS in 7m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afcf0e68608f464e887eda9dd8b993db : SUCCESS in 23m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f9ed8b70122418d893e14aac320551a : SUCCESS in 29m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/20890de445ff4ba7b9afa1bff024ea03 : SUCCESS in 59m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/524dcc98e0224a3a8f170941158b1a6b : SUCCESS in 53m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1ab73c038eca4bf49fcbd518cf4be702 : SUCCESS in 2h 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c9d062a2457f50cf9e9656a7f6cb33baa3ef1829","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-26 09:21:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"944690df4db58826319aef7d3352cdec4f4a0087","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 09:22:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4eb081eac7ec2da49140a6ec144e41d9dd2625f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:51:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cedfc5191d4d43328fa9d8ad2f5fe5b9 : SUCCESS in 25m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/22c6a79461124994a40ad6e31327e9f7 : SUCCESS in 59m 53s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bc5f58f6dde74912abbe4b271bac10fd : SUCCESS in 1h 03m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/069cbbfd9e654290ac52a8eb93cd9fbe : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8177fd0490a7458a9b7b2f289ff4387f : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd1cd6ec8e40417e8048610a310005d8 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd548a95b6144adb2e7d360aec6911e : SUCCESS in 6m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/43e197672a4243bd95e66073d6934801 : SUCCESS in 26m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/81f5e89db3e4424fbc7ada9393486a8b : SUCCESS in 56m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f1633c42fddf401aabb6c26a0d834b90 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3196d117fcbb4295824ea40bb3cefa90e5c2e0fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:51:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"96afd7d008f61070d952206334ef198b88ca700d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:53:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1190bdb070ce4650a54aa93f25218c1b : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6e528dfbbae13e91099af08b645fca65c03be34","revisions":{"a6e528dfbbae13e91099af08b645fca65c03be34":{"kind":"REWORK","_number":1,"created":"2021-01-21 16:34:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/46/771846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/771846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/771846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/771846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/771846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/771846/1"}}},"commit":{"parents":[{"commit":"213452b4ba6f015a863e7bbcf28ae9f44094e2c7","subject":"Merge \"CI: Use Docker proxy for pulls\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/213452b4ba6f015a863e7bbcf28ae9f44094e2c7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-21 16:33:50.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-21 16:33:50.000000000","tz":60},"subject":"nova-compute: \u0027pmdk-tools\u0027 is x86-64 only in Ubuntu \u0027bionic\u0027","message":"nova-compute: \u0027pmdk-tools\u0027 is x86-64 only in Ubuntu \u0027bionic\u0027\n\nVictoria+ branches use Ubuntu \u0027focal\u0027 where this package is available\nfor aarch64 as well.\n\nChange-Id: I2076ab4a634418df5863174588a72e6c2ce8f05e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6e528dfbbae13e91099af08b645fca65c03be34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6e528dfbbae13e91099af08b645fca65c03be34"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
