)]}'
{"id":"openstack%2Fkolla~771994","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ia5234c7355669a4aa9bb1c45660930aee553e220","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Ia5234c7355669a4aa9bb1c45660930aee553e220","subject":"nova-compute: provide needed deps for debian/aarch64","status":"MERGED","created":"2021-01-22 15:23:36.000000000","updated":"2021-01-26 20:35:44.000000000","submitted":"2021-01-26 20:35:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771994","meta_rev_id":"cef380fb95859212e86c9a3486aadd4f486bc6f6","_number":771994,"virtual_id_number":771994,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 20:35:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-01-26 18:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-01-26 17:37:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-01-26 18:51:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 17:06:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 17:37:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-26 18:51:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9d9c2b43499731deb7a049f7f0ac847cafa681da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-22 15:23:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45794796cdbc656f977c0fbcc233df2745d28708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:06:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/50c0c8f4c7bb4f628eb149669e284a2b : FAILURE in 36m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a2716564be064cd4a843df2d83355eb8 : FAILURE in 48m 01s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9ee6692241a849c0a3c3c740e63676ff : SUCCESS in 26m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d156956ddd9f4e40b983b093fd820431 : SUCCESS in 31m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7f5bbdafa45d4180854f55dadc5c8f83 : FAILURE in 6m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1c49a4ca27674472a7d65a0234750b9b : SUCCESS in 54m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/30cfac9d15eb4293b578eb379f75aa05 : FAILURE in 8m 20s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1d655ba8e362425bba0c21f79ffc4232 : SUCCESS in 1h 02m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1e4b47f30c4246dd8730f9eaae477af1 : FAILURE in 5m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6ddf1d38f15b4d8fa1dc48aeff920684 : SUCCESS in 50m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/78ec402be77442e78e3a414cfb8420f2 : SUCCESS in 23m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/60d6c0966b2f41ccbdef7cccca169389 : SUCCESS in 24m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b1b5c83022547dc810e11402a562236 : FAILURE in 5m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096fc9207f5b414a9bcd1a43bc96a94f : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1234d1a5448a4cd2b30105ddae0ee608 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2094b47c4af84de18cb6df536284e3ad : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d04e4c1c90484c14832dea2ea99a7c0e : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad8f377cce5415383dff275cb369e25 : SUCCESS in 6m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87f0654269ba48968982aea2f794665e : SUCCESS in 21m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5205db09979b412e91e2e1ac61ac0bb6 : SUCCESS in 26m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f435a5041c249ec92fd7cb236595229 : SUCCESS in 42m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0061fc37c9ea4ab9be1cfeff1b1efbf0 : SUCCESS in 46m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e8ecd1458c9407eb19c21767562b6b5 : SUCCESS in 1h 10m 35s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"63c5582683b17c382eeed66d8542c424074e37a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-26 12:00:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d619e6eb0a1bc8e8e964be583bb9dd2170e59ab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:17:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/62e7863e7ce4430493618c63d5bc0210 : SUCCESS in 36m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/39059741f5c94c699970c2a5e6176df5 : SUCCESS in 37m 07s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7ad52822dcaa462893b0d5e262207394 : SUCCESS in 48m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6194a47754eb4117b36f494919bfd728 : SUCCESS in 27m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/93f3babc1d534b6cae15d7f40e158192 : SUCCESS in 1h 03m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f0dc5fd8b9e64f91a15fd251019d66ea : SUCCESS in 59m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/523dca0c567648279bb4f4a6db3cac77 : SUCCESS in 1h 25m 55s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9f4134f8c6484bf0b2e8c7c1e265e5b6 : SUCCESS in 1h 20m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/58edd33a394947709f5d8613607fbfdb : SUCCESS in 1h 35m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7193f06c8376469d9c183f116b1bcd19 : TIMED_OUT in 2h 08m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fc2d413f995e40a998fc445f9373bda9 : SUCCESS in 25m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e09401bb799749a6808ea5d58f29f169 : SUCCESS in 17m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ea5eb2e53be4e018cf089f1cb7124f7 : SUCCESS in 59m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87dbf19b4c0f48e7b270f2810847d80d : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a8e9073bde04e62a970b24d0ca946c8 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/079d403f688447ab98c463aa907f9b43 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d314d393459440ca8c3c32170432ca59 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb69e26a7302461abec0b97bef091e96 : SUCCESS in 6m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/615efecb98cd45a9959634da6b5e6aa7 : SUCCESS in 22m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/178f399b647b476ab2b60954bbc3f3c6 : SUCCESS in 25m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a64fd3031244c0f8849ee10689ac77c : SUCCESS in 1h 03m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fb679dc1b854487da29eac08d8dc14e0 : SUCCESS in 46m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de5f815a6c1948329aa35535ac458854 : SUCCESS in 1h 40m 33s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"09e7517648ae6a709c6c1299469e532d24cb41b2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-26 14:44:43.000000000","message":"Patch Set 2:\n\nrecheck\n\ntripleo...","accounts_in_message":[],"_revision_number":2},{"id":"41ae7248ab10159f13885f418b2bdeeda4b386d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 17:15:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/326b65809747481fad4bb3e334d82792 : SUCCESS in 37m 46s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4386a7aa0fd34f8e9004f611618b9996 : SUCCESS in 47m 17s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7f98484f00f74b10a328befd69669600 : SUCCESS in 25m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ebd7f91c0751422ca43b8f67cbf78141 : SUCCESS in 30m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8495e3feab854363a63c41b1cd9df8f2 : SUCCESS in 1h 08m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7a487b9ba6264540980f7faacd63a98a : SUCCESS in 57m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6790c8b8e0244fb78bdfa868ecd3ab40 : SUCCESS in 1h 14m 54s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/87bb9e1f713c41269358989375f4d3c7 : SUCCESS in 59m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d38d8836479f4f25aa4bb33f8f764603 : SUCCESS in 1h 46m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/480be0deb1c446d9b88199ef411f157d : SUCCESS in 39m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/943b8e4fdf4447fb82bc53ef6f7d475d : SUCCESS in 19m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8ad5dbf9c7d24023a9ac7b0eb1d1e52c : SUCCESS in 19m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d2077579c631456590b334be29ecad40 : SUCCESS in 47m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0705c0731ff242299cef0fc9d2c74adb : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44e2c053261543668bf6467912078689 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e81823482d6f45c8a4887926e609fd0b : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/550052a935694d20b1ea70cd928c0b3f : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c34a71e3017443f288fc8a3e5ea80cd5 : SUCCESS in 5m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43728d3ba8694c5c90a1df0196f9f512 : SUCCESS in 31m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/806200e903264558a6e821a610b89dd3 : SUCCESS in 2h 20m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e505a79d6e54ee0b3e640694b4f00b1 : SUCCESS in 48m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ce1e67b13f3463385b2da993d41d08b : SUCCESS in 51m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/875b52e42b38454dbb7f8e877a980db3 : SUCCESS in 58m 18s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"e0e01a0af27331d4639f25a19249903d1f2dbb6f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-26 17:37:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"02bd55a60a293b12eae4d90814dd6efb4a69fdc1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-26 18:51:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7c3173f988a559a339c5c14e932e7ed920e212b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 19:02:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"211829262b03ad68ee278b4b7e43fa706ee6e6bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:35:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/31d516730faa4794abe11ce76c477736 : SUCCESS in 48m 26s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ee857bb2185543e081a7d2975ac2aafa : SUCCESS in 53m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/68bfcaeb49de4185ae69384126e8f024 : SUCCESS in 1h 10m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bae563367c114891a2372292fdac9243 : SUCCESS in 1h 00m 55s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/605c8db9575343d1b03465fe947484df : SUCCESS in 1h 26m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3e6d01ea079f4885b5ce92752a672464 : SUCCESS in 48m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce076a9354f1403fb67041b1e0d69a3d : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1532529028f44f7b945ce05b80b1687 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd709c2ee9614232ae4710b1438a850a : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2778368bc52b4c4fa11f8762010656c8 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64604edbb55940f3be322012eed6b85b : SUCCESS in 6m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b1145b76f762484ea4abbb4dd848d21d : SUCCESS in 28m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cfa8414bafb7416db8788c71c7d41053 : SUCCESS in 1h 08m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6c32240ff0d741a0955cdf0269ad2cac : SUCCESS in 1h 00m 09s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"cef380fb95859212e86c9a3486aadd4f486bc6f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:35:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a9e8cb106cf8d05c7f52d86fdf97dcc780f4a2b2","revisions":{"e150b3fa2958946473c770bb6cd897e45586734d":{"kind":"REWORK","_number":1,"created":"2021-01-22 15:23:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/771994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/771994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/771994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/771994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/771994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/771994/1"}}},"commit":{"parents":[{"commit":"e457fe15c43dfd641b644e16b3c90fb366dccc8a","subject":"docs: drop Fedora info, mark Ubuntu one as Debian / Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e457fe15c43dfd641b644e16b3c90fb366dccc8a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-22 15:20:57.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-22 15:20:57.000000000","tz":60},"subject":"nova-compute: provide needed deps for debian/aarch64","message":"nova-compute: provide needed deps for debian/aarch64\n\n\u0027daxio\u0027 needs \u0027pmdk-tools\u0027 which needs \u0027buster-backports\u0027 on aarch64.\n\nTrain only.\n\n- backports repo was enabled in ussuri with move to ceph nautilus\n- \u0027pmdk-tools\u0027 we pinned with move to QEMU 5 in ussuri\n\nChange-Id: Ia5234c7355669a4aa9bb1c45660930aee553e220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e150b3fa2958946473c770bb6cd897e45586734d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e150b3fa2958946473c770bb6cd897e45586734d"}]},"branch":"refs/heads/stable/train"},"a9e8cb106cf8d05c7f52d86fdf97dcc780f4a2b2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-26 12:00:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/771994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/771994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/771994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/771994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/771994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/771994/2"}}},"commit":{"parents":[{"commit":"f6c4fc69d93754843709f045d0bf9ebf6c416cba","subject":"Merge \"Improve pip install process for offline deployment.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6c4fc69d93754843709f045d0bf9ebf6c416cba"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-22 15:20:57.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-26 12:00:40.000000000","tz":0},"subject":"nova-compute: provide needed deps for debian/aarch64","message":"nova-compute: provide needed deps for debian/aarch64\n\n\u0027daxio\u0027 needs \u0027pmdk-tools\u0027 which needs \u0027buster-backports\u0027 on aarch64.\n\nTrain only.\n\n- backports repo was enabled in ussuri with move to ceph nautilus\n- \u0027pmdk-tools\u0027 we pinned with move to QEMU 5 in ussuri\n\nChange-Id: Ia5234c7355669a4aa9bb1c45660930aee553e220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9e8cb106cf8d05c7f52d86fdf97dcc780f4a2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9e8cb106cf8d05c7f52d86fdf97dcc780f4a2b2"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
