)]}'
{"id":"openstack%2Fkolla~772206","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I7f85f17e7b9a8a6d7bc656abbc39ba3dd2deda3a","project":"openstack/kolla","branch":"stable/ussuri","topic":"revert-virtualenv-workaround","hashtags":[],"change_id":"I7f85f17e7b9a8a6d7bc656abbc39ba3dd2deda3a","subject":"Revert virtualenv workaround","status":"MERGED","created":"2021-01-24 16:11:04.000000000","updated":"2021-01-25 14:45:39.000000000","submitted":"2021-01-25 14:36:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772206-revert-virtualenv-workaround","meta_rev_id":"03ccfcbc99bc26bd1670595579c4e4814a362e09","_number":772206,"virtual_id_number":772206,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-25 14:36:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-25 10:44:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-25 11:25:18.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-25 11:25:18.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-24 17:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-24 17:24:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-25 10:44:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-25 11:25:18.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"e709ccdb0baf84a8d64d9b58718f4b3f71866c45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-24 16:11:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ed2a5d8cdc2fef2f375efe507797506ca732f1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 17:03:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9dd1a91e6d394978be1533baf3fd9b06 : FAILURE in 44m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa6348862fe5c30196bb2c557993d98f4894f4c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 17:24:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/534732129937461399ae72fa20b553bc : SUCCESS in 23m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5afc4ab0c6234dbc8e0a2f51b39fb1f3 : SUCCESS in 27m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5c7073871ba74f9a83dbcd0f2fb44943 : SUCCESS in 56m 35s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0465da0279054d8aa4fcf924f7d2d9a6 : SUCCESS in 57m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/952bb0ca622b49d2950e300b10485fb5 : SUCCESS in 1h 07m 18s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c4192b6fabb74ba1b5fe2f8461518dbb : SUCCESS in 18m 15s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5e25354228334789adb0323e6f7d74f2 : SUCCESS in 22m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9797bec1604c4afe84b8baa4ce9a55c8 : SUCCESS in 26m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/90450dd52aee427286db778bfcea3282 : SUCCESS in 54m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ccf50889c34f779f51302a00bd5a14 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eef88f886f54d4c917ee9c056516836 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c71235c7c2c42ad8be68616e89484a3 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81899b001c143f29bd451ea2d90bd66 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b51d6ac72914d02961bc16e7bd56e29 : SUCCESS in 23m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/116787d534694e4bb347f19e2c6a1c8f : SUCCESS in 27m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/62c0202436e746739ca810344cd8b66c : SUCCESS in 55m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f9b69147a67241cdb3f23ed709714430 : SUCCESS in 52m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da62f9d94fa448008501026e9f3f20b8 : SUCCESS in 1h 03m 07s","accounts_in_message":[],"_revision_number":1},{"id":"976fdbfc2f8d8778ee8e29b33a79c85c66be7f06","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-25 10:44:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e434a20acd78a1c6bb224ae5b2b86350730333c3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-01-25 11:25:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac9f1ba5255efa16496bc26a0af5985b4c9a879e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 11:25:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e3a7d2b80870c8d944775af9761d08d431eef8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 14:36:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ee6f92d623a64060a27ad90d19256426 : SUCCESS in 26m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a4dd09cd5c9a441587c91fb24ccbc505 : SUCCESS in 1h 01m 58s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d74f98c9da7e4b2f92aaa105dd0c86f6 : SUCCESS in 1h 05m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d1e0632557c458bb8c2f9ffbb238c5c : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/310212e0a95d42a6b4dc95ef5f2f3a31 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b46c97e7170541959f92b6009c9ab71b : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25406da326154ccfb2ab4a20431c75ff : SUCCESS in 6m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b92fa23567694a67b2600a831a28b2ba : SUCCESS in 25m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1137c11c0394d028e3713fec5413fe9 : SUCCESS in 55m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b3027d55d8e4348a5352ce1b03d54d8 : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":1},{"id":"629740505bfa306df764446d50263519c3d09093","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 14:36:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"03ccfcbc99bc26bd1670595579c4e4814a362e09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 14:45:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6836d6122bf94db99e8dafbfe2a113fa : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d555005ba01c08fe85a55ef779f6a1e7832edfa6","revisions":{"d555005ba01c08fe85a55ef779f6a1e7832edfa6":{"kind":"REWORK","_number":1,"created":"2021-01-24 16:11:04.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/06/772206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/772206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/772206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/772206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/772206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/772206/1"}}},"commit":{"parents":[{"commit":"213452b4ba6f015a863e7bbcf28ae9f44094e2c7","subject":"Merge \"CI: Use Docker proxy for pulls\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/213452b4ba6f015a863e7bbcf28ae9f44094e2c7"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-24 16:10:26.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-24 16:10:26.000000000","tz":60},"subject":"Revert virtualenv workaround","message":"Revert virtualenv workaround\n\nIt should no longer be necessary.\n\nUssuri only.\n\nChange-Id: I7f85f17e7b9a8a6d7bc656abbc39ba3dd2deda3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d555005ba01c08fe85a55ef779f6a1e7832edfa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d555005ba01c08fe85a55ef779f6a1e7832edfa6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
