)]}'
{"id":"openstack%2Fkolla~772396","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I7ad2373327be29cd307db5ce4f7db022829e3b35","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I7ad2373327be29cd307db5ce4f7db022829e3b35","subject":"Bump up rally and rally-openstack","status":"MERGED","created":"2021-01-25 16:49:44.000000000","updated":"2021-01-26 20:13:27.000000000","submitted":"2021-01-26 20:11:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772396","cherry_pick_of_change":769909,"cherry_pick_of_patch_set":1,"meta_rev_id":"3a10dd896d7e928b5cc627d0f2d2f16b666adf36","_number":772396,"virtual_id_number":772396,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 20:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-26 09:43:39.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-26 18:34:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-01-26 18:34:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 17:47:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-25 19:56:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 09:43:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-26 18:34:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"ec93d7fd128b334f45f254d071756cc1933f576e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-25 16:49:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7300365a43f5c9ed8a4cc5deda3b5670ba1dd1b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 17:47:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15c11326c58351ecf0145297c2a28dd8e5498a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 19:56:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b77d37fe785f429588dddb8b5fb96515 : SUCCESS in 34m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e2c67b949aa94d80a55aa5142c145099 : SUCCESS in 30m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d827c79d1c7c496aa6ff3197917ff2f6 : SUCCESS in 1h 02m 12s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/dcb6b972b76a457593cf0fb441b86e29 : SUCCESS in 1h 01m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c83e101f3f7a4ecd9e171d939c1d57d9 : SUCCESS in 1h 24m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0d0d39857f5e490fa487d210ba94c42e : SUCCESS in 23m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9eb9038760364b408671b620f3ba89ee : SUCCESS in 33m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f4fdb74e359c4898a977f6043c3adf86 : SUCCESS in 1h 00m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb63e1af6ba4d5ab041d1bd19812a1f : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/521b6387be4d4e2b8b3895438ea497e5 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfe4c91eb99d4b3187cecd6b5b2f10e7 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667bb9cbd7f44802968b062aa3da5b40 : SUCCESS in 8m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fbe0ae8782c5443b8c3ec7b9a098e5fc : TIMED_OUT in 3h 04m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d979c67d929240acb47bf9c56e1b57c3 : TIMED_OUT in 3h 03m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1583465bfe864f1dbcffb9196dbb43a6 : SUCCESS in 1h 24m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/51130dc6eb74449ba3966a1465d67be1 : SUCCESS in 54m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fd4b81a0fb854075999b1dafa86fbf69 : SUCCESS in 1h 08m 27s","accounts_in_message":[],"_revision_number":1},{"id":"59039176b015d2d80295a051f53e77786c536c72","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-26 09:43:39.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1f0515182a27f7dbcaa29bc616cf176fc7c028c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:55:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/027ac5009bc44f7e8bb5a2893e9970c0 : SUCCESS in 22m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/daab4d86a1e34e8ab875c884c25c570f : SUCCESS in 30m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a14f5a3dc0604ec3abcb54486ce751a6 : SUCCESS in 1h 04m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c01d32fee82f4654bb7c34521daed765 : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7351d2feaf34405385ec1755a763825b : SUCCESS in 1h 07m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7bab759cd162488cb4723d540a4b7e7b : SUCCESS in 22m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c9af861f3afe41f5aa7905f8a36b5905 : SUCCESS in 18m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/24a7215e709b45328cbce936bde5f298 : SUCCESS in 55m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a326ea4bbc584a4b8f818e9655233498 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5a1ffdb999f49fd867173949449d01d : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1aa137da8cb84d9fb7c2cd59afd6d0bd : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3723167aa14454977369ae0378d176 : SUCCESS in 5m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dca4b401ae654cdc8389f65021f3a5a9 : SUCCESS in 36m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ce13aeaee3246e98aa7902b313118e6 : SUCCESS in 23m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc83a03157ee44b8a79fcfefa103a5b4 : SUCCESS in 57m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ef9a08c39ef424a91e7bffcfd1cbbe7 : SUCCESS in 52m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b7c2f611caf7432483357883969b1a23 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f4df90831c0b724b536f2092d8160a35603a32a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:56:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/dda3cdc444c24920a00d64fb2625f04a : SUCCESS in 1h 11m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/73972916df2b482083850234ce28f7c9 : FAILURE in 1h 08m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e42ea356be744cfa383a6b9d688994b : FAILURE in 33m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/85ac2546fd4b4ff697426ff5b83499bd : FAILURE in 1h 06m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"635d1fb4ef81c4a2e1b9d50cfd9185b00b581446","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-26 18:34:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f46ab720843c887b3620d6892ea81b41279f3b8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 18:45:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa90f2feecd4599111089f05b8f46731a5b89ce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:11:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0480b3d871264a4cbb41712bf5ba3a81 : SUCCESS in 28m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e1485cd5187e43c1b0b744af3041f498 : SUCCESS in 57m 33s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0239fabc719c419cab92d0e684654dc1 : SUCCESS in 1h 08m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d26f8d0f5b46479dcc9732b2988529 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2e801eadb9840819aea6c96c82746ca : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbec137fe43e4b07a9095e1011af5a8c : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c116dc3860b24a68a3280744a5daf020 : SUCCESS in 4m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/04e19773f9034eab8047b3fbc19db95f : SUCCESS in 24m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c9421cbd045460aacb27219ad740bdf : SUCCESS in 54m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9bc824f9f7314995a552dd91e0814aaf : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5d3b0d6413c80aa12cfb1f24732df8a4d65cd3ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:11:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3a10dd896d7e928b5cc627d0f2d2f16b666adf36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:13:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44521dd444b42ba90dcf861b2a2715c : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"636a2438f52252f2675bd7914cc0fab0ad585234","revisions":{"636a2438f52252f2675bd7914cc0fab0ad585234":{"kind":"REWORK","_number":1,"created":"2021-01-25 16:49:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/96/772396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/772396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/772396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/772396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/772396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/772396/1"}}},"commit":{"parents":[{"commit":"aa42f8501484013bf265044685624dd6f73002d5","subject":"Merge \"[CI] Stop checking TripleO\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa42f8501484013bf265044685624dd6f73002d5"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-08 11:02:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-25 16:49:44.000000000","tz":0},"subject":"Bump up rally and rally-openstack","message":"Bump up rally and rally-openstack\n\nNew rally and rally-openstack versions released due to incompatibility\nwith latest oslo.config.\n\nChange-Id: I7ad2373327be29cd307db5ce4f7db022829e3b35\n(cherry picked from commit c0eef769db3d951f1e051af49ffcd95b6d6dfb87)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/636a2438f52252f2675bd7914cc0fab0ad585234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/636a2438f52252f2675bd7914cc0fab0ad585234"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
