)]}'
{"id":"openstack%2Fkolla~772507","triplet_id":"openstack%2Fkolla~master~I0d8417fdeb2f02138d03cc57bd3b7b95bc43fc8f","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I0d8417fdeb2f02138d03cc57bd3b7b95bc43fc8f","subject":"base: print release version in case of running unsupported one","status":"MERGED","created":"2021-01-26 12:10:29.000000000","updated":"2021-03-25 22:12:17.000000000","submitted":"2021-03-25 22:10:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772507","meta_rev_id":"ba99918253995e07bc1379b1940cb3f19b0da99d","_number":772507,"virtual_id_number":772507,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 22:10:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":1,"date":"2021-03-25 22:12:17.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":2,"date":"2021-03-25 22:12:17.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-03-25 22:12:17.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-03-17 17:16:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 14:15:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-26 14:24:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 15:59:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-06 10:48:04.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2021-02-09 16:48:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"9814f1e316095b6ab61485482c85b889adadcd9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-26 12:10:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d3a54ed26f815cb458c8540da75b97d57fc337f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:15:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6946a928a1944f7a3fa32a2ef2e3456 : SUCCESS in 1h 12m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0eb36cdab33c4527ad08e32130028ae7 : SUCCESS in 1h 09m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b680c3c55f849d4b840855d17835b24 : SUCCESS in 1h 58m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4de81d28d1834a7688942657d1ac54a6 : SUCCESS in 1h 04m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e330709f30846eb3e948e9d3aa86321a0bd484ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:24:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9499292ca8e6457290b0fb0cf26d4e3f : SUCCESS in 26m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b1e608c2d3be4c5db8ef406762c74948 : SUCCESS in 43m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9e4afa51d748458bb83377f650cf77c2 : SUCCESS in 1h 06m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f658dc9d9a4f42999498465c24df58d3 : SUCCESS in 57m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d291ab0e87c444ddae1b5f3bd26caa8a : SUCCESS in 2h 04m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/af084999b3744727a0d961b22b6dfc1e : SUCCESS in 54m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/56db8400c1a54b9c868ddeada338b014 : SUCCESS in 18m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3c435ffd1b424ce4a73c3b5149b8b473 : SUCCESS in 54m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97273e7e4d2e4f10a630a9b72fe2a9df : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284391c1e83744c7b69488c57ee27ee3 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7017af58cfcd4a559b6341d513145823 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/543e88f146a34122804bda19055d018a : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fbc7dd5845444da827f8351251835b3 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dff8f1bd97664c6995c2b0a17b2314d4 : SUCCESS in 21m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7458ffeea4145e5a93f52df18191010 : SUCCESS in 33m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e380b18c7c3946a5b26373e820bc3143 : SUCCESS in 53m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0a91be81288a46278c0757adb2c122a5 : SUCCESS in 55m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1fdbf81b03bf41fdbf0774ac31365a23 : SUCCESS in 1h 37m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0bf61f5f12c377f51fd10427a74c7f1838b3019d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 15:59:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bca5eafe2d8aeca9f12d8ddd4bd96ef0106a8d39","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2021-02-06 10:48:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c90d4d46c8a4f27f4439ac65001e2764f10c8005","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-09 16:48:11.000000000","message":"Patch Set 1: Code-Review+2\n\nnot W+1 because CI is unhealthy","accounts_in_message":[],"_revision_number":1},{"id":"106d8469d5e2d347030be6e5a544d4de60488dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-15 13:06:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d1fea0b9cdfa3de05d6d310cc8985d49020026a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 15:46:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/18ea835a31b74d3890b92b5f6302e59a : SUCCESS in 2h 33m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6da07065083946cbb9691ff86b101003 : SUCCESS in 1h 34m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/301eabf066834313ad8bfdfd6411bba2 : TIMED_OUT in 2h 05m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2547159a12564236bd4b34df764bc8b2 : SUCCESS in 1h 44m 06s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/403a463de878491690cb23c095b02525 : SUCCESS in 1h 17m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/66610dc58d5d4a03be2ba37708a90f25 : SUCCESS in 31m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7965932cddd143728086a6fdc366a888 : SUCCESS in 29m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/61afd6bf0e30487c9129b94bfd1cf8a6 : SUCCESS in 53m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4442204810364922b849acd6d9611e7b : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/386cedfeb09c4e1bbcbd5a266eb68371 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0ddbf73562f4722b192b57d78c9fe9b : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01695683b09547d39466f83f1b4de2c9 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccacca551fb84080ac5f63ff9546aef5 : SUCCESS in 6m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f0006d38ba84bf8bf496936268cb898 : SUCCESS in 23m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd180dc39972461d96e3d8fecfe8add0 : SUCCESS in 39m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff5197789e6d4cd2a5ce204dd97b2813 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91f4d1642b8648fb9b2bc7f4604a54cd : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d3a7c307101c47f6919ebeceb7a7dd29 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":2},{"id":"c334c6647c61258a82338dc87ca607abffec3c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 15:47:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/18ea835a31b74d3890b92b5f6302e59a : SUCCESS in 2h 33m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6da07065083946cbb9691ff86b101003 : SUCCESS in 1h 34m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/301eabf066834313ad8bfdfd6411bba2 : TIMED_OUT in 2h 05m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2547159a12564236bd4b34df764bc8b2 : SUCCESS in 1h 44m 06s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/403a463de878491690cb23c095b02525 : SUCCESS in 1h 17m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/66610dc58d5d4a03be2ba37708a90f25 : SUCCESS in 31m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7965932cddd143728086a6fdc366a888 : SUCCESS in 29m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/61afd6bf0e30487c9129b94bfd1cf8a6 : SUCCESS in 53m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4442204810364922b849acd6d9611e7b : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/386cedfeb09c4e1bbcbd5a266eb68371 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0ddbf73562f4722b192b57d78c9fe9b : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01695683b09547d39466f83f1b4de2c9 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccacca551fb84080ac5f63ff9546aef5 : SUCCESS in 6m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f0006d38ba84bf8bf496936268cb898 : SUCCESS in 23m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd180dc39972461d96e3d8fecfe8add0 : SUCCESS in 39m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff5197789e6d4cd2a5ce204dd97b2813 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91f4d1642b8648fb9b2bc7f4604a54cd : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d3a7c307101c47f6919ebeceb7a7dd29 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":2},{"id":"84839d6174ef9de248da2f75a7ff43aa01afa930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 15:48:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/18ea835a31b74d3890b92b5f6302e59a : SUCCESS in 2h 33m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6da07065083946cbb9691ff86b101003 : SUCCESS in 1h 34m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/301eabf066834313ad8bfdfd6411bba2 : TIMED_OUT in 2h 05m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2547159a12564236bd4b34df764bc8b2 : SUCCESS in 1h 44m 06s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/403a463de878491690cb23c095b02525 : SUCCESS in 1h 17m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/66610dc58d5d4a03be2ba37708a90f25 : SUCCESS in 31m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7965932cddd143728086a6fdc366a888 : SUCCESS in 29m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/61afd6bf0e30487c9129b94bfd1cf8a6 : SUCCESS in 53m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4442204810364922b849acd6d9611e7b : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/386cedfeb09c4e1bbcbd5a266eb68371 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0ddbf73562f4722b192b57d78c9fe9b : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01695683b09547d39466f83f1b4de2c9 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccacca551fb84080ac5f63ff9546aef5 : SUCCESS in 6m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f0006d38ba84bf8bf496936268cb898 : SUCCESS in 23m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd180dc39972461d96e3d8fecfe8add0 : SUCCESS in 39m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff5197789e6d4cd2a5ce204dd97b2813 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91f4d1642b8648fb9b2bc7f4604a54cd : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d3a7c307101c47f6919ebeceb7a7dd29 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":2},{"id":"c5724aeaf8f2dbd4283862cb7b13919089698aac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 18:26:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef61fbc106104a938db3c5f5f45e517b : TIMED_OUT in 3h 17m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cbddae56845c42208a54fc2884140683 : TIMED_OUT in 3h 10m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b34c50c0546b45acb9fb2f51b2529cb9 : TIMED_OUT in 3h 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a2b85d0a0b5d6740b49a2a16f0dc0b6e5f6e0cd1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 14:57:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1134dc6a3a32f57db879411e77184fb8cdd100d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 15:58:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/274861f0c3d442eab376812426ea25a6 : SUCCESS in 43m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7aa516f4784d4f828fed9dc83786f285 : FAILURE in 48m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/02a48d2c7baf4d64ac24707858a8890d : SUCCESS in 53m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84bfe635adc3ef375ce8c71efc85ad765c795302","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 16:07:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fefa166f6c05f04b4a6821b746a38da7e87ab499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:31:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2e75544d199944c4bf69c6056ecab032 : FAILURE in 1h 00m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/04eea161d9784f668f6fe3229e91e1c3 : SUCCESS in 1h 19m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8f6935ce81604886a2c5653edd03394b : SUCCESS in 1h 05m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/51c143397cdf4ef98aa5c2f03b30714b : FAILURE in 27m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7ea84e7132a94509971658da1a1349b3 : SUCCESS in 1h 20m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b8f4b0f02c584be3b0f749a26bbab801 : SUCCESS in 40m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3f59f65df1924ced9f9b61d645a4e1ff : SUCCESS in 21m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/97469983be3d49b1a4dd20d1094a3853 : FAILURE in 45m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/760a5bb316324f348329f4eca8cf1620 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba0b8461ebde4afd8511f4a99de3a258 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6b497d8484549158fad456841d95aee : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6598f42fd8604278a4c7c73ef5aefdf3 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9456aa397544d22807f836b8f4be777 : SUCCESS in 11m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f7f1897602bc409587b410b670733f7e : SUCCESS in 49m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0beba8cc3dc4edcb613910011756dfd : SUCCESS in 36m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/91e77bd34c6043b6a513f5a6509af95d : SUCCESS in 55m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/17a07774b7704e31b7ddd0a76dc42ea9 : SUCCESS in 48m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cf5a61999f284194bebd9d75f6515ea8 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":2},{"id":"96cd5ecc495cf30f66a84902c9ea417c0ae1330b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:10:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0fd61ffef59f4ca1b14c0b4be35f6441 : SUCCESS in 49m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a50528887dd24312ab995c837aeae248 : SUCCESS in 57m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ee9df8fea559574cc7f49704689cd2c45496898","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 17:16:10.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d1247ae78e94e3117bdbca31c8a7db86d9a48c97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:17:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9a4c9dadb9e99f9e09091fdf846f052646b5358","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 18:35:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8e3949fd645343078685c8c9e6eff34a : SUCCESS in 45m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d40d353f00484b72a1fae3a0136ae02e : FAILURE in 20m 07s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bbc83b8f2c114bb5802cc4f5bddf05a7 : SUCCESS in 1h 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/262eef6e2c38422facbff080803ea0d9 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58addfb4dc4a42efa59170378705f55e : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0e3277a69d24fb7ad48ffdcf9eb370a : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/715a5260613945f894b294f936fda771 : SUCCESS in 7m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/440680203fdd45b19b46b7e37783240c : SUCCESS in 42m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa751b42f848418398975cfd67a961b4 : FAILURE in 19m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/695b992700cf40eda68d2464c4e7fc4e : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":2},{"id":"782cf70b05f9a829db51e6e663fbda56e432173a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-18 10:20:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"eb0ea5a443b676e5840aba2a6bff99f102c0f82e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 11:28:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c69fb09fc9b4051a4613d697a3f7e0c : SUCCESS in 40m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ae0c81ea799987aa77465bf118ed9b6e816e751","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 11:29:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c69fb09fc9b4051a4613d697a3f7e0c : SUCCESS in 40m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12535f6738c2515a4af5bcec413ac9974faed20c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:07:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a110844fe4b04aef90e9cb10b2b5962c : FAILURE in 26m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d6570eb5a1364ce89a9c0875738c9289 : SUCCESS in 38m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/940ae2abacea4923bc4a8a5035603fc5 : FAILURE in 17m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/519cb1c504cf4804ad8fb11546fa77a1 : FAILURE in 15m 20s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/386045ac95af4f629b552d4559b5dbc0 : SUCCESS in 1h 01m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/055b3aef62224082ac5b59ca19ca80e7 : SUCCESS in 1h 20m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0c27b9717bc74c578b52c137fc8607e3 : SUCCESS in 23m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/64ed2239546c486ba9fbdc9382bcea87 : FAILURE in 23m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b755565592d149d59d5d413527296f25 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58de004680ad4a5f9090c0d3fa879006 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2408e8d0a6f14f9aa5bc88f656028c21 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06fe71757c374edfbb90bb26fc07094e : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf475e724c643d9a681d75f6b2484df : SUCCESS in 7m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e3d6c48092d496c94e9d3a7bed6495e : SUCCESS in 26m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ec3b25b0fce45969ea53c0cb2143700 : SUCCESS in 33m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c3b306db536b4e67a03451a5fbefdf9f : FAILURE in 16m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/338320709d48450a8d4ad127932028b1 : FAILURE in 23m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cb7f1d1655894395b0013ee8e740617a : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":2},{"id":"bb5fdcd0de2b76d79255175bb0dbfc569ac2c120","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 19:38:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"98ef355b78398a10294ec4dc88245d78bc6ee1ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:43:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/52202625575d4c6ca94f9b37ac5668db : FAILURE in 1h 00m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab2d13116a684c5cb5bcc3486bdaaecc : FAILURE in 29m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/92b9d21e374e492c91893b5537890a9a : FAILURE in 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e49acffb395eec3885488d0b80f59bbc29f58660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:47:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3fe90503afa9483b926758466d790a47 : SUCCESS in 25m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5262af7e1400456ca17d560cb243649d : SUCCESS in 39m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4777a3a9b10d4fcbac2e951f007d79cf : SUCCESS in 1h 02m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b22b506a68944c42b41f5120daaf78dd : SUCCESS in 59m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d7860daa630d457ca85c6b5afe27a04e : SUCCESS in 1h 07m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2910ae69495440b3b84c152a9fed835a : SUCCESS in 35m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b207587e8f2f4206b478775e7834807b : SUCCESS in 22m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/990666ed540a451e88ad10821530af6a : FAILURE in 38m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f145bcd4d24f6f8c063acea8fab9e3 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/366d5a063c4648628c31d1b2fab27d3c : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c148740cfe174ad7af67b0ce054a26dd : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5da54dddd34a4ad7af59179815645934 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c2a3e317034e51a6535c01bfb547ad : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d85d2fc51ed34280b142244683d135eb : SUCCESS in 24m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/68382b9ebde64ef2a4ad3b4fceb23a1e : SUCCESS in 45m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/10f35da5cce840608ab02ca9b3e5bbf2 : SUCCESS in 1h 04m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9729457ee714bbaa956d778a6a48266 : SUCCESS in 44m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/03183fee66fb4ad59794f1a3d65bb7f7 : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":2},{"id":"5f2e5c65071feb9e4880618d125dd42947070fa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:47:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ffc475d74b7aed98d903e5ee5b30e0804163aeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 22:10:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/39682b3f74754855a56fad8a921b5418 : SUCCESS in 50m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9a5bbf8ed6ee4c719a7ea5abfaa8c491 : SUCCESS in 1h 16m 29s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/56f6d752187a45a4b0a96491f0d7a33d : SUCCESS in 1h 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0fa23125cc94e69a2696b8774ae458f : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56b293c38bc9474c802a900e824fdb88 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64002fe012ce41478c3c29bc2528294d : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8f6c47d1d0d41578f2c12ddfa9df2c0 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbec5e5ec32b475fb09d5471359d4c5b : SUCCESS in 45m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e8618982bff64296a9aada846fe8df0d : SUCCESS in 1h 01m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7913c4f45ebf46eeb30ad7d3c34d9f5f : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":2},{"id":"47a4cbbdcecfa1b6649e8e590f886a2a1da0484c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 22:10:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2bdeef0a231370faee541d350bd3a7d956256f0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 22:12:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5ba70d165d4ce688848b76c5d7aba7 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad3c18608f2eb02db148faff15eb3ac7c99b7526","revisions":{"e9a7aaa2ca75165860688064e7ee4f05cfba2bc6":{"kind":"REWORK","_number":1,"created":"2021-01-26 12:10:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/07/772507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/772507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/772507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/772507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/772507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/772507/1"}}},"commit":{"parents":[{"commit":"8031118c631d883b8dd33c8c6292314abcbfb783","subject":"Merge \"Remove two missed unicode flags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8031118c631d883b8dd33c8c6292314abcbfb783"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-26 12:08:22.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-26 12:08:22.000000000","tz":60},"subject":"base: print release version in case of running unsupported one","message":"base: print release version in case of running unsupported one\n\nUser may provide own base image but it needs to be supported one. If it\nis not then we print info and error out.\n\nSo provide user with info which release is present in their image.\n\nChange-Id: I0d8417fdeb2f02138d03cc57bd3b7b95bc43fc8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9a7aaa2ca75165860688064e7ee4f05cfba2bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9a7aaa2ca75165860688064e7ee4f05cfba2bc6"}]},"branch":"refs/heads/master"},"ad3c18608f2eb02db148faff15eb3ac7c99b7526":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-15 13:06:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/07/772507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/772507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/772507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/772507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/772507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/772507/2"}}},"commit":{"parents":[{"commit":"60bb16520c42c3cbe8d720cb90bf452223971adb","subject":"Merge \"Make kolla_set_configs open files in binary mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60bb16520c42c3cbe8d720cb90bf452223971adb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-26 12:08:22.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-02-15 13:06:05.000000000","tz":60},"subject":"base: print release version in case of running unsupported one","message":"base: print release version in case of running unsupported one\n\nUser may provide own base image but it needs to be supported one. If it\nis not then we print info and error out.\n\nSo provide user with info which release is present in their image.\n\nChange-Id: I0d8417fdeb2f02138d03cc57bd3b7b95bc43fc8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad3c18608f2eb02db148faff15eb3ac7c99b7526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad3c18608f2eb02db148faff15eb3ac7c99b7526"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
