)]}'
{"id":"openstack%2Fkolla~772615","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I776af0daf6478505590a187a394991b8ef9973ed","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I776af0daf6478505590a187a394991b8ef9973ed","subject":"[DNM] add rbd-iscsi-client as a dependencies","status":"ABANDONED","created":"2021-01-27 01:56:13.000000000","updated":"2021-07-23 17:46:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d8d010018d67e1254a58ec2d7c62e06ca9c360bd","_number":772615,"virtual_id_number":772615,"owner":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-27 03:10:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 02:45:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-27 03:10:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed491f6de7b46009c680d56d4cbc1b40ededfb0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2021-01-27 01:56:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d655bc52319841f1924761156134676a0879cde8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:45:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/4352cef911bb4c05953e910e22edc2a7 : SUCCESS in 48m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c24b1838694465b8274d2e4683ba0e8 : FAILURE in 43m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/34e217540f794ea49e28b8a6871d7e7f : FAILURE in 25m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/38b8f84541294594b3767153236736f4 : SUCCESS in 45m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33f0e282509af3109e70f7e4459708964603ed21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 03:10:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e002ceffdbe44b88ab8cad1faad78f26 : SUCCESS in 29m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/71b63f3fefd94ee1a67d12de7f0035d2 : SUCCESS in 40m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5adc81399ce84a12a64ccb63983f73e6 : SUCCESS in 1h 05m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c87fbbacdabc42ec8869294a69349cb5 : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b137b224a52e4a77955e5a4bf9509a5f : SUCCESS in 1h 11m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dc21122d4b714ca9abf7815e3a1bb1e9 : SUCCESS in 30m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5b872758ffa8458f869dbc1633b1ea37 : SUCCESS in 26m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aac264514a1b452e839a8f2207b42cd7 : SUCCESS in 1h 00m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b5d6bffaaa4bfcab979ec7812ab1e1 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17a3c864a40a4ad397ea6de210c59f0b : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4601b4989db24c6c9d02aa8e9a307459 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd9c66f29d048449f642defabe60a43 : SUCCESS in 6m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/240615b6b62c4337addc4aa08e34bc39 : SUCCESS in 23m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d6496a487ea45789b5c833f56c33546 : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/47b1a67ab3da4b7ba1b505729b1a722e : SUCCESS in 59m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ac6f85bbd80f4271ab3c2972979c4497 : SUCCESS in 52m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1e2a3fae513a4894956ab97a7b4fb9dc : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d8d010018d67e1254a58ec2d7c62e06ca9c360bd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-23 17:46:15.000000000","message":"Abandoned\n\nold DNM on stable branch cleanup","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b50eaa227fe0459bde4675cdf293f0d5d2ab3f7","revisions":{"3b50eaa227fe0459bde4675cdf293f0d5d2ab3f7":{"kind":"REWORK","_number":1,"created":"2021-01-27 01:56:13.000000000","uploader":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"ref":"refs/changes/15/772615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/772615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/772615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/772615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/772615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/772615/1"}}},"commit":{"parents":[{"commit":"114e728a1378287ca711314a2821b233a4cba50a","subject":"Merge \"base: install setuptools to fix build on aarch64\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/114e728a1378287ca711314a2821b233a4cba50a"}]}],"author":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2021-01-27 01:54:09.000000000","tz":480},"committer":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2021-01-27 01:54:09.000000000","tz":480},"subject":"[DNM] add rbd-iscsi-client as a dependencies","message":"[DNM] add rbd-iscsi-client as a dependencies\n\nJust test, since this patch[1] is not merged yet, we lack\nof the dependencies to run ceph-iscsi-cinder driver.\n\n[1]: https://review.opendev.org/c/openstack/cinder/+/662829\n\nChange-Id: I776af0daf6478505590a187a394991b8ef9973ed\nSigned-off-by: Kevin Zhao \u003ckevin.zhao@linaro.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b50eaa227fe0459bde4675cdf293f0d5d2ab3f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b50eaa227fe0459bde4675cdf293f0d5d2ab3f7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
