)]}'
{"id":"openstack%2Fkolla~773013","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I4a13bbe959e3324f05d5cf78a3be2415eb2ba7ec","project":"openstack/kolla","branch":"stable/ussuri","topic":"mistral-respect-uc-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I4a13bbe959e3324f05d5cf78a3be2415eb2ba7ec","subject":"Fix Mistral source images to respect upper-constraints","status":"MERGED","created":"2021-01-29 20:01:21.000000000","updated":"2021-02-08 12:35:50.000000000","submitted":"2021-02-08 12:33:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773013-mistral-respect-uc-stable/victoria-stable/ussuri","cherry_pick_of_change":773012,"cherry_pick_of_patch_set":1,"meta_rev_id":"1ade3d0a372f06870578403d1b4b9559b5ec32f3","_number":773013,"virtual_id_number":773013,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 12:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-02-08 11:14:07.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-02-01 11:04:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-02-08 11:14:07.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-29 21:46:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-01 11:04:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-08 11:14:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"e88544574e2a7fff31f0ce65c5baadae3043b383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-29 20:01:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"e45e3e941bbb91b68274750e5cc4db1581c78cde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 21:46:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/866970cee532499294433556c408c77c : SUCCESS in 30m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7c5dd9eef45041d3a1a96aa4cd690661 : SUCCESS in 34m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6d2716357b3b422a97eb959238d250cb : SUCCESS in 1h 42m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8b856bfe70af4f49a8005e0ec75ebd6f : SUCCESS in 1h 13m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e2a0f1bdd1f04f6693f118fab25d695f : SUCCESS in 1h 06m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5cb6475b93d14c4bb3f15be1924728e6 : SUCCESS in 21m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/652651f0bb394491ae514745ac890194 : SUCCESS in 19m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1f6045bda3904354afa98841d4a9e394 : SUCCESS in 51m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd917069123045b98fff8efeca623d82 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1966b49bd5cb483dba01f31158221f20 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caa75318cac54b6a98fdec33d5ff8081 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb932b7db2bc4eacb9fb017cd091210c : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e9abda668ff4b268e84d3088af067b8 : SUCCESS in 3m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/10372501055544de84a815f301fd08a0 : SUCCESS in 37m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c0cab12b4f041a5bb20fdd961a368b9 : SUCCESS in 40m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0ff3dca1aa34720a6c6e6746b9cabaf : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c92a39d942784fb79fbff0612b25cc38 : SUCCESS in 1h 03m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/628508931cea41328975b0b2ee2ee178 : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fc654e749fcb94ab0fe292eab201b374a1748c4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 21:57:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/dd2b2bbc71374ce7b2c6a49437af64ac : SUCCESS in 1h 45m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/336f6331824149a58ffb39e35d3b61dc : FAILURE in 1h 44m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ca4f365e4ea4c5da183c33d4b4b4dc1 : FAILURE in 1h 02m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9bba44d8830d4727b7a2be3db902018f : SUCCESS in 1h 45m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ec8af5ba6aa89620b25a43b18ebf73fb47f331e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-01 11:04:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"619c9c3c670deb528a3409d39433db70bacc370f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-08 11:14:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5eb466d9ef32201381f395da0f6dc5be35b690ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 11:14:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4d3fbbca8aface599c68f95d493f21b1a0a817d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 12:33:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/21c56d493f5d45dfa55a4d2adae9f170 : SUCCESS in 31m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8f66cc405ac94ee5a1a5bcda8f74503d : SUCCESS in 1h 03m 42s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/19e779fcafa9432da688938ac865bbf2 : SUCCESS in 1h 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e891fb5f494424adf22b23a1473c32 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61f61519e33c4784af2facbf355b627c : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c352b03b46584099a2f51354d10c6036 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4bc802dfbc4d5a8a80d9759b9ebbe4 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ae8864d83614f6cb16f838f9db48041 : SUCCESS in 3m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fdc4512cfa61417aad27630fffc78357 : SUCCESS in 28m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/52c71aba2cde40a2b1bdfaf700ecfb5c : SUCCESS in 52m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0489d5d1695941aa9bc9af3f2ddf0c0e : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":1},{"id":"33d2954f4aeb669a6b28b9d5096bfab2463b4811","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 12:33:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1ade3d0a372f06870578403d1b4b9559b5ec32f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 12:35:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c697021952774b98a9a26edceb5ec968 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab435779bc7d48e4ab8a049ec2c59406 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a4044afb1f063d4d24ce782e546bd71d047c549","revisions":{"7a4044afb1f063d4d24ce782e546bd71d047c549":{"kind":"REWORK","_number":1,"created":"2021-01-29 20:01:21.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/13/773013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/773013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/773013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/773013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/773013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/773013/1"}}},"commit":{"parents":[{"commit":"114e728a1378287ca711314a2821b233a4cba50a","subject":"Merge \"base: install setuptools to fix build on aarch64\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/114e728a1378287ca711314a2821b233a4cba50a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-24 16:04:54.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-29 20:01:21.000000000","tz":0},"subject":"Fix Mistral source images to respect upper-constraints","message":"Fix Mistral source images to respect upper-constraints\n\nBackport to all supported branches.\n\nChange-Id: I4a13bbe959e3324f05d5cf78a3be2415eb2ba7ec\n(cherry picked from commit 004d9eb5d575fcb140b4d90fb30ced0a81dd3e66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a4044afb1f063d4d24ce782e546bd71d047c549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a4044afb1f063d4d24ce782e546bd71d047c549"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
