)]}'
{"id":"openstack%2Fkolla~773257","triplet_id":"openstack%2Fkolla~master~Ic638516eb92d24ad247a7866fd1b5e2ac0400388","project":"openstack/kolla","branch":"master","topic":"bug/1913952","hashtags":[],"change_id":"Ic638516eb92d24ad247a7866fd1b5e2ac0400388","subject":"Make kolla_set_configs open files in binary mode","status":"MERGED","created":"2021-01-31 16:10:35.000000000","updated":"2021-02-15 11:36:34.000000000","submitted":"2021-02-15 11:34:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773257-bug/1913952","meta_rev_id":"cbb0653f2c668d12d3ab337c316f992f4f63e370","_number":773257,"virtual_id_number":773257,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-15 11:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":1,"date":"2021-02-13 05:08:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":2,"date":"2021-02-11 10:25:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-02-15 10:10:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-02-15 10:10:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-02-11 10:23:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-31 17:33:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-31 17:33:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-01 11:06:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-02-11 10:24:24.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2021-02-11 10:24:24.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-02-11 10:25:53.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-02-13 05:08:11.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2021-02-15 10:10:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"b79dde7edca43b983bd35240a6f0e71cfed4a3dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-31 16:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68c18c488d11f73d25a6b1e31aeda9f01de8d4cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 17:33:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/23df14b02d3042e9be9f87c81cc2acc8 : RETRY_LIMIT in 8m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/17684bbecc584339bd81539004a709bf : RETRY_LIMIT in 6m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b18f782268a34a8cb896c60a3e525448 : FAILURE in 1h 14m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c1ce10a02b33451086a9a408bfc8682e : RETRY_LIMIT in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b84a76c34f81d5a703c23cf77c87540ace98b1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 17:33:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d9bee9d85c9a4a1f97297184f3572e32 : SUCCESS in 23m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2eba2144efb443fbb3c982b7496cc21f : SUCCESS in 53m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/df12cfb63f6b456d80844465a6f98046 : SUCCESS in 1h 05m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b4b2575e95404974b68288ca69742d38 : POST_FAILURE in 1h 06m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f8d7e4a49f9442f8af45d7d1e15e6ed9 : SUCCESS in 1h 05m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/995ebd1932da4a9dacb6446f7ae7a4ba : POST_FAILURE in 33m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dac1459b1e614e0db63f9e24461dd054 : SUCCESS in 32m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aa925acf8f584c10874b904c96f0ea3c : FAILURE in 58m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d52d7170bb3d4c5aaa8ebcf308b21251 : POST_FAILURE in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c99c0714102b48dbac3d9e2fb5950a87 : POST_FAILURE in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92152cd37891494ab7c3e74373daa3a6 : POST_FAILURE in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8794de193cc749eb85a358e0f2ef8b2c : POST_FAILURE in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a417c2116c42f897c6f7a61173b03a : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34ef6aa8c8174cb398522709a02bfefd : POST_FAILURE in 6m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94835d994de14cffa6427146c5aad705 : SUCCESS in 28m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6757bae7d67742439108145eedc5e19c : SUCCESS in 34m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4215fd185d0549a49780519b6bdbd4ab : POST_FAILURE in 1h 05m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/370936194d764ba49f99003b6eded1ce : SUCCESS in 48m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cd60947aabed4303b7cca590516e23b3 : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":1},{"id":"111085ec23d72aaf94bd5fd7133912750ca039ca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-31 17:53:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4549427429ed35e54155a55b0cfeae21b443bb75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 19:01:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/14f552e58ea74a67b58ecde3e8aa7495 : SUCCESS in 23m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/386068a36c194cc595f2f56e96192c6e : SUCCESS in 37m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b869d72318be4f2dbb5c8b39acbcc531 : SUCCESS in 1h 01m 27s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/55e245e2a03c4809818bec849dc919b1 : SUCCESS in 53m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1b4854afc31f47bcb951166cf9c5e147 : SUCCESS in 1h 07m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/29bbb7e4cbde486d9ffb27742bcb7299 : SUCCESS in 29m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8ca96fbe466449758cfc42775f86c314 : SUCCESS in 19m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ae83a7d072b42e4ab6367cf45f03455 : FAILURE in 33m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa71b937c0d44afd9bbfac8957ffbc14 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f44f8657747446a87f63d6cb10cff46 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ff18916dacf40e08dde0f1ded7816f2 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8e6fc92a94848ecbe17242306f4b213 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4462bbc5f2fd4f7a9db081b0cf545d75 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78eb23c760a242129dd93de6d9517dff : SUCCESS in 3m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d131ee031474bb0a9d13bfdd616ceb3 : SUCCESS in 21m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2fd8a9af14b47adbfe9bbdbd5420e18 : SUCCESS in 34m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/594a17c056b84cf8953b349744880a1d : SUCCESS in 50m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/88af296ad82d4177a79553fc13600710 : SUCCESS in 50m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9346e09bdaba495e8552632af61ff4a1 : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a9a512899e99a4a41e690da6b933a88d5c4b437d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 19:07:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/755b553cdf4849b8b97c92205656c075 : RETRY_LIMIT in 5m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7a99b2b9765416687a2a51d79387113 : RETRY_LIMIT in 3m 59s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/27da9bb04a434eb387f767b4ce91bc5b : FAILURE in 1h 03m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a722bec18f3c4f7f957cc3aed0ab2e47 : RETRY_LIMIT in 3m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7784ed1854c43cf25dd222320a83abffb3d9dd7d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-01 11:06:17.000000000","message":"Patch Set 1:\n\nCould it be covered in a unit test?","accounts_in_message":[],"_revision_number":1},{"id":"07538de8c9f16835227fec06e57b92db5dc3f1fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-01 11:16:06.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Could it be covered in a unit test?\n\nWe have 0 unit tests covering the _cmp_file itself. Perhaps we should have some indeed.","accounts_in_message":[],"_revision_number":1},{"id":"731ae73ef505b0d1f180aabffbfa3b38f1611a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-10 14:58:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c1e5e881de659a74433728b8b26bdffe51c4079","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 15:56:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d3dc999bb9a450fbb04e0a1e73e8a59 : SUCCESS in 55m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c75ca2b5dac4935b2329afc8d818d1b : SUCCESS in 57m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/178ffee64bac4682b4cd912ed8effb3c : FAILURE in 6m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2470036e38f4405fb370cfb796b27e36 : SUCCESS in 48m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"35f08bb52c1a0c19ba0beab903974cf272330c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:32:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/438dd6735c594d93acf9e225032d2ce8 : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/05d34313565c4f17a74fbb07de02faed : SUCCESS in 1h 27m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/22feac2b342644d4bfa0e6ccd0529b66 : FAILURE in 4m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f95cfc7a5deb4e969e8934556408959a : FAILURE in 4m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/dd0927bbbd86479aa07d3f2794b38e6f : FAILURE in 51m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d1f9742bcee341178a511601b886f9f9 : SUCCESS in 25m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3033a4ee102d4705827a01498356e9e5 : SUCCESS in 19m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18a82f687e45441292b34bbd11bd8fcf : FAILURE in 7m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53417010cdf84bb581ab66cf09e1b6aa : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bba564a834d14c608c63ba8202dc29a3 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a17d649875b7481f8bdb7e168ad87015 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74fa5ab6f8b14c4da9bc49bf68c37229 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00cd3d08ea6744e18315d8f3ec83f5e2 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60ce952877804d69800ad2764192d281 : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91f89d2f91684c608ab37ee1e90a51ec : SUCCESS in 21m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1272fb21d5d54708821ba1907fa3582f : SUCCESS in 35m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0358a70ee612437284e9399c2b953a6b : SUCCESS in 50m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e4cb4b774ebf4443a8d3991fa3c4ef66 : SUCCESS in 50m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b94ed5a5409498db48f69ff9f3e3b54 : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":2},{"id":"0ebae1221f976c9b1708136eed235efd928fc348","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-11 10:23:41.000000000","message":"Patch Set 2: Backport-Candidate+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"802920776a769342de5d70b3f388ebf3652ff1fa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-11 10:25:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cfe67452539e4b996040643ab70efc74a525f317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 12:41:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/88cdc3add5fb4e6d9a759e8ed47095bb : SUCCESS in 22m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/42d6b2ca7a9d4fccb3a3641560816a36 : SUCCESS in 47m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0a4b21d0f67844c682a36021a9f194de : SUCCESS in 1h 01m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/32360a6e645249ba8aaaf2779339daf3 : SUCCESS in 1h 14m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bdd1e20951f04d6aa2374033ab77ff13 : SUCCESS in 1h 10m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/34c8895b2ab148f39ac477a20d738eaf : SUCCESS in 48m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/db55642d34f14dcca5186ba6a71063e9 : SUCCESS in 18m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1a0d4163776949d48236bac32b575a6c : FAILURE in 55m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6433961175d84f5d9fca030e24c4549d : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f31805c6796046c5b0208462a92cfdae : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dac5a8c1f8e457c8ab1b76142841133 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8265d4dfe3ca4cf89464982a2d9d413e : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daa528264e5547edbb1dd9f8dcd4ceb9 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d160ed2f44eb4789a6a6b76e0b026e7e : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1dbdc1e376fb4bf78686c227d07afda9 : SUCCESS in 22m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fcfb3175985b4958a264f28d6faf9183 : SUCCESS in 41m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/96d0ed789e1341a48d7de2f7a1321d33 : SUCCESS in 56m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/932c059e84db438781f358f624f8a651 : SUCCESS in 51m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4c65cb0d500446d28af50e34491a20e7 : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":2},{"id":"74feee48d4eba7ef5c81c6b790864c6d601400ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 13:52:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/62fd5f1a714b44baa1ec5a4a649952a5 : SUCCESS in 2h 21m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c657b3dcdcf4e33a2dd4cc2f4df8a4a : POST_FAILURE in 45m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8fa71fa723b4af3c3114a730c6e34f6d7a38874","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2021-02-13 05:08:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5ddd1b14c383f49850f7b63f00f3a39067709c99","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-15 10:10:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ac0579e5d2fceabc107d10ef55080ff037517cbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 10:11:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"044473d3cf6f4f6cc9bcdafa3d0175e20a67de38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 11:34:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e5a804955d1e461abd1ba5eb43a76258 : SUCCESS in 41m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5cfacf522a1c4fe7bd3b686e69bc0ba7 : SUCCESS in 57m 34s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cbeee493c330493daea9a7d1d567063f : SUCCESS in 1h 15m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e9abf6ca9841848ff5b089172b4206 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e531a1ec3d4745a686dd31d4242370f5 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85baad19fa024665bdfcf419c9c36956 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843e0f16c5a94442b3175878cffaeaa2 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b10067428b04d0cacfb14dbea7a6f29 : SUCCESS in 3m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e39ffa5b36dc480b8c92c63b64714f11 : SUCCESS in 36m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b63c58743a2440c8bbe7041f0b722b92 : SUCCESS in 48m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/76a80d399a034d91b6c1304cbae14f29 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c71a45c0bc4ba8f2a93d975ecbe995f84b4c626d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 11:34:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cbb0653f2c668d12d3ab337c316f992f4f63e370","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 11:36:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266c76c99c8444758bfdf74054e98994 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fb327a1f6884600a4474131c4ee031f : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8e3027c5427ecfb89fe48ad38eb8189b1608d0fc","revisions":{"91234661b3a239eaa46a3f3aec55bbf3c92f52d5":{"kind":"REWORK","_number":1,"created":"2021-01-31 16:10:35.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/773257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/773257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/773257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/773257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/773257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/773257/1"}}},"commit":{"parents":[{"commit":"849b31303f4aad7d6fb190a80cf95514e8755279","subject":"Merge \"Fix Mistral source images to respect upper-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/849b31303f4aad7d6fb190a80cf95514e8755279"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-31 16:09:04.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-31 16:09:04.000000000","tz":60},"subject":"Make kolla_set_configs open files in binary mode","message":"Make kolla_set_configs open files in binary mode\n\nThis fixes comparisons when files are not Unicode-encoded.\n\nChange-Id: Ic638516eb92d24ad247a7866fd1b5e2ac0400388\nCloses-Bug: #1913952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91234661b3a239eaa46a3f3aec55bbf3c92f52d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91234661b3a239eaa46a3f3aec55bbf3c92f52d5"}]},"branch":"refs/heads/master"},"8e3027c5427ecfb89fe48ad38eb8189b1608d0fc":{"kind":"REWORK","_number":2,"created":"2021-02-10 14:58:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/773257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/773257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/773257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/773257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/773257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/773257/2"}}},"commit":{"parents":[{"commit":"849b31303f4aad7d6fb190a80cf95514e8755279","subject":"Merge \"Fix Mistral source images to respect upper-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/849b31303f4aad7d6fb190a80cf95514e8755279"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-31 16:09:04.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-10 14:57:58.000000000","tz":60},"subject":"Make kolla_set_configs open files in binary mode","message":"Make kolla_set_configs open files in binary mode\n\nThis fixes comparisons when files are not Unicode-encoded.\n\nA relevant unit test is included.\nIt can be used as a base for other _cmp_file method unit tests\nif the need arises.\n\nChange-Id: Ic638516eb92d24ad247a7866fd1b5e2ac0400388\nCloses-Bug: #1913952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e3027c5427ecfb89fe48ad38eb8189b1608d0fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e3027c5427ecfb89fe48ad38eb8189b1608d0fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
