)]}'
{"id":"openstack%2Fkolla~774122","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I4d20f23a9b26364943bf967908255d82c8f6621b","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"I4d20f23a9b26364943bf967908255d82c8f6621b","subject":"get rid of traces of CentOS 7 support","status":"MERGED","created":"2021-02-05 09:18:41.000000000","updated":"2021-02-09 12:23:17.000000000","submitted":"2021-02-09 12:19:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774122","cherry_pick_of_change":774098,"cherry_pick_of_patch_set":1,"meta_rev_id":"e89ef3c27430c11753719b486b07dffed67ec2e5","_number":774122,"virtual_id_number":774122,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 12:19:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-02-09 10:03:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-02-05 17:15:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-02-09 10:03:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 11:02:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-05 12:24:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 17:15:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-02-09 10:03:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ef536df35f7c247727d4a2d57178207cd6c2f150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-05 09:18:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b2f166449ba324a7964028484070e517063de934","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 11:02:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d23499b06fa4b9bb8c56b117a8def20 : SUCCESS in 1h 25m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/748307c8408a47ba8a9d16e51ff6e6ce : SUCCESS in 1h 18m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86e804e1682a462db7c84c7aa688dc14 : FAILURE in 1h 34m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8179044691264608b2901dcf69c0ccc2 : SUCCESS in 1h 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd8eded71c6e32c14c9588832863497c4325fb61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 12:24:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/94b17c0dae1d48db879d151d14a39ba1 : SUCCESS in 1h 12m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6702ca7df780420da6ebbfb346d840f0 : TIMED_OUT in 3h 01m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9d3842b4ceb14131ac261ee0c40e7beb : SUCCESS in 59m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6c7a7f58488147cb88bb0fdd725e0a61 : SUCCESS in 58m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9a568361c51f44d79809ceaa04ff7204 : TIMED_OUT in 2h 34m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ceefcf128d8743b796ebde9ca35c294a : SUCCESS in 25m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ae500275ade147ee911fe7c31c4db32e : SUCCESS in 18m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/694f8b7cfef44a7cbfc8adae7068e7a5 : SUCCESS in 59m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae24cd2966f946df8f419e1b0962480a : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88d6794c9a0047c29482d1804dc77db3 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68a9f173705e421fb42c7dd7d1cd9ad3 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8aaf1df775a4c66bf353f0eddeda34e : SUCCESS in 5m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a6eef4c7fec1470cad4c75626628fa7a : SUCCESS in 19m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d07c2704548438d8e3beaf7fa7ee43f : TIMED_OUT in 3h 01m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0bd3f847447437e88198d3cbffd2382 : SUCCESS in 50m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f96e3fd56c434e2f97545e4321f326a9 : SUCCESS in 50m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/98c1ba909fa14a558f1fc275a0ea83f5 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8d65500e1bc3722ef3a1d1e1ac2e9489076d2072","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-05 15:47:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"01044f5e801c1d2dd23a736813fc99c192dc0eda","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-05 17:15:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98e158715ba87955e26fa75638de39ebde5b4265","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 17:37:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d4628304a6249bda2cf3b9ffd6150e0 : SUCCESS in 1h 28m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c25cbfcb2985444db2d387880a9dcc8d : SUCCESS in 1h 33m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a5060a1d3f24c549f91fd47f3e209c7 : FAILURE in 1h 41m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ae51ccae57d54f10b694eaa21207aa19 : SUCCESS in 1h 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3dcd43688c78b3424bc647b9e361182f7f2d030b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 18:27:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c0e90da208ff4e6e9c42cf06ad447957 : SUCCESS in 37m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3dd2920f39c74b09917f7bac54811681 : SUCCESS in 36m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/df4ac65c83d54d758355975431cc24ee : SUCCESS in 1h 25m 15s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/21b1d5f2c9814b2cb53691744147c8c3 : SUCCESS in 1h 14m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0a6a928c5c8f4b4198193b56da2c9065 : SUCCESS in 1h 26m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/67357724d548492ba2ac8b8b5eb64b76 : SUCCESS in 31m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7ea9fa251a5d4554a4dd94e4631580d3 : SUCCESS in 35m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f3d6db158c8f45b49759f70d445dfd51 : SUCCESS in 1h 09m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c06db204af42a19478f1383af3daa7 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b41a3ed7d10b409180f3709c4334b386 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e171ec90d4b647d6b508f4c80ac806ce : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c7054dbfa341d3a25ebfbceabd2073 : SUCCESS in 6m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d1a803ff4054f8cb87d6752b4a1a541 : SUCCESS in 22m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e341764bf28044f29e8e0dfcccf9dbaa : SUCCESS in 24m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/48d5d6f1292e444bb8019208db2d4b28 : SUCCESS in 58m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4437dba3314e495fa1c36dd65fb284ad : SUCCESS in 57m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/91ca44048a314170a167ecbf2fdf0e07 : TIMED_OUT in 2h 35m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e2af0dad5e3e355eaf0baa74323d10a381700427","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-05 18:34:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e40743f13f0453af9ce2c96df30db1471438c84b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-05 18:35:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"27c5363e6a559ca1c0ac2925832f414b5f1f4321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 19:49:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/866270c143454b2fa5a7f9ef89c95fd2 : SUCCESS in 19m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/33970934b6f546d7a63e18705819bcbf : SUCCESS in 27m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c7ff438e215d41938ce7425f85acd8d1 : SUCCESS in 57m 29s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/912dd6eec7bb410195a4d1a4ffb0f504 : SUCCESS in 58m 13s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7c6a9e7704e64d9893dec500f37a5ef9 : SUCCESS in 1h 07m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0cfa1e1b3d744a08b194676431f7484c : SUCCESS in 19m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/60150aa11b25497fa9e3d78db334b7f4 : SUCCESS in 17m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/62144dd183fa413f99c261ac441f5474 : SUCCESS in 57m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3097b9872d2a43a5a140e617755b4d0e : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ecc54068095485b94678fea13164952 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ab79202157f439e9bd80205c5828eae : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1aec682f08f4a11bd96d05bb09752cc : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/175c77dc37274527a8c3ca60b50bbbf7 : SUCCESS in 20m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/952f60356289494e9eac4ebf23773ff6 : SUCCESS in 21m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbbb98f1fd574dae8809e75e654f38fd : SUCCESS in 54m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/949470b1144e4de09835282ccff1ed80 : SUCCESS in 55m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1e0e46c379b844bcb48a068785956be4 : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1ec9080124e8a46369928a87a6b4f3a12ceae813","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 20:08:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ae1b5457f36e4edf9d730e69279e0c6f : SUCCESS in 50m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ca2aa302fdf4fae81b9442bd8dd5be4 : SUCCESS in 43m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d89a05084aad40c397e1b4af16cd1d77 : FAILURE in 1h 28m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f47f084625be4afdb2d728d3c6fd3210 : SUCCESS in 44m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"496a77b62f09ab6db09e38e4b5d8530f1ff69204","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-09 10:03:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b15b4a6a34371120974dad3ff4e10b9f282903c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 10:03:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc1e71149cede5d2b8e3f8c7c1afce9172bbf89d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 12:19:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1529f9a27ed84ecf80b431fa0e60557c : SUCCESS in 23m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/21595753720e43f293a08590bb861745 : SUCCESS in 1h 03m 17s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ee95ad2e8d1746999d7b7f39d9c7127d : SUCCESS in 1h 06m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0bba957ddd4d158e135ab344cce549 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6168928dd1e74512a716f8719e167956 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd16253f49154da3bb034c4fd74c39b3 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0000bd7beebd4343a63da186a21ec08f : SUCCESS in 6m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84b11a2bef5c43c0a3bfcd7ef2e1397b : SUCCESS in 22m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cdd5f63b20e24e858427dba374d3c93e : SUCCESS in 1h 00m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3221358240f84c80b4106b8617f1dde5 : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5f5deee496273c082bb91c5ab356a978b0f8c2c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 12:19:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e89ef3c27430c11753719b486b07dffed67ec2e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 12:23:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb21806a4f94112b92c9547f8ec2bed : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65870d9650ed3141a84f7cf5173371f2f415fbc0","revisions":{"65870d9650ed3141a84f7cf5173371f2f415fbc0":{"kind":"REWORK","_number":1,"created":"2021-02-05 09:18:41.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/774122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/774122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/774122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/774122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/774122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/774122/1"}}},"commit":{"parents":[{"commit":"4e3b178e6e0d897beb5387d375c5563edc948946","subject":"Upgrade the Monasca Fluentd output plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e3b178e6e0d897beb5387d375c5563edc948946"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-02-04 16:06:28.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-02-05 09:18:41.000000000","tz":0},"subject":"get rid of traces of CentOS 7 support","message":"get rid of traces of CentOS 7 support\n\nChange-Id: I4d20f23a9b26364943bf967908255d82c8f6621b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65870d9650ed3141a84f7cf5173371f2f415fbc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65870d9650ed3141a84f7cf5173371f2f415fbc0"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
