)]}'
{"id":"openstack%2Fkolla~774127","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I0cfde46c29eedce77c70c1c677220814a801ff3a","project":"openstack/kolla","branch":"stable/ussuri","topic":"remove_sensu-stable/ussuri","hashtags":[],"change_id":"I0cfde46c29eedce77c70c1c677220814a801ff3a","subject":"Remove sensu images","status":"MERGED","created":"2021-02-05 11:00:26.000000000","updated":"2021-02-05 22:04:46.000000000","submitted":"2021-02-05 22:03:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774127-remove_sensu-stable/ussuri","cherry_pick_of_change":739440,"cherry_pick_of_patch_set":3,"meta_rev_id":"9ea1cd969bd54d7e1eb3648efc7ea6c503086eed","_number":774127,"virtual_id_number":774127,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 22:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-02-05 11:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-02-05 17:14:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-02-05 17:14:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-02-05 12:33:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 11:37:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-02-05 11:56:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-05 12:24:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 15:46:18.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-02-05 15:46:18.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-02-05 15:46:18.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-05 15:46:18.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"d73f37cdbff72c46f9740ecdf5833263463ea154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-05 11:00:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4da42b59bb5d48f6638e6b5d62eab2e3c1602bb1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-02-05 11:37:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8986d093fce0fef5ea393b266f416e932a052158","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 11:56:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c1c9098897cc45b58e5a71472e51f173 : SUCCESS in 51m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/893c0c8a86164f0a9b77438440623b12 : FAILURE in 48m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d76eece89f34861a156bd42e7d96958 : FAILURE in 34m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2d461bf7b3cf455096f689d6bc547b0d : SUCCESS in 49m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a08dabecdd35414acc784e3fa8be3c598bd812e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 12:24:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/de6cae4e0e494fd69fafd3937fb34c1c : SUCCESS in 23m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/228819b1c11f4459a79e9172bee1da17 : SUCCESS in 24m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a0f3f08d0e0d416985aca91a4c39c87d : SUCCESS in 58m 32s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/53aed66427a44779b45d1aff956e4ff6 : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/21e0095a2d414633af8e8a9d8b3f7bc4 : SUCCESS in 1h 16m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2b8ada642ca4462b83116c3c6eb08882 : SUCCESS in 21m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6f9bef794b5e4fa4b581fe5155a6ee68 : SUCCESS in 17m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e562052d7854f16a10f9960542beb53 : SUCCESS in 59m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c70f3b58e9e4fae8feb014381676f53 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b7a973457d748d5b7253e5f146352fa : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b711fdc3954b45c1aecbc033e1961316 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d20a3f699c8f4cf3b839a22dcaa13203 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cd73343ffa24a00a2699e7f1bb375a5 : SUCCESS in 3m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ce84a9fdc7845adbf46526be6fe7076 : SUCCESS in 19m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/53e7c55469324e00868b60c1a2790777 : SUCCESS in 22m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/53ca73229ede4066848a10e808c94515 : SUCCESS in 53m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0e793ad3756c4ba59f412e3da3338dc9 : SUCCESS in 50m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/119eaf35acd34e64a837f04f450b788f : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3278b99ea292dd224ae2bb2590e58977072d22bc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-05 12:33:08.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"3cac46ee873e3c47f5aed8152bfde4c7ec5fc09d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-05 17:14:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nbut need to reword the reno to mention that it is an emergency; best follow up with this fix","accounts_in_message":[],"_revision_number":1},{"id":"36a0883ff8f2d389a68b0bf38a1d248d9e6e5ea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 17:14:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc6b1b7200003aab191d15c488e9cf1c52166259","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-05 17:14:46.000000000","message":"Patch Set 1:\n\nand what about train?","accounts_in_message":[],"_revision_number":1},{"id":"143bfd43a7a368493e85892c99ef606262ce8617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 19:22:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b1ca5f5f669a42169d9c5007c06fdeb1 : SUCCESS in 27m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d2e334b1dc1c4e678c576bb5c1af9bad : SUCCESS in 1h 01m 06s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/85574ef4fbbc43fa9df2cda6fc98d5e5 : SUCCESS in 1h 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afad796ca20a4378918acc651b921312 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bd0272673b744208430e2a81b838754 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c957f573fa5e4cf482dc85d20396affa : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1c7f6737764de3bb14296950adce2b : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3de945bde110413e98adce83a5f5d782 : SUCCESS in 4m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed46d0e1e8e044fa94e90a7eb5063ab0 : SUCCESS in 22m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e4f844f07434c17869f0427952e4c15 : TIMED_OUT in 2h 04m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/17880f5f01884bb9992cd4d3e41f67f9 : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"86975e8b1d1e04df362a225ad0d14fee514ebf6b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-05 19:31:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18bd24df63454480bdbec28f48cf0a52d54052f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 20:30:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b366293acde7438ebb540813977d8c64 : SUCCESS in 42m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58432d7fffef473bbd2362bd0059725f : FAILURE in 50m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/24d6816591d8496a89e972a62f7d61a1 : FAILURE in 29m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8da9b8b6c48f43d9ab755dff156ce457 : SUCCESS in 44m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b7fa0e497e6fae779d05fd4b4304be2d9c510d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 20:47:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/546d26c4bdd34f9fab3c96a26d4393e4 : SUCCESS in 25m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ea26beae022a4bbd8de1d2da2a61acc2 : SUCCESS in 29m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b317c3cd6b0040dd8fe8561fb8c86b3a : SUCCESS in 1h 05m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d6a1cc6ea53d481c9bc955b6b39a812a : SUCCESS in 59m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/022605cb08524aa596d6db6dfcaa9d7b : SUCCESS in 1h 11m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/be8822fc7c58499795b9b4be971618c5 : SUCCESS in 22m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/35e5af073575411188899ef9365d93e4 : SUCCESS in 22m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/977f11e7535743ca9694ace95c0166ec : SUCCESS in 52m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e564e017f9964bc797fc2dc5bbb5ed70 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/109bd853c84641959c294715a29fc3ea : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8941e6737e3045f8a006e63fbd28cdd2 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ba68905073498b89a0edd1c63148cc : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1a98470f638412cbbdfe5306bac0f37 : SUCCESS in 4m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ab83bf3b46ef4ca48d763cee6f46c3ac : SUCCESS in 24m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a90ded625f1641f4843b03786fa6dffa : SUCCESS in 25m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/39a9634f109843ec83843b530830b9d6 : SUCCESS in 59m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f08357bdd1f45c1baffb1038cd1504f : SUCCESS in 54m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/040b78353b9640029d78388dbfe2b0cf : SUCCESS in 1h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"98587198c7074f915d4a3ceae0acb87c411d07ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 20:47:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ed716cd45765a606c4bf904ee14c6349b7eac1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 22:03:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f8ae7db5a8ab41cd9fe427a19f63359c : SUCCESS in 28m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/84a6b0bb757f4ecca085424479082f7e : SUCCESS in 1h 03m 39s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2725beec03ae41f0a57f5a531c287f9d : SUCCESS in 1h 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71878b1939af42eda1738207f13bada4 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22c7d37dae374f57bfea1a17dbe75be4 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce7455d53c7148feaa7334976d73be4f : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9192420f8a64f47bf55151f0d116b45 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3eac72aa5334003a6c2e6a38724674c : SUCCESS in 3m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ce7580b66354adc99d4ade0ff406493 : SUCCESS in 28m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b414552c8cb44e09a71091b023f226b4 : SUCCESS in 56m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9c7f58dfa18f4bc0bb948b1fbdabe0b8 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e834f87689f8379adbd780ac20da32b6f6691c2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 22:03:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9ea1cd969bd54d7e1eb3648efc7ea6c503086eed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 22:04:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45293e2ba768473a8874f6e59845b246 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1f9e00183b947f69f055c6a488b9e56 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27b2914b43afe64d1e7c218aa8b6876851e9d9d6","revisions":{"27b2914b43afe64d1e7c218aa8b6876851e9d9d6":{"kind":"REWORK","_number":1,"created":"2021-02-05 11:00:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/774127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/774127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/774127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/774127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/774127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/774127/1"}}},"commit":{"parents":[{"commit":"114e728a1378287ca711314a2821b233a4cba50a","subject":"Merge \"base: install setuptools to fix build on aarch64\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/114e728a1378287ca711314a2821b233a4cba50a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-07-06 09:33:24.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-02-05 11:00:26.000000000","tz":0},"subject":"Remove sensu images","message":"Remove sensu images\n\nThey have been deprecated in Ussuri [1].\n\n[1]: https://review.opendev.org/#/c/711636/\n\nUpstream dropped it:\n\nIMPORTANT: Sensu Core reached end-of-life (EOL) on December 31, 2019, and we permanently removed the Sensu EOL repository on February 1, 2021.\n\nChange-Id: I0cfde46c29eedce77c70c1c677220814a801ff3a\n(cherry picked from commit 0cfd36292cccb8d3af9c65a1dff9696f50870df6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27b2914b43afe64d1e7c218aa8b6876851e9d9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27b2914b43afe64d1e7c218aa8b6876851e9d9d6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
