)]}'
{"id":"openstack%2Fkolla~774602","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ice125e101a689fd3df9d1cd9a093a026f9747a3e","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Ice125e101a689fd3df9d1cd9a093a026f9747a3e","subject":"Fix several issues (Train)","status":"MERGED","created":"2021-02-09 09:05:31.000000000","updated":"2021-03-17 19:22:05.000000000","submitted":"2021-03-17 19:20:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774602","meta_rev_id":"14dcc933b8406f977c7d32cb9bdf3440c117c1fe","_number":774602,"virtual_id_number":774602,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-17 19:20:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-02-17 09:52:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-02-17 09:45:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-03-17 17:35:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-03-17 19:22:05.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 10:04:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-09 12:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-09 14:14:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-02-11 11:36:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"6cebedb7bc16761d65dda8e500c8c747e7c0d6fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-09 09:05:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a809952383ca92adfaf003518bd993be240030c4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-09 09:06:10.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2\n\nstein needs that too","accounts_in_message":[],"_revision_number":1},{"id":"226d786fd09e0c3a9678fe194932cae164e6f4af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-09 10:04:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f6b8255aed1b7b0776cb609a57f582cd30e5be36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 12:08:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2f0bda502f8540dd8ab3f24593c2dc73 : TIMED_OUT in 3h 01m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/36ed21e7e71444fdbdd85ddf6da0a753 : FAILURE in 2h 21m 14s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6185df79a8254c72aedae8a9ed1a1d03 : SUCCESS in 50m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/58d421946c5a4fdea349b8504241c259 : FAILURE in 36m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fd7cd75e613b4357b7eea28259420a42 : SUCCESS in 1h 14m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/122925c70bc84e8fab6e8355b8a120af : SUCCESS in 51m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/337bc540bf144d7d87831c17ccf5765b : SUCCESS in 1h 14m 38s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5d94a0102d2e4942818b1dfbf419eaf1 : SUCCESS in 55m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7c8a5a6c9edb4cb9b9648f7facca3063 : TIMED_OUT in 2h 31m 23s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f43f9bf3a91245449aa960db8e950692 : SUCCESS in 55m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2ddf142878fc4d4f9d97b777c94e97af : SUCCESS in 17m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e706773ac2294ce59ea09b96090855c3 : SUCCESS in 17m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a438338618aa43078cf7bf8a609daa6d : SUCCESS in 1h 42m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2a3460a1efe48cba653622a5f9728cd : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2afb24e6c5a649b3933b835e03d3567f : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d80283cfb9114eca835cd0179c2fafaf : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36dc9b1c510744468fc5d9127cf4bac6 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b294b34bdec647ca8586c114fe3ce5eb : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db728a6b7b0c4976a7a024bc9d1b63c7 : SUCCESS in 18m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3ae00d3fc4942a98a3419369b0c0eb8 : FAILURE in 34m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/49f4701d65b348008c25d15941db60bf : SUCCESS in 47m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1420bc5f8dcc4529b4e12727d4d26e38 : SUCCESS in 45m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/378b26e8426b43bb8c18b1387fb776d4 : SUCCESS in 55m 35s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"e4c0ca59524f5536fbd4f12b2eabafe250fd31c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-09 14:14:33.000000000","message":"Patch Set 1:\n\n*cough* reno *cough*","accounts_in_message":[],"_revision_number":1},{"id":"fdab642cda614543242fc247407423b8d9a03c06","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-10 16:54:03.000000000","message":"Patch Set 1:\n\nFix for bifrost is in progress: https://review.opendev.org/c/openstack/bifrost/+/774237","accounts_in_message":[],"_revision_number":1},{"id":"7e29aabe2edd03a902638096efee584de93d86a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-11 11:36:09.000000000","message":"Patch Set 1: Code-Review+2\n\nnot approving because bifrost b0rken\n\nand I would love to see a reno here (followup works well because it runs only the reno jobs)","accounts_in_message":[],"_revision_number":1},{"id":"47566f1494861cd0c951be19a92508d2ea7265e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-17 09:43:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"04a160d97027ec0d684f30ec8899c31129f93997","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-17 09:45:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ab92ca8936426760c2be8d182eb5ef3258103417","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-17 09:52:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"276afd57424b3bd585d66c54ad1cdb22205e21a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-17 10:32:23.000000000","message":"Patch Set 2:\n\nFails with Elasticsearch on Ubuntu/source:","accounts_in_message":[],"_revision_number":2},{"id":"0aaf2b452452d650a874c1290680b33edfee8561","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-17 10:32:32.000000000","message":"Patch Set 2:\n\nINFO:kolla.common.utils.cron:E: Failed to fetch store:/var/lib/apt/lists/partial/artifacts.elastic.co_packages_5.x_apt_dists_stable_main_binary-amd64_Packages.gz  Hash Sum mismatch\nINFO:kolla.common.utils.cron:   Hashes of expected file:\nINFO:kolla.common.utils.cron:    - Filesize:183773 [weak]\nINFO:kolla.common.utils.cron:    - SHA512:fb44640e4de0dedb137341ebdd7e0d3c6e22f84f96269a1322f0e863af5ebc774e6a62a6fce0d54357c2d494574864b9ad3b99ba113cacefc033215756ac854d\nINFO:kolla.common.utils.cron:    - SHA256:6b2ddbb7d7a0fe7a60f2e7693ed1ea747b9d10dee2db0962000f64816af89e99\nINFO:kolla.common.utils.cron:    - SHA1:6a2ad4dd253eec7eeccb1dd9647c290938dd20ea [weak]\nINFO:kolla.common.utils.cron:    - MD5Sum:db65b0fed147907bef2a5ee35f84f9e5 [weak]\nINFO:kolla.common.utils.cron:   Hashes of received file:\nINFO:kolla.common.utils.cron:    - SHA512:a6b8f7f314895805e17bed18d8382b8091bea6d7c9c2e23d2ddf71fef863dd901bbcc8406361437b80fd912ef03444861f1c2496ce134becef16eb2d6ab3c2c3\nINFO:kolla.common.utils.cron:    - SHA256:9c0e0ccbfd143c94c33656f9a0dcb315822a9f6b5098c9d56b138f119315d7f1\nINFO:kolla.common.utils.cron:    - SHA1:77c16a3270cf2738faf205e2f8837b1b72e69724 [weak]\nINFO:kolla.common.utils.cron:    - MD5Sum:5d3c3bd960acae47382a002278071ad7 [weak]\nINFO:kolla.common.utils.cron:    - Filesize:83482 [weak]\nINFO:kolla.common.utils.cron:   Release file created at: Tue, 19 Mar 2019 14:51:51 +0000\nINFO:kolla.common.utils.cron:E: Some index files failed to download. They have been ignored, or old ones used instead.\nINFO:kolla.common.utils.cron:\u001b[0m","accounts_in_message":[],"_revision_number":2},{"id":"d892be4a038723b514411efd7390378058e179ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 11:18:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/dbe867add66f4ba4a62589523e7e96ca : SUCCESS in 36m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5cc1c879ed9643e594bd373ac2a73416 : SUCCESS in 44m 50s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/09729b6aa57c40cd87a3bdc59df63548 : SUCCESS in 30m 50s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6a5906dc41be4c7988bf2e8eb66ed6ec : SUCCESS in 28m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a6a81496bf63418ea2ccc4e90ae69e12 : SUCCESS in 1h 07m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b1a15badfd5547b896533dae109f48db : SUCCESS in 53m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3cd5816990ce4a6ebf06acb45ad09afc : SUCCESS in 1h 06m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/86e7a128943b47ddbe457804d87cda45 : SUCCESS in 1h 00m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8975cae882234b94aca2d7b443337807 : SUCCESS in 1h 21m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/42deb510aed04f758d7a83987043e9a8 : SUCCESS in 44m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c6ac5767765740cc8c02c3cb2730eea4 : SUCCESS in 29m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a26cf6ef118e45c7a215005f41a3f467 : SUCCESS in 19m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f6353b8cbdd945a2bb29115b5abd9711 : SUCCESS in 50m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a5bd20d79c42da9bdf599fe733a270 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9bd87311475941b087c8b3a8268885a0 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f366d6440bd4344ae84b4053364c01b : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c61b7517b76472fba4973715ab6dcef : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fbc21ee3c144ac08f9af2b0cecec91d : SUCCESS in 6m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3bdd06a5b174dfa987fc697211fc3a8 : SUCCESS in 26m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c30b93f58e9d4405b1cf44eacadfc145 : FAILURE in 11m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe15f23d33a649fca468fbaa7a1fc79b : SUCCESS in 49m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/da7ee032129b4da5ade9c1a6c39e83b8 : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5748df4d599f4e0eb39ddec25da8f06a : SUCCESS in 1h 05m 26s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"fbe1a8ab8eb279b2c3d7fa6a53affa965fb75fc1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-17 11:56:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c275f1b3f0cbd6f09d0fd447a78f767affba7e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 14:00:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/856a04b5ac094926be94e8c8b2e35122 : SUCCESS in 36m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/38a370ae17184a48826931b2921fb67a : SUCCESS in 53m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/59cf271ea4df4da1be83c4eb2405307e : SUCCESS in 34m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f6b6c53cdbb04fe8a632ce4f38f9d5f2 : SUCCESS in 30m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9ca850c7f87a4431a82a6ecc84a3ed14 : TIMED_OUT in 2h 01m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ccc27537609045d394d8b5ee5a12522b : SUCCESS in 1h 01m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d7684ebd8ad84f4695bf3bbf283b4f18 : SUCCESS in 1h 02m 03s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9fdcfd92582d44229caf3ad4b7954045 : SUCCESS in 1h 02m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c735896bd5d14b2099fab7ae0984ca55 : SUCCESS in 1h 41m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/22363aea3c564032b7ae59e93eecb9a8 : SUCCESS in 1h 11m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2171e5b408614b699f5d06bfc44e7327 : SUCCESS in 23m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5e0cd446515f4d119e9d315faac20f14 : SUCCESS in 26m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1365d6804a58408d96e6e895be6453db : SUCCESS in 45m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01fb61bc5def4085baf037ea5855f145 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/312abfe0e7094f07805855b2b0a4ff11 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c495be1dbe142b1ab35da805e7242e1 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/301a2e7bde0f440b9b32609cc168a8fb : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e7a077a6084ceba6158a33e17e3cd6 : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99684361ac7742deb41d156ee79693c6 : FAILURE in 10m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/605012e4c7f84c07b35c592e012b84b4 : FAILURE in 31m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/be9376a8274b4fb6848f451b472b9a0c : SUCCESS in 54m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bdbb2b35daf64093b882ed837d1267d5 : SUCCESS in 47m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62e6e05528ad424faa1433b4da8d7740 : SUCCESS in 1h 17m 26s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"f99deec278b2af357e3f37cef2b0c818bb2433ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-18 09:30:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ffdb49501dd4f3402b82a584868995b0bc4c8557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:56:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2ea8e260aa1e4f13a3b9db5e6f85223e : SUCCESS in 45m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9576f5624245420596d4f8083469cbef : SUCCESS in 1h 02m 51s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0795082033974a9ca5687ad8a51bf8d2 : SUCCESS in 23m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/629b1c6d860a40368cc8a23c35217cf3 : SUCCESS in 34m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/33ae52749794477b884b9d23c445cb44 : SUCCESS in 1h 19m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7370006ac19e491fbac33c3262e462a2 : SUCCESS in 1h 00m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c266119d25b04c8790227cec4b7ecaab : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/31184bf9e70a41b99669d1f9947eb20e : SUCCESS in 54m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c5a95baba6194ce58dffba3c05d4b640 : SUCCESS in 53m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ae4c42e42d284592b4bd7850e55985f8 : SUCCESS in 53m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9c483966125c4317a3f8a15ba67b72ba : SUCCESS in 19m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2308844eec2e486e94df8e26b43e63e9 : SUCCESS in 26m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7a16f516c879492aba6b483af3221c44 : SUCCESS in 50m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c32ece748914c5eba5e4bd3e4314f26 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3efd2b8604ad47feae63c68591585ded : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4c0fef16218472788840df5cb10ca31 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7721f929f4e042d08594220dfe18a533 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e000edd524f4aa8a896e9c209520db0 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b26d9a44a8044b0c81faa432354db48f : SUCCESS in 33m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b897992cf89e43d1b460bb441be41515 : SUCCESS in 33m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c208be771c6641a3b734047c78fe1489 : SUCCESS in 58m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eac97efb0acd40388416d103a32390d9 : SUCCESS in 48m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ab2427a221ce41538e2a975c068035af : SUCCESS in 1h 03m 03s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"f60e625d2077bd427202364b384207de30e2ff11","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-18 11:09:33.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"865364c789d5de33bd9dfabdb8625ac98f638897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 11:09:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ed5ad1a251fa977ad34f6daa15b1c6662514f0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 12:33:38.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3334143d120f43a098bde90ec210a47a : SUCCESS in 55m 15s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d741982c437949adb8cf65e989820fcb : FAILURE in 4m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ac62277343cf444f8b8a11e688a96f55 : SUCCESS in 1h 10m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7062dabc54ce4d308641abe62e872b6c : SUCCESS in 1h 10m 53s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0320869adac545548e643173f99d3908 : SUCCESS in 1h 19m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/be19acc980b04681b051284ac7e4c0e7 : SUCCESS in 49m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03c23d6f57464bc388c99f3165fdff59 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88a9f4b7a4ce4017ab7ed0f3b0388683 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d0002f850394622a1e46490715d394a : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c9bf1d546b64c16bbe9fe5d46cea574 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c1a325069f4a7c8e96bc67ce3f7302 : SUCCESS in 5m 47s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4415b39e213b4acaa14c40c93eabdc6b : SUCCESS in 23m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1862ee02b9e41458d56b1c214571369 : SUCCESS in 47m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/91a6ac7b846b4235b321179144f50d36 : FAILURE in 1h 04m 30s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"8a7d9119f39e4e59670ee766f987f80cd8824531","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-18 13:12:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bf6f42655cba6f89f9adc54b994a7d2c8464dd27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 14:50:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d7c5465cd62b4eb4918652cada0bcce5 : SUCCESS in 42m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a239651b9af540528560cc8a0ba1e317 : SUCCESS in 45m 56s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b91bfa08435945fc9a4604e526c7bebb : SUCCESS in 32m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5db29579079244c8a1464f4fa44697f5 : SUCCESS in 28m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0c7d6dc395f34251b50d9361c6eecbe1 : SUCCESS in 1h 04m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8d3bbce019144bd2af53f7ee2a1e6518 : SUCCESS in 52m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/71ec5b2ec25f4fdcadeeb2fe74d7c394 : SUCCESS in 1h 04m 34s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d8e0bb84f79f4adb97a2affd6da1cc7d : SUCCESS in 51m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5fe48b793cca4f96b5902cd8ab946f81 : SUCCESS in 1h 31m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cef9a16c894a442e8672e28d642ae5a8 : SUCCESS in 52m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/343b21f8ca2c42fb9a8c8f23b9861443 : SUCCESS in 34m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c209d3dbf0614112a7446b43008c3cee : SUCCESS in 18m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6190ca1fea4f42a3abcaafbe2bdf4dc9 : SUCCESS in 49m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dbe10bbdec2467ea1e9fc5e7df440b6 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08305cafe8a7470aa5b2d463a1f6a157 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43a79c637e8a454e9a5f1bb877bdca56 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7032f61a27264b06874164e242622e9c : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f162487b2e9442883045d9842cfd76b : SUCCESS in 9m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa090a174bb84b7b82d103f1ed04e528 : SUCCESS in 27m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f21e26eaaec4727b07d27d74288ffd9 : FAILURE in 5m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c235a6b2442744809054cf757130fbb7 : FAILURE in 7m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/12276bcb93f64f24aeea477663409057 : FAILURE in 7m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d075fdf43e6d4679af446f59f15e9b97 : SUCCESS in 1h 05m 58s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"032b0bfcb867fc7f043da5455b09deba01fb0c33","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-24 15:07:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"416a9a76364c9d2aa9b21faae50e692d4c4886b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:50:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7164c495574346f2a9fe6788aab79e61 : SUCCESS in 38m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e1d83d303b824a09b745ba3b0628ab78 : SUCCESS in 43m 44s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a1940e5a29f9442782406e544b772553 : SUCCESS in 24m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5158035bcee647d388a456a8716044b4 : SUCCESS in 45m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cc8361e2ea1747008146aebdaa992965 : SUCCESS in 1h 07m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9c996495f2e6476c8876b35e9fba28be : SUCCESS in 1h 05m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/69fcb5ba2da844fd973d5f92c40b060e : SUCCESS in 1h 19m 25s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/63c2a9c7ea534b60a31bb4146da018d8 : SUCCESS in 1h 02m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c29326083e57455bab30ede68e48c8d2 : SUCCESS in 1h 33m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/769eedc6529045c78740ee10ea135ea6 : SUCCESS in 1h 08m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/806e1fdce9c247eab7811ff475d765f3 : SUCCESS in 21m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b7a7e66c1f8c4a4eae8196ff4dfa19cd : SUCCESS in 20m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d2ed9750d0f145ef80a6acda0425444a : SUCCESS in 1h 12m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab7dcc2890e4de3b0a569432905f305 : SUCCESS in 7m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ee68e1ba0dd46cca4d6869f9034b7d9 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f3a989a17e34bb4a139109f88a2f14c : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/055db60031ec4dc7a27dd1f377e62c0e : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37a2e998a14a4a508afcc4770b6d40d7 : SUCCESS in 6m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d39757fafce54a698d4c8a9110fbca96 : SUCCESS in 54m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/88667ebf787a4e4cbfc488ba4458a444 : FAILURE in 38m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/548d05297b444f58b1bdefd08419b59a : SUCCESS in 50m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/428bf60d08f94a6d94d8ee8160ecdcd2 : SUCCESS in 47m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/258b5ef5a0634abea7ccdae5fe808ddc : FAILURE in 10m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"f0b3b116229a9567de0743e2e18919fccac61860","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-01 11:37:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cf50e8483d1977651a2e5b34de1735905ad7dc66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 13:20:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/58253eca0b974e8d9cd3c71372225ebb : SUCCESS in 45m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/66abe499a8634849a50accc2cdd23b77 : SUCCESS in 1h 09m 01s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0d2f2468a04e4dd4b81093acaf22a439 : SUCCESS in 28m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8de5242922254323a7e8e2aadf99c30e : SUCCESS in 34m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4b8c03cc18ea49f7948692e3b79f12cb : SUCCESS in 1h 03m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f73e75d80e264964ab20f37d888ca2f1 : SUCCESS in 52m 06s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b63ac8f241a34e9993a4372f8cf6f35c : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/27c5ec6c239e477bb2d24afbbd8059c9 : SUCCESS in 55m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a7042366d5904839996cde1d56253d55 : SUCCESS in 1h 33m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/795aad3483454cebaad8abc6b82dc807 : SUCCESS in 44m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/40d04184655549b0b837350962204bc8 : SUCCESS in 30m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d3e2cc3fe403413197879a6979b5a968 : SUCCESS in 47m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2a6075ab501a4210ac49d9320a158000 : SUCCESS in 56m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad3c3ee18094165a0864d5af9f87115 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b29d0e6689d477cb7f9f96ee91abb53 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8a8693c3bbc4b64967b7355ce58d55d : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8775e806c00b417d8a564ea5b8794c6d : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4baee71d8d114e388dc33df60f0b75a4 : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e028ab18774d493db01b78597e289c97 : SUCCESS in 30m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/434aea6953ef43e99d68d11c8991c2f8 : SUCCESS in 33m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/abc525bcb28f44e282fb04a24629acb4 : SUCCESS in 51m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d8039d7a90264af097c6d4aea9b6661c : SUCCESS in 57m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4baf28afc0b948ab80f1a2e39753021c : FAILURE in 20m 42s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"38daaaf64379903c6d0689cf12f257cac41f824c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-02 10:30:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0a7cf194a8e2cba71850386c06dd70b2d874628a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:27:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0eaf4c7039a148ddb31171d7e9a981a9 : SUCCESS in 46m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d9c9b65d77fc47828e877e35d92fd976 : SUCCESS in 54m 54s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5fb8c41de75c4ca9a099ce79b502a92f : SUCCESS in 41m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/61312a368cb347bc9301dc43cf7a85b9 : SUCCESS in 41m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/02055779bbf7449380a7f517940f455a : SUCCESS in 1h 48m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/019e584021bd41709cdbc4cfe1575b4b : SUCCESS in 54m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0f4d315a47ff4aa4abae79f52001ff22 : SUCCESS in 1h 15m 37s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6577932d96324e87934ab809549c4d6a : SUCCESS in 58m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b637319d0a554b9d8450007b156cd21a : SUCCESS in 1h 24m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a26e67297249435bafdd9766f35dca70 : SUCCESS in 51m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4308a9b11b4048c789d69e2015efcfe3 : SUCCESS in 44m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f531b0bdca0a4065a68b94f3ec714e21 : SUCCESS in 42m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c8090b9b71a8474e8d15ecb6fde59d31 : SUCCESS in 58m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaed01a2331e4e1e96858e4b333786f1 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/29e18f7700da405c9c894c40271e8182 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ab99b5a46754b778fb8723eec5ba249 : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/700aba9323054d14ab150e92c965965f : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0226fff018604d7fb52066ecb709ebed : SUCCESS in 7m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37e70e16edc74a85a7f3091228ea0835 : SUCCESS in 44m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c5799cb046d4e10bbbfa599d6b0783c : SUCCESS in 47m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a773ae1302a840cf984a15c05ecb6a60 : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/08ca052ba3dc407b847672f64ad647b2 : SUCCESS in 55m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f6988a8a350544dcb8958ab17deca6fa : SUCCESS in 58m 17s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"79b7ae1cc5b2c72cb77911f8ce080c30e0fefe0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:28:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b09decd07c4f2d0b8e4eec788aa592140a0dc076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 14:27:59.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/45f87da592734fa7bd3a0c04711f853b : SUCCESS in 55m 51s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a547d1dcdfce4659a5127caab4c8fcd5 : SUCCESS in 36m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dd029849f4714275a8d626b7a7855f28 : SUCCESS in 1h 13m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fbf2bafe1d6b47e3a2e3e06b92649197 : SUCCESS in 1h 03m 39s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e3f01b45ad2a40ab833dc4022ff1388b : SUCCESS in 1h 53m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0ed9fdb9173b40d19d2728ffda825054 : SUCCESS in 56m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/184d3dd6e0054e75bc183ae0650558d1 : SUCCESS in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/715e7bfcc3bf47e0946c9263c18f137b : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/151a859692b4442c85f791de45493372 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e36b77bb0fa4ff48ef86b92a1e8021f : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1037fe2a65547c7be020cafb9389aa8 : SUCCESS in 8m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc8715e11086471f828dd05b3e33aec2 : FAILURE in 5m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/83ebf490610c477c9012cd6f87f76b7f : SUCCESS in 48m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70452b06de7d4a6a9edbd8a197809a45 : SUCCESS in 1h 01m 47s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"356d90196eebce5f18e434c4833d0d6e393cb8e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-02 22:22:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"63d62be96c3af5490dae7e6dd728c64ca0282ff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 23:38:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d4ed2dd97bb244fbbbdbffb20cd74db3 : SUCCESS in 36m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2d13f82cecd84f2d98a2dbbe86a2bc88 : SUCCESS in 45m 47s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ad9b63977ec54f90ab00bb2155ee8711 : SUCCESS in 21m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/40f5420a3e4b436f947ec97277539a9b : SUCCESS in 32m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/67a76ebf999f4ad8bd8a95c1d4a80b50 : SUCCESS in 1h 02m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/baf55fcb0a6649f7a3a71f8d257879fe : SUCCESS in 57m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f1934dffee9e481fb98a7166b0e554f8 : SUCCESS in 1h 15m 33s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9595603cc2d640d0915e49e7341a0c4f : SUCCESS in 52m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6ffd2ee35f4347a4ba3a22b33ca24a6a : FAILURE in 39m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/db027d4eb07741a1a7c6cdb3cebe9c9d : SUCCESS in 17m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7322a334593b454c9e19dde2208e4496 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/055f5241d8fc495a8d2d21e7e18269aa : SUCCESS in 21m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8f8db94e9cbd4062a1f0e8636060d8c5 : SUCCESS in 52m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a64967965fdf4e709d287a94e3d1770e : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af95d93b6c284f8d8daefc5407c8abac : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa007ff0729c497a927feb98cfbee497 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cce0a5892d88429c8774fd4e426cbf7f : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60514035cca14d6a80783b4c3c7a05db : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2dceee7a2a834bd7a1a4424b7c4cedf7 : SUCCESS in 26m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/50811d81138a4acdbc0e1e7caa1fbab6 : SUCCESS in 27m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/930814ac67e74d2583145b17cbbb83fb : SUCCESS in 54m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b6670beccce04c87ab5fc04142969222 : SUCCESS in 47m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4eec01933f5e4943b5f4ffbba828bb45 : FAILURE in 20m 23s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"9abd25c5a6a18471963e7148e3625219a0d75bea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-03 15:10:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"aea4eaafa6ef135eab5484372391c87c8b981006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 16:55:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ff773668f9a94162b7d37c1ab2f0f4cb : SUCCESS in 44m 00s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4359fe268c754169a9b70e92c427bcbc : SUCCESS in 45m 34s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/db5c1d9aa4fe48868995b390f71c93a1 : SUCCESS in 41m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1d790ac9c66a4c44a7274ca4c0ed2c00 : SUCCESS in 36m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/af7409c1e15b4c118e3808d59fc6e1e0 : SUCCESS in 1h 24m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2f4263b35bce4ee4b092fd0bb7e400f8 : FAILURE in 8m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3ea4b649615e4c028acc1e1c2ae28834 : SUCCESS in 1h 12m 13s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/563701e28e714f75a737172cc7ab3533 : FAILURE in 8m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c749835259e3433aa51fbd5acd69b5d1 : SUCCESS in 1h 21m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6aed546f35d347ccbd8c18ccf2ab258f : SUCCESS in 59m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a5cc9adb5ea64019aaf32de2fff00c4e : FAILURE in 29m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5f8b71e211f74606896e9d28a23d3a1c : SUCCESS in 34m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed0df70d0c714c18b8ec5a089cbb48ed : FAILURE in 29m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca698c6c5514c5da4fcdd2f4b81cbae : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8b3e982699040af945f3fec3e4bfae0 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/243c02df470f44869b557db68f01b778 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c24be2818673404b9c1996e8878e6162 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e73d019ae94d5492aa7501213aa424 : SUCCESS in 9m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/acb819b6a68f43c1b4f755e730c0ca34 : SUCCESS in 27m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f9a853ca5094ea6ab3704cc9abba789 : SUCCESS in 28m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/934c0ac786124b889743a51e6747ddf4 : FAILURE in 22m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2d4236df847b4443b37cdbf3f6ffb3b6 : SUCCESS in 52m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fb525665b9c948ee8c1629b875542cbc : FAILURE in 19m 21s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"216c950a92ff77d615274cf6fa1706afa5dd0a96","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-06 15:22:03.000000000","message":"Patch Set 2: Workflow-1\n\ndo not recheck, there is a valid error here:\n\nERROR: Cannot uninstall \u0027PyYAML\u0027. It is a distutils installed project and thus we cannot accurately determine which files belong to it which would lead to only a partial uninstall.","accounts_in_message":[],"_revision_number":2},{"id":"a9069cdd1410f5253d6cc73e321b8b64e2768321","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-17 15:08:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f569a03f19bd3d4e3938dc7aa285b6924fafccc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:02:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7305e1110a314ea6a132d91e22909060 : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cc87c14a214e4f1c8f9949a63388d446 : SUCCESS in 1h 05m 51s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9d8cc55edd2a4373ae6704ed8bffc981 : SUCCESS in 38m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e361da1fe130408194a877f50b8f1d46 : SUCCESS in 38m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1742eae4b47b4f559d8b497eb18b4e45 : SUCCESS in 1h 37m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/380c9a25db7d48ce9b77906cef4d0453 : SUCCESS in 1h 06m 30s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/679c39f07ead44838c91b0ea04112e7f : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/82808265a69f4e4a8e79f8dc4781fdeb : SUCCESS in 57m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d05d7ca42d74410d88db48d1c851f691 : SUCCESS in 1h 48m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cbaf2e9b00d44c728486c5af5947b2d4 : SUCCESS in 55m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1edd3756568d423e85e10d3ee3ff801f : SUCCESS in 22m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/43da154e182b411fb7fbc01b5ff453c7 : SUCCESS in 22m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f315350f97814d23ad8bd5a94c9df7a5 : SUCCESS in 50m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60fc0581bb764020bd30d0d92d1a3e92 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84b5087ea5fe41628a4d981de5393a33 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/080bea5df13840ae873fc0f8f5c1aa97 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8de6d1ab97614af2ab0df183c999eec0 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bdde429c3df4df5bb237c4021d8b3b6 : SUCCESS in 6m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2b098dbc4a504920ae6678b928d2b552 : SUCCESS in 25m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/91fa49b432c1490faa3a6a38a90b92f3 : SUCCESS in 30m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/215f08272083442c98aa83a0653616c1 : SUCCESS in 49m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e9acc4257264f53abf2f76510e02ed9 : SUCCESS in 1h 06m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4156b1c892584ed78d7d2d56acfce263 : SUCCESS in 48m 09s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"5a01a726a09ba7ca84b5ade834abf4750ae8423f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 17:35:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d9f4c98913f968342c83acaf7faaf7e814368739","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:35:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"476832aeb2ac52354b56f46a778ca9a9ccc7aeb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 19:20:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/08743ad13e9d4b32ac77fbe9dfd66da3 : SUCCESS in 1h 10m 18s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5bc3dccbe8ab401b947e6a9038c10c7e : SUCCESS in 32m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e349be9aeb264a4289f2b4bbafd6a46d : SUCCESS in 1h 13m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9716581520c248b8af134fde059f81fc : SUCCESS in 1h 00m 02s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/83e64e085b5e4c22a286e43b37c00cbc : SUCCESS in 1h 26m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/32f42f7cc34243b7a5c6b7a56cd06079 : SUCCESS in 44m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cefc151ca944f91bac8bf252f989f20 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6d3ab0e82fa4b68a08416ef43364945 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b7939061eb6429d91e821b38ea3069c : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9766818660a48b18489d2f06eab1220 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063c9fa3352b4738b093206fd01b6feb : SUCCESS in 7m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c9259b45d1d42e9afbfc48a158e4c1f : SUCCESS in 27m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/93898d1a082a4ca1a7dfc71e8bee11ba : SUCCESS in 49m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/44e7a0935c684e37ae39f31b2bbd1af0 : SUCCESS in 1h 29m 33s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"fc853234d288b45f2750f9620ff6b6f260776c83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 19:20:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3e5a52dbd8d9b90684612cd439f9cde42d388604","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 19:22:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5125099dd8fb4ca390a7bad415473e0a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"03adc9ddf88e0e66075df254aa0bdf922fbcb74a","revisions":{"dc018f4e8d1a77ff17563069ce796da94dab55c5":{"kind":"REWORK","_number":1,"created":"2021-02-09 09:05:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/02/774602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/774602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/774602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/774602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/774602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/774602/1"}}},"commit":{"parents":[{"commit":"fc0ebe07c2e88ee000caf43d7ba67ceed7d7a842","subject":"Merge \"nova-compute: \u0027pmdk-tools\u0027 is x86-64 only in Ubuntu \u0027bionic\u0027\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc0ebe07c2e88ee000caf43d7ba67ceed7d7a842"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-02-09 09:03:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-02-09 09:05:14.000000000","tz":60},"subject":"sensu: drop Ubuntu support","message":"sensu: drop Ubuntu support\n\nUpstream removed Ubuntu repositories.\n\nTrain/Stein only\n\nChange-Id: Ice125e101a689fd3df9d1cd9a093a026f9747a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc018f4e8d1a77ff17563069ce796da94dab55c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc018f4e8d1a77ff17563069ce796da94dab55c5"}]},"branch":"refs/heads/stable/train"},"03adc9ddf88e0e66075df254aa0bdf922fbcb74a":{"kind":"REWORK","_number":2,"created":"2021-02-17 09:43:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/02/774602/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/774602/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/774602/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/774602/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/774602/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/774602/2"}}},"commit":{"parents":[{"commit":"fc0ebe07c2e88ee000caf43d7ba67ceed7d7a842","subject":"Merge \"nova-compute: \u0027pmdk-tools\u0027 is x86-64 only in Ubuntu \u0027bionic\u0027\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc0ebe07c2e88ee000caf43d7ba67ceed7d7a842"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-02-09 09:03:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-17 09:42:01.000000000","tz":0},"subject":"Fix several issues (Train)","message":"Fix several issues (Train)\n\n* sensu: drop Ubuntu support\n\n  Upstream removed Ubuntu repositories.\n\n  Train/Stein only\n\n* Bifrost: pin IPA builder to 2.4.0\n\n  Train \u0026 earlier\n\n  IPA builder master branch now requires diskimage-builder 3.4.0, but\n  Python 2 is supported only up to DIB 2.38.\n\n  This change works around the issue by pinning IPA builder to 2.4.0,\n  which has a Python 2 compatible dependency on DIB.\n\nChange-Id: Ice125e101a689fd3df9d1cd9a093a026f9747a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03adc9ddf88e0e66075df254aa0bdf922fbcb74a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03adc9ddf88e0e66075df254aa0bdf922fbcb74a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
