)]}'
{"id":"openstack%2Fkolla~775594","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ic638516eb92d24ad247a7866fd1b5e2ac0400388","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1913952-stable/victoria-stable/ussuri","hashtags":[],"change_id":"Ic638516eb92d24ad247a7866fd1b5e2ac0400388","subject":"Make kolla_set_configs open files in binary mode","status":"MERGED","created":"2021-02-15 13:54:12.000000000","updated":"2021-03-01 10:56:10.000000000","submitted":"2021-03-01 10:54:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775594-bug/1913952-stable/victoria-stable/ussuri","cherry_pick_of_change":775593,"cherry_pick_of_patch_set":1,"meta_rev_id":"4b095f412d434c93cac4d430d172b72b580f0d7d","_number":775594,"virtual_id_number":775594,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 10:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2021-02-26 06:38:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-03-01 09:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-02-16 10:26:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-03-01 09:32:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 15:39:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 10:26:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-26 06:38:29.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-03-01 09:32:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"12a022d7dce9808e61fcc0335f2fe1221ca51f61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-15 13:54:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"517f8c412f7c4e449b88b5b883bfb1e822c3e41d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 15:39:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/60c74ff9f0064099ad9f2535a598a05b : SUCCESS in 22m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fc07b295239e43bc93d1c2e06a911941 : SUCCESS in 29m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a3e10ae9fdba4a8f888440e8da9261e3 : SUCCESS in 1h 02m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7389806623e44eeca42bff0d26f78099 : SUCCESS in 59m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c1d9464a8535464f8a00385dbb7fd134 : SUCCESS in 1h 06m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e7fd6a8d4cec4a589e131d6969378b0a : SUCCESS in 25m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f2b5856f7a68488b85e8b2d7e54cdc2c : SUCCESS in 21m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/12a16bdae4004d0690aaed0893c28c32 : SUCCESS in 59m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa8ada5ce1d41cdb8edf06e7f96b019 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8e53484e4764747a09e1023021a8c41 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52f5d9f52cae496b9df5f867f0dff545 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e49d6df0d94a648a6312a0fa7d0573 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a783f1b64f244159a38b8c396ad538a0 : SUCCESS in 3m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d9f3ee0c53c4e898ee2d2207e387ded : SUCCESS in 23m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a3f5c09bc1b4a6da32958a6ca98a4b9 : SUCCESS in 35m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f7196574cdd4ddc83a30a99a76db57d : SUCCESS in 50m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/60aa746c717341dfaddd4c616ff6c5a2 : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d136b177f206400eb8cb9fa1517d60cf : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c432eb9e9dc73faf20d7514abc925347610d6a58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 17:38:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/16dd78ad74e94b21b202610bf17f55a6 : TIMED_OUT in 3h 04m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c9087ffe5374f0886b9e156705615b0 : FAILURE in 26m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03804a59b5d74a5bbbaf86420a804e2b : TIMED_OUT in 3h 02m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/36791dc12a4c402294847e55bb2aa4bf : FAILURE in 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75044f5443f8c7188fa5853205f4aaf9d44167b4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-16 10:26:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"abf4cf8f793d7da6953a914d3a07114e58b70f1f","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-02-26 06:38:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7711e15b40fd581714f21779bedb1c0d8628a0bb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-01 09:32:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15dbdaf955d477d0d4ce8e340ea6c58f8107c5de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 09:33:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"440d351099b46362015ef809b2a7da533d086aac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:54:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f838d2d1fe2b410dad2dfdc3b1672727 : SUCCESS in 45m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/46e0788e5a6a4c6bba8b0deb3af9c2c7 : SUCCESS in 1h 06m 19s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7d993d0e3927404c8788bc84a9f72013 : SUCCESS in 1h 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae63070c9d5d490fb726247534793ddd : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38f3f5826a5b4f3fa5b68c43d845a94e : SUCCESS in 7m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57aba65293b84a7ab5da2f6e73c2610b : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718f0ba2c73b4431b09f8cf471197544 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50506efeaea349dea0e2965a20b41649 : SUCCESS in 4m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3608ab5f74c8434498b9dbe949554527 : SUCCESS in 53m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb3cb146248e49b0ba316fd1beb0430e : SUCCESS in 52m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2114bae2ea54434896c6930afbccc2db : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":1},{"id":"081f42395e46da8d3146e84afb0d4b9cc0f85682","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:54:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4b095f412d434c93cac4d430d172b72b580f0d7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:56:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64281243bc714676871b8d291b59749f : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1f7bca81c3f4acb9ad4ea8c4ef46657 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc5fe2f0eb19601236a54deeeee4ebc0ccd519d9","revisions":{"dc5fe2f0eb19601236a54deeeee4ebc0ccd519d9":{"kind":"REWORK","_number":1,"created":"2021-02-15 13:54:12.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/94/775594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/775594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/775594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/775594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/775594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/775594/1"}}},"commit":{"parents":[{"commit":"0e260fe703d971f2b2ff7bba90bf883da887304f","subject":"Merge \"Drop lower-constraints\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e260fe703d971f2b2ff7bba90bf883da887304f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-31 16:09:04.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-15 13:54:12.000000000","tz":0},"subject":"Make kolla_set_configs open files in binary mode","message":"Make kolla_set_configs open files in binary mode\n\nThis fixes comparisons when files are not Unicode-encoded.\n\nA relevant unit test is included.\nIt can be used as a base for other _cmp_file method unit tests\nif the need arises.\n\nChange-Id: Ic638516eb92d24ad247a7866fd1b5e2ac0400388\nCloses-Bug: #1913952\n(cherry picked from commit 8e3027c5427ecfb89fe48ad38eb8189b1608d0fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc5fe2f0eb19601236a54deeeee4ebc0ccd519d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc5fe2f0eb19601236a54deeeee4ebc0ccd519d9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
