)]}'
{"id":"openstack%2Fkolla~775595","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ic638516eb92d24ad247a7866fd1b5e2ac0400388","project":"openstack/kolla","branch":"stable/train","topic":"bug/1913952-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"Ic638516eb92d24ad247a7866fd1b5e2ac0400388","subject":"Make kolla_set_configs open files in binary mode","status":"ABANDONED","created":"2021-02-15 13:54:36.000000000","updated":"2021-04-01 11:44:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":775594,"cherry_pick_of_patch_set":1,"meta_rev_id":"f308e62ed5ad3e8a122be91783bd24e5cc3ce7e8","_number":775595,"virtual_id_number":775595,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-15 19:06:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-01 09:20:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 16:30:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 09:20:41.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"00b6f97d423fda7f64365abfee562191cc2fd862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-15 13:54:36.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"92cc8be8153938984c915460cdf1b9ccdf128cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 16:30:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/82b27f97fda44957825662b19918c589 : SUCCESS in 40m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9d03bdeb7f2149fa8cff6b6307e5e7d3 : FAILURE in 1h 42m 26s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/875c1656aa5b43f4b2b4c736c133fd9c : SUCCESS in 29m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a0a508c8866e448daf06e2d0c5657086 : FAILURE in 42m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/003a4add16e3488ba21366d02a83b464 : SUCCESS in 1h 10m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/100e387a64ec4506898995d1480972d3 : SUCCESS in 54m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fb8678e466e040ebbe3c280da1b5a2ff : SUCCESS in 1h 05m 38s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b5ef694af18b48e4949efcb95becee3a : SUCCESS in 53m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d32ceffed9c440458c73c2a7d8cf599f : SUCCESS in 1h 19m 01s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d4790f36d7dc4af680e0b9d74cdcefc3 : SUCCESS in 50m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/258fe729650343538e0804f87f4f910e : SUCCESS in 21m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3b1b76ca39154bf3b3101ce83e3d058f : SUCCESS in 22m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f6b06f5ea30e48b385caae141898e7ae : SUCCESS in 54m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa883fd30aa544c091ceb1c29ecc82f6 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d131a68731de435a885401e3473738f6 : FAILURE in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0826d6ae2d44b90882a1a2b1bea506f : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19b5be284c7d44d39e7237ece9f32f73 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db3c85a19ce465b8e3552d159b0e259 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10fdc9bcc7ce4762916535ee6d18b3fd : SUCCESS in 4m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/504d8a5f01d143e8b02ae0454613ea2b : FAILURE in 21m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/75ff1a4f80b74367a647c5473b02d66d : FAILURE in 26m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ecbb7358ec9b4b44b5e15427368805da : SUCCESS in 46m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94e1aacccb6042f4a8732dfde3c34a70 : SUCCESS in 46m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d05a8111708447508a1d36638b043258 : SUCCESS in 1h 00m 27s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"e765cc87c8afe8e969526cdb66dd04e1d45ebf72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-15 16:38:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"da083c907b7320a85a900f6ad109e369b3175f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 19:06:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3bd0ee402eee4ed98f76f82ae0861247 : SUCCESS in 1h 03m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cee98d155d31494d8c7ee141647441b3 : FAILURE in 1h 53m 02s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/602f4afa3e89492a9a7c164337b03dbd : SUCCESS in 53m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/838b56b8d84a4c6691a33882a719ccc0 : FAILURE in 36m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/063cf82217ad4e2eb662c507aaa902b2 : SUCCESS in 1h 07m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/90b1c3a377ac4a5cba4bc2e2550bcefc : SUCCESS in 1h 28m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9bae3fbf70ba466c9b19c34d3d20b412 : SUCCESS in 1h 15m 16s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9e8c91f9af5045c7b94dc306e3d9e985 : SUCCESS in 1h 10m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/78480532ca6545c1bc08beee909f7d41 : SUCCESS in 1h 59m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6a69415d0ce6441ab3d0736cd0af044d : SUCCESS in 50m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/09290ab54ba1482a825ce7914f081848 : SUCCESS in 20m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f3e071af35384fbfb82a67a144bd3ab6 : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/39262570353d4d3c97b25f84c3096f6f : SUCCESS in 44m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/297f2874ba094e1db06371ef89e11132 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21b8fcc02dd7423593aaa07241e325cf : FAILURE in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a89a5ac688d4eadafd404347c928b08 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d20f9651d86d44ce959809ab020a3622 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cff03cde8c742d19475c4ac9fa3cb53 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf359dcedc6d49ff98fd3ba2e2e6b175 : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f1d4574870a47e3bf3c59645d1640f4 : SUCCESS in 52m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d10fe16484ea4a81bec7086ba6e83a28 : FAILURE in 1h 03m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/431f46902cd74799a1af1b7cadc1333e : SUCCESS in 1h 11m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f74b9f8da4d42f79be4b7e05d00d51c : SUCCESS in 53m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6a5517724ab14f33927c8dbf62190b28 : SUCCESS in 1h 05m 55s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"01eabcb0cbf7b856a600fcad9c9ea8c7bcbd4b3e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-15 19:10:23.000000000","message":"Patch Set 2:\n\nhmm, the mock does not patch py2.7 properly","accounts_in_message":[],"_revision_number":2},{"id":"8a1ec879f61794eedc5e494a5beca85dd475b6cb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-01 09:20:41.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2bfedc69066a8b8fb24a6d1352b2d9272d98b7d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 11:06:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/49951277d5dc4ea1ad75f1d5878bc79f : SUCCESS in 43m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/055c7cd161a84c0eaff0c3741337b66d : SUCCESS in 51m 15s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a05fffa55cbf4404a844c0ff2dd01314 : SUCCESS in 31m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e557787fa2ef46b2b0c522d09c0824c5 : SUCCESS in 39m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6b7dace736384ff583ce17e94f785ad4 : SUCCESS in 1h 06m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a3d98619859a42558baf6217f49cb43a : SUCCESS in 1h 06m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/88f931a267c44f5891aa40d987d198c3 : SUCCESS in 1h 18m 42s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/72d60208685f45d993526143154734ff : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c7a92d9f242c4fad893744891aa019a4 : SUCCESS in 1h 30m 20s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c95cf5cbb56e429c860c268f4a152850 : SUCCESS in 49m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0ecc94010b1841479598e7d9c5a3d87f : SUCCESS in 44m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b79c42071f0c4af8b1dbda50ff1390e8 : SUCCESS in 25m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ff515d11873f4ef4a8c076bc5d0c36c4 : SUCCESS in 50m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a97379f6d8b42948fdde16e59cefacd : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da80fa8ad2774d5ba1de243d6a1fed0e : FAILURE in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a4b0454f2694994bfa0b71afdeb5bdb : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0ddb06186f144188cce70711ee20139 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d3ccca9711f44d5bc6947efb0422bc6 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63e96b89b6ed497a8c31cd37add83fcf : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/956833acf283401193f47a6501694f22 : SUCCESS in 26m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3eee65d3e9c8456fbbb30f1d835a51d9 : SUCCESS in 32m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/42352a07a9fe4544aa18a45663076f81 : SUCCESS in 52m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/adb276d6051c4313af877cd501303a7a : SUCCESS in 49m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dc6c381945cf41fdb750c0b230f82011 : SUCCESS in 1h 07m 25s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"f308e62ed5ad3e8a122be91783bd24e5cc3ce7e8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-01 11:44:11.000000000","message":"Abandoned\n\nnot going to fix for Train","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab7727cadbb3741cc1407502b722a578278297d9","revisions":{"a3c88301c0e5a8f67100d1e2c798e389ca8c3daf":{"kind":"REWORK","_number":1,"created":"2021-02-15 13:54:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/775595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/775595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/775595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/775595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/775595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/775595/1"}}},"commit":{"parents":[{"commit":"fc0ebe07c2e88ee000caf43d7ba67ceed7d7a842","subject":"Merge \"nova-compute: \u0027pmdk-tools\u0027 is x86-64 only in Ubuntu \u0027bionic\u0027\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc0ebe07c2e88ee000caf43d7ba67ceed7d7a842"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-31 16:09:04.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-15 13:54:36.000000000","tz":0},"subject":"Make kolla_set_configs open files in binary mode","message":"Make kolla_set_configs open files in binary mode\n\nThis fixes comparisons when files are not Unicode-encoded.\n\nA relevant unit test is included.\nIt can be used as a base for other _cmp_file method unit tests\nif the need arises.\n\nChange-Id: Ic638516eb92d24ad247a7866fd1b5e2ac0400388\nCloses-Bug: #1913952\n(cherry picked from commit 8e3027c5427ecfb89fe48ad38eb8189b1608d0fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3c88301c0e5a8f67100d1e2c798e389ca8c3daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3c88301c0e5a8f67100d1e2c798e389ca8c3daf"}]},"branch":"refs/heads/stable/train"},"ab7727cadbb3741cc1407502b722a578278297d9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-15 16:38:02.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/775595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/775595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/775595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/775595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/775595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/775595/2"}}},"commit":{"parents":[{"commit":"dc018f4e8d1a77ff17563069ce796da94dab55c5","subject":"sensu: drop Ubuntu support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc018f4e8d1a77ff17563069ce796da94dab55c5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-31 16:09:04.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-15 16:38:02.000000000","tz":0},"subject":"Make kolla_set_configs open files in binary mode","message":"Make kolla_set_configs open files in binary mode\n\nThis fixes comparisons when files are not Unicode-encoded.\n\nA relevant unit test is included.\nIt can be used as a base for other _cmp_file method unit tests\nif the need arises.\n\nChange-Id: Ic638516eb92d24ad247a7866fd1b5e2ac0400388\nCloses-Bug: #1913952\n(cherry picked from commit 8e3027c5427ecfb89fe48ad38eb8189b1608d0fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab7727cadbb3741cc1407502b722a578278297d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab7727cadbb3741cc1407502b722a578278297d9"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
