)]}'
{"id":"openstack%2Fkolla~775637","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ice125e101a689fd3df9d1cd9a093a026f9747a3e","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"Ice125e101a689fd3df9d1cd9a093a026f9747a3e","subject":"sensu: drop Ubuntu support","status":"MERGED","created":"2021-02-15 14:14:41.000000000","updated":"2021-02-16 11:50:12.000000000","submitted":"2021-02-16 11:42:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775637","cherry_pick_of_change":774602,"cherry_pick_of_patch_set":1,"meta_rev_id":"dffe34ca7879307f29a932461d62b2ed1754379b","_number":775637,"virtual_id_number":775637,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-16 11:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-02-15 14:27:48.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-02-16 10:39:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-02-16 10:39:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 14:14:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-02-15 17:11:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 10:39:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"7d703df6b626b738ff531562e5b7ee105156fe15","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-15 14:14:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.\n\nThe following files contain Git conflicts:\n* doc/source/matrix_x86.csv\n* kolla/image/build.py","accounts_in_message":[],"_revision_number":1},{"id":"d4a082a849b80be97b1469ac89ff9100610496fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-15 14:17:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"899ea4f29a4b54bf3e8abafb131a77bec95d111d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-15 14:19:07.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"7fec39631369d2e032bcd23565362e105f01ee25","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-15 14:27:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"deceaad63ced9ddfa3ac8b8291671c90766bb850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 17:11:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/daedd9f247dd472a95d8be44ff3aa0a5 : SUCCESS in 38m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b8f855c6393a463599fbb311bb52b2dc : SUCCESS in 44m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6e026e3fd733431f8746c7d10a477dd7 : SUCCESS in 1h 12m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e3e77fe19ae2461098ad5c924af6413a : SUCCESS in 1h 20m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f113e7f108cd4029aecae2b62112d744 : FAILURE in 33m 53s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0c26fd99b0434f2d84a1e1a98fcbecb3 : SUCCESS in 46m 44s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9bfdb4e1b19540c0ba550a6d059be6d1 : SUCCESS in 39m 08s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7a19c51a6e574841bce0e0e0a355df8f : SUCCESS in 39m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c7114e5d7fa4e999858c7d51b14a09b : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3e3b68568964cb2a8f117a6e2d01933 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5299863dce945d89815f6c7e22a00cf : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/144bb4dd2e804f779b9027e43127ba13 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3b8ddfe24248e2b6a0efdba037c87d : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2972b19e73314740bfbcea14cff7e21d : SUCCESS in 35m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b58f9b10f98a46b98160110bf7a28c9c : SUCCESS in 25m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7dda6047356a46e384885e8a562ddb74 : SUCCESS in 51m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c238fecdccf49d88e2bc34570887cb1 : FAILURE in 17m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/14bb05ec750048b9aaeb8d308fd15965 : FAILURE in 25m 04s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"d11f1507819debf0c5e6b088111c5654dec18390","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-16 10:39:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d07fae760398b2855c560e808cd0a7853851170","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 10:39:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b91ef7b87621b25542acb4bb3773deab6764e795","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 11:42:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/89932409aa724a02adcd52b87d36175d : SUCCESS in 47m 52s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b516251fd3d541efb2723a73554c858a : SUCCESS in 37m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be9d4ceadee84745b418b8ec83701488 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8ce12c098ac4b00890b104f5cb7cd97 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06689bb03bc34fcc9f5deda0456e28e9 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a9110f2098048ee8130d8fc957f1693 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97b8011fa9924e47b7e18b959cf63232 : SUCCESS in 3m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ac4b849972a48938b4c36751a9c8aae : SUCCESS in 28m 34s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"e2e38af6a6c35b6244a1e88fd7cb581f2d98fdf2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 11:42:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dffe34ca7879307f29a932461d62b2ed1754379b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 11:50:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/895ab7b793bb457db2c75dc6baf774d4 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b0f74e5d5581a4f802b40d47936e25a54d2c7b6","revisions":{"1aa4c20eea671a3521dc212ca7a20a85a7d617fc":{"kind":"REWORK","_number":1,"created":"2021-02-15 14:14:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/37/775637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/775637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/775637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/775637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/775637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/775637/1"}}},"commit":{"parents":[{"commit":"cde47e352f0f1393e9341776f108e45767672858","subject":"CI: Use Docker proxy for pulls + Ruby fixes + get_pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cde47e352f0f1393e9341776f108e45767672858"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-02-09 09:03:51.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-15 14:14:41.000000000","tz":0},"subject":"sensu: drop Ubuntu support","message":"sensu: drop Ubuntu support\n\nUpstream removed Ubuntu repositories.\n\nTrain/Stein only\n\nChange-Id: Ice125e101a689fd3df9d1cd9a093a026f9747a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aa4c20eea671a3521dc212ca7a20a85a7d617fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aa4c20eea671a3521dc212ca7a20a85a7d617fc"}]},"branch":"refs/heads/stable/stein"},"9b0f74e5d5581a4f802b40d47936e25a54d2c7b6":{"kind":"REWORK","_number":2,"created":"2021-02-15 14:17:39.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/37/775637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/775637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/775637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/775637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/775637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/775637/2"}}},"commit":{"parents":[{"commit":"cde47e352f0f1393e9341776f108e45767672858","subject":"CI: Use Docker proxy for pulls + Ruby fixes + get_pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cde47e352f0f1393e9341776f108e45767672858"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-02-09 09:03:51.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-15 14:17:34.000000000","tz":0},"subject":"sensu: drop Ubuntu support","message":"sensu: drop Ubuntu support\n\nUpstream removed Ubuntu repositories.\n\nTrain/Stein only\n\nChange-Id: Ice125e101a689fd3df9d1cd9a093a026f9747a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b0f74e5d5581a4f802b40d47936e25a54d2c7b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b0f74e5d5581a4f802b40d47936e25a54d2c7b6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
