)]}'
{"id":"openstack%2Fkolla~777062","triplet_id":"openstack%2Fkolla~master~I6889555820adcaee8a86e70b5c3ea31c15e3192c","project":"openstack/kolla","branch":"master","topic":"story/2007938","hashtags":[],"change_id":"I6889555820adcaee8a86e70b5c3ea31c15e3192c","subject":"add openstack-python3-xena-jobs-arm64 job","status":"MERGED","created":"2021-02-23 08:51:21.000000000","updated":"2021-05-22 19:31:42.000000000","submitted":"2021-05-22 19:29:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"777062-story/2007938","meta_rev_id":"2a4ee303489b99a3ab947d3e69861edb3f5c6c16","_number":777062,"virtual_id_number":777062,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-22 19:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":1,"date":"2021-05-21 06:32:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-05-18 08:34:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2021-05-18 08:42:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-05-22 18:19:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-05-22 18:19:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-23 09:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-23 09:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 12:23:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-09 06:22:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2021-04-09 08:07:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-05-11 15:53:35.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-05-18 08:42:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-21 06:32:06.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"81bdc5559b7defa435b3b6a2149b2019fac1cf42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-02-23 08:51:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ff79007f37b30741f058b92e149ff239ec610ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 09:08:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7185eb661afd4725a22827e29f378100 : FAILURE in 5m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82470699f0df4a4896ca2cb26a7149d1 : FAILURE in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f85fa53e63fd15bc11deb040f367e5855e783ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 09:08:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8daf832d184d4cb08a5c62cac4b12d : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c5da57611cc45b3aa38f1e82cda757d : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c1b57fc572e4b3e9197a4dd3d17c4c6 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36f7f56e3bef4e3886bd0f80785bc7d9 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7df5671b8356461aa0b4684648287ed2 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":1},{"id":"89f18a0458ffdab85b607180206f411820606b7f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-23 12:23:33.000000000","message":"Patch Set 1: Code-Review-1\n\nneed to fix the worker tests","accounts_in_message":[],"_revision_number":1},{"id":"92f6c04e0bf53d2faa767e6bfb9158d4b2318a08","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-03-29 05:59:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41ef2cbab9f4c99c46fde2cfc3a16abf4c42abb2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 06:22:44.000000000","message":"Patch Set 1:\n\nKolla testsuite does not have architecture related parts so I do not see a need of such job (especially with limited amount of Ci nodes)","accounts_in_message":[],"_revision_number":1},{"id":"09a381eda8ae1f33c3a9ecd94e17523f72fb62f5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-09 08:07:55.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Kolla testsuite does not have architecture related parts so I do not see a need of such job (especially with limited amount of Ci nodes)\n\nWhy it is failing then? :)","accounts_in_message":[],"_revision_number":1},{"id":"3e64663699d4765354385fb8568d40d6b4cb1a47","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 08:48:42.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Kolla testsuite does not have architecture related parts so I do not see a need of such job (especially with limited amount of Ci nodes)\n\u003e \n\u003e Why it is failing then? :)\n\nhard to tell without logs","accounts_in_message":[],"_revision_number":1},{"id":"a985c341d90a9c88e11ddea8a0e5a06d574fbd29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-11 17:55:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ff39a44d2ad7531f45b1b8302de08eec7653ff55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-11 18:07:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3dd13f83c4644acca95003cc8f3bca5b : FAILURE in 6m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16b46d978acc438f907b6763373d32b3 : FAILURE in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4d8839f81d4e1c548e348be2f992c46838f6b051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-11 18:09:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55819435f525478694f4151a6763445b : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ccb2f9a71f44865a62553dbe5dd0532 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb38d5ce94fe47c881b292711743b4bb : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e7bbceccfc141239a9ae88680b6f50b : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf274b20c814480ebf10bf571c3a1475 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a305d1d9a8aabe7710b10b70c1ae1ce6f825fd63","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-11 19:18:57.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Kolla testsuite does not have architecture related parts so I do not see a need of such job (especially with limited amount of Ci nodes)\n\u003e \u003e \n\u003e \u003e Why it is failing then? :)\n\u003e \n\u003e hard to tell without logs\n\nenjoy the logs: https://18638f8cd3f8b71dfaee-eb83075bad77f107ed9b57803fd20c1f.ssl.cf1.rackcdn.com/777062/2/check-arm64/openstack-tox-py38-arm64/3dd13f8/testr_results.html","accounts_in_message":[],"_revision_number":2},{"id":"2dc01966bd6c069cc3548d845953c6015f08edf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-11 15:53:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0ece5d478f978d858e9118107256e095428e2ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:54:10.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The project template \"openstack-python3-xena-jobs-aarch64\" was not\n  found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-xena-jobs\n        - openstack-python3-xena-jobs-aarch64\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - periodic-stable-jobs\n\n  in \"openstack/kolla/.zuul.d/project.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":3},{"id":"8563b4ac8638c137424faed761d54b38071787de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:54:31.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The project template \"openstack-python3-xena-jobs-aarch64\" was not\n  found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-xena-jobs\n        - openstack-python3-xena-jobs-aarch64\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - periodic-stable-jobs\n\n  in \"openstack/kolla/.zuul.d/project.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":3},{"id":"2c3c085fd75af39106fe0a4d5572d2b185062bc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 07:04:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2e318bd7273a490597b0f3332aee3fbae3693acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 08:47:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9c61fa10fa6f4bdf95cefd6f96656b48 : SUCCESS in 21m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/68fed930d90e44a7b2032cbd5be448dc : SUCCESS in 37m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/085c187a99af45d3b659bf1105e2337b : SUCCESS in 56m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b1642e85c7f84b9eb5a7f4e17b490d14 : SUCCESS in 1h 26m 19s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7f3d780acdb64ff6b62100ddf9d7a80b : SUCCESS in 1h 02m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9568790d1cf24060a626b442a750160c : SUCCESS in 35m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/292e35ba1dd049928e2eff11dbcce4e6 : FAILURE in 16m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fbd08e0621e14345ae9c35ba35566d4f : SUCCESS in 1h 03m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24cbde5e70fe4dc68ab7a9d3e72c3dc2 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70c77d3f571547da9a0d0af6e60ff5e3 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/090b7ccd7d2744e8bf180b9decc06133 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d54f7b3c75ff4de2b21e738389b55df7 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e7727fa3844246af491f0716f01d9b : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84e6bdf6585f41e69667011bc8397e2b : SUCCESS in 23m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee7993562fc9479291889948b689493f : SUCCESS in 38m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b83f968fb17463293a8557095edb94b : SUCCESS in 51m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8853001d7f1144f09a7a15c504d3bb53 : SUCCESS in 56m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/527e0dd12ed54c5691a4a7b3649c3019 : SUCCESS in 1h 39m 06s","accounts_in_message":[],"_revision_number":4},{"id":"d75033ca4b38b31b2073414edded28da956b8040","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:31:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dfd74362512249c79cb3010502d3a4ab : SUCCESS in 1h 54m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dda3e144b3b04e4cb9c630f1648c6da6 : SUCCESS in 1h 56m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c0c992fff2c46f8b1dae700e5a6e9bc : SUCCESS in 2h 13m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c948d82689dd4f6695c5d62f04fd91d9 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd76084841f1452591a2cae6f76a6433 : SUCCESS in 6m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c29fc93b36ba41058115a0d1b71f0a31 : SUCCESS in 1h 51m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b1b4e732486bbf6f0c2b1a7230670f783438086d","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-05-18 08:34:30.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"86535edd23acd5e90104ad1ba32da663bd07fab9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-18 08:42:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"40dec338c5efd5cebf900ae5792e0dc2fed9f0c9","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-05-21 06:32:06.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7d19ed93d400f9ad672e748f52bbec8289312f8c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-22 18:19:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ca83ee7eb482e29237194d93ac50cc580154fdec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 18:20:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0dbb8248354275c7adfe5a50c8399601b204dde2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 19:29:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/171760ff0fe6441db9092ff7568ecf51 : SUCCESS in 47m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/226d58c0a98f4b14ba276b18e9995d56 : SUCCESS in 1h 09m 30s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7fa029e33eda4520af596f451a4c027f : SUCCESS in 53m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c033663467a4e7fbadc8393bdc342ce : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c3764dd4d484453a8da62da82df6111 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a919808ef04feab31429c4b640e6a5 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa95308aa8f2473bb9469e8510d665a1 : SUCCESS in 7m 01s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f44d3f4ddb7442baacfcc901802b78d : SUCCESS in 29m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a13355ecefe4240bde34067bbfdc672 : SUCCESS in 54m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cbc1fc7e69b144a998d7aea028936dee : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":4},{"id":"35a1a37c128e6897300c7641e5584c78c6fd23d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 19:29:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2a4ee303489b99a3ab947d3e69861edb3f5c6c16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 19:31:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ecf0eee7d3841c9bb9c50b66fd7bc69 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ea71584c6545b9f414635040188633046fb7e6fc","revisions":{"02908497d2d0edd43665d51cb5a14f51cd1ee6ff":{"kind":"REWORK","_number":1,"created":"2021-02-23 08:51:21.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/62/777062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/777062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/777062/1"}}},"commit":{"parents":[{"commit":"e9c11009f5f91f12711605d30e6562419fce1e06","subject":"Merge \"doc: describe \u0027allowed-to-fail\u0027 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9c11009f5f91f12711605d30e6562419fce1e06"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-02-23 08:49:34.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-02-23 08:51:18.000000000","tz":480},"subject":"add openstack-python3-wallaby-jobs-arm64 job","message":"add openstack-python3-wallaby-jobs-arm64 job\n\nThis is a non-voting job to validate py3 unittests on ARM64\n\nTask: 41911\nStory: 2007938\nChange-Id: I6889555820adcaee8a86e70b5c3ea31c15e3192c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02908497d2d0edd43665d51cb5a14f51cd1ee6ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02908497d2d0edd43665d51cb5a14f51cd1ee6ff"}]},"branch":"refs/heads/master"},"9fa7121d1b11fc9dfd61dda34e86c63aefc8282e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-11 17:55:48.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/62/777062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/777062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/777062/2"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-02-23 08:49:34.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-11 17:55:48.000000000","tz":0},"subject":"add openstack-python3-wallaby-jobs-arm64 job","message":"add openstack-python3-wallaby-jobs-arm64 job\n\nThis is a non-voting job to validate py3 unittests on ARM64\n\nTask: 41911\nStory: 2007938\nChange-Id: I6889555820adcaee8a86e70b5c3ea31c15e3192c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fa7121d1b11fc9dfd61dda34e86c63aefc8282e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fa7121d1b11fc9dfd61dda34e86c63aefc8282e"}]},"branch":"refs/heads/master","description":"Rebase"},"6df08929f99a37a0707ed646dfb7dd8d036e4ea4":{"kind":"REWORK","_number":3,"created":"2021-05-11 15:53:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/777062/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/777062/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/777062/3"}}},"commit":{"parents":[{"commit":"73c456f85b4e861e4be12282a67ca59f5f319357","subject":"Merge \"docs: update supported OS distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73c456f85b4e861e4be12282a67ca59f5f319357"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-02-23 08:49:34.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-11 15:52:21.000000000","tz":120},"subject":"add openstack-python3-xena-jobs-aarch64 job","message":"add openstack-python3-xena-jobs-aarch64 job\n\nThis is a non-voting job to validate py3 unittests on AArch64\n\nAlso fixed test units to use source install type with Debian as Kolla\ndoes not support Debian/binary on !x86-64\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nTask: 41911\nStory: 2007938\nChange-Id: I6889555820adcaee8a86e70b5c3ea31c15e3192c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6df08929f99a37a0707ed646dfb7dd8d036e4ea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6df08929f99a37a0707ed646dfb7dd8d036e4ea4"}]},"branch":"refs/heads/master"},"ea71584c6545b9f414635040188633046fb7e6fc":{"kind":"REWORK","_number":4,"created":"2021-05-17 07:04:02.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/777062/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/777062/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/777062/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/777062/4"}}},"commit":{"parents":[{"commit":"73c456f85b4e861e4be12282a67ca59f5f319357","subject":"Merge \"docs: update supported OS distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73c456f85b4e861e4be12282a67ca59f5f319357"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-02-23 08:49:34.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 07:03:58.000000000","tz":0},"subject":"add openstack-python3-xena-jobs-arm64 job","message":"add openstack-python3-xena-jobs-arm64 job\n\nThis is a non-voting job to validate py3 unittests on AArch64\n\nAlso fixed test units to use source install type with Debian as Kolla\ndoes not support Debian/binary on !x86-64\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nTask: 41911\nStory: 2007938\nChange-Id: I6889555820adcaee8a86e70b5c3ea31c15e3192c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea71584c6545b9f414635040188633046fb7e6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea71584c6545b9f414635040188633046fb7e6fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
