)]}'
{"id":"openstack%2Fkolla~777142","triplet_id":"openstack%2Fkolla~master~I20f50d236d74b32e9a0454d74376c9f70919accb","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-05 09:13:02.000000000","reason":"Change was abandoned"},"32186":{"account":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"last_update":"2022-04-05 09:13:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I20f50d236d74b32e9a0454d74376c9f70919accb","subject":"add lsscsi due to change in os-brick","status":"ABANDONED","created":"2021-02-23 14:57:30.000000000","updated":"2022-04-05 09:13:02.000000000","total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"47ae4092a496a0487d44d8718afd3e41d9361fdb","_number":777142,"virtual_id_number":777142,"owner":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-04 17:21:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2022-04-05 07:41:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2022-04-04 15:58:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-23 16:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 08:49:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-03-25 10:48:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-04 15:57:59.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-04-04 15:58:33.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"cc1b1b95a3af31c224776fb61d85344769451e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"date":"2021-02-23 14:57:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0eba0eeea249b5111037e46a543e3fc739866ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 16:20:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4a5cf53679974cd5b490d5390c181531 : SUCCESS in 28m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/34c91dcedd744153afa092966954d4b3 : SUCCESS in 46m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9fc29a3351a54eb6bd342462ec820fb2 : SUCCESS in 1h 03m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6795246c2eaf4303835a12aa376f8cc1 : SUCCESS in 1h 13m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4e4238efd4bc42948b12d758afa342df : SUCCESS in 1h 11m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eba257ae10574c7ba6510e186c77446e : SUCCESS in 37m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/942af9e28dfc4dec91fca1ba5dc6a561 : SUCCESS in 20m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4be16c804d7c4c5f8a66e710b4e2549a : SUCCESS in 50m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4723a86bd494a7cb2d86b550b536a24 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90e2c4c0fc048d686f556940c844261 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f38ee72b2604eddbce01ac03be22639 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e40c48b75c414d15b34ecf1ce12f5b05 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3f0b2522924557ac7eb656fbc1e27f : SUCCESS in 5m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1cb94db2f51a4982bf8036be3bed3ea4 : SUCCESS in 23m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/56c1c789cc344e19b09839c79ec520d8 : SUCCESS in 43m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/494d50910f5349d888ea9e065f345702 : SUCCESS in 52m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb41653de626483784a5a240a138df4f : SUCCESS in 1h 03m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cfad91ac2c584650886ab72ec0b8375c : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3098f0d04f6258f645f9c360147d29efaab245da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 16:52:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d540623627004a489f60162212f91a21 : SUCCESS in 1h 11m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7a96c9d0e364ee79266e9dfa5ddfd79 : SUCCESS in 1h 05m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/44ac6cd248cc42e7961df106cfe1c32c : SUCCESS in 1h 48m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/208432f58e5d427893e1ab5843117367 : SUCCESS in 1h 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d9ae47552806b251ff39e1d7946cad6ae2a633b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-24 08:49:21.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nPlease add a release note.\n\nhttps://docs.openstack.org/kolla/latest/contributor/release-notes.html","accounts_in_message":[],"_revision_number":1},{"id":"604140020e0221e64aa7cfac25e3549fa648d90b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"date":"2021-02-24 12:57:36.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7a06171e33bda96652ae8170525fe95d5960ba80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"date":"2021-02-24 13:02:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ff227ecb7289b17e250d97f964258bfed10a011c","author":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"date":"2021-02-24 13:04:36.000000000","message":"Patch Set 3:\n\n* Release note added\n* all lists are now in alphabetical order\nThank you","accounts_in_message":[],"_revision_number":3},{"id":"06f74cddd6086392eeeca1cb11a9da9ffe68b037","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 14:05:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/900b152b899b4503a53442768f3df0c8 : FAILURE in 55m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/63f7b9b67fc14415b0af4e1ccb8204b2 : FAILURE in 44m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d1b90a962ef343229faef2f152f85b0d : FAILURE in 33m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7239926f15c940e8b6db456980ad24ab : FAILURE in 55m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb4a5aad4875f95549f7f82949f6b53e4420b90b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 14:10:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/995b393096e4439c8e0e82a67ab3af16 : SUCCESS in 30m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e03d8d2c620c4368821061e1cf22cf97 : FAILURE in 48m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/99e77514fe1942dca890745a062a983a : FAILURE in 28m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/564e240718d247a1b74166af7506034a : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fa61b980ff5b4b3b92535d731b593673 : FAILURE in 28m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/80c0b4a1b75843dbb2d0e0ed3fa46989 : FAILURE in 41m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8a1a5ae6a34e499daa7a8fe6ac5c5ef6 : SUCCESS in 28m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4dc063e1e6624cd4bbf303095e9c2133 : FAILURE in 24m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bb6f20065cf4604b68845c10f2bebe6 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53847e73cb9640348737392e5e22c789 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a5f7bb880244b4a9d90f274888a7af9 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20bfa7beaf5b43998a09ee93a10b5573 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94fb41a584034659828a1ad70636f628 : SUCCESS in 7m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f8735acb7b948379224339809d0da75 : SUCCESS in 30m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f7ed226e5b34721b026e15fa30cee6b : FAILURE in 34m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4ef9ca9152440d7ac61e0782638bbfa : FAILURE in 22m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/309c57d7a5b44680a537cd8f80e7965f : SUCCESS in 1h 01m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a2c2897d67fb43349fa153a64dd7697e : FAILURE in 17m 13s","accounts_in_message":[],"_revision_number":3},{"id":"dac47e1044635515083f5d56586369bcff723760","author":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"date":"2021-02-25 09:43:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9fdd5b4dff0689409dbf720ec3457ae9a969a9ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 10:49:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9e856ca68b748c881b02a965e1b0d12 : FAILURE in 1h 00m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9a975a909754f1996053476fc11abfd : FAILURE in 58m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1cbd80f1d7640e6ab92f6bb77aea559 : FAILURE in 37m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1353f2990eab45379eded72c6289467b : FAILURE in 1h 00m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"97559f05c77d1aafbac86b955fdc19b7849074b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 11:02:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1a4b1f87374b433d94d37d350dbece3d : SUCCESS in 35m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/65516abbec11481d925b30636f258f21 : FAILURE in 40m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8885c4c532cc412a9a77f836ef9d0566 : FAILURE in 35m 32s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/009564ea240b4e9c89a0f97c2cf9da63 : SUCCESS in 1h 14m 06s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5c01d00c99dc44ceaa995dab50447891 : FAILURE in 35m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6387d103d8b64b128d5bf061166b4b50 : FAILURE in 50m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3efcf962399b4a68bb4201f96779a9f5 : SUCCESS in 36m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c891210383ac48618555401a348401f2 : FAILURE in 33m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363394d6d16b4a89960dafbe316bebbd : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0aba321d260e4774918042105faac842 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5fb63ca426e4ae78f487776ab70507a : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/532989cf1894499aa4014513b9bafc3c : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61a875cfbbc6408a9d9c03a43ba6bb53 : SUCCESS in 6m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f4061a4d46764f5590f07803d4cd84f3 : SUCCESS in 21m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/41ffe3a4416d45b480dcd5101012959d : FAILURE in 36m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4bfa2afab4f4bff97d68c918a3ab92d : FAILURE in 20m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1f76aa43f10a40b9a7ae9ead16becc74 : SUCCESS in 1h 03m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/144f2eda60344753b376db0604b32250 : FAILURE in 15m 37s","accounts_in_message":[],"_revision_number":3},{"id":"a9160980c334eea433c5c98008faa67704651d22","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 10:47:26.000000000","message":"Topic stable/victoria removed","accounts_in_message":[],"_revision_number":3},{"id":"4e3ba159636fd0a10ecd4e1a34e856386e405b3c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 10:47:57.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e * Release note added\n\u003e * all lists are now in alphabetical order\n\u003e Thank you\n\nI don\u0027t see a release note. Did you forget to add it?","accounts_in_message":[],"_revision_number":3},{"id":"240699ef487515d4d351c326d9420c71769eae0d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 10:48:02.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"8a513533e7dc739e031b6e5be799fccabb5a2fe7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-04 15:57:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9c6030f5e67d796de9beaf13dd4dbfbdeb9940d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-04 15:58:03.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c4b9d64919f2b2aed8aa4bfd48f75e4beabdd8a9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-04 15:58:33.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c177e61f9c08eb516e7ff4022674fb1db1cca1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 17:21:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ec9f65f6870943c98c195c136ccab44e : SUCCESS in 25m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b5a57260933e4f1f9345465f6455c227 : SUCCESS in 40m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/66a440756ac94b97b8a0a3ab86057167 : SUCCESS in 1h 10m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5f8e73dd4d484e87b249bb89dda45d9f : SUCCESS in 56m 10s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4c7cc6c952d7487ab774f80521938821 : SUCCESS in 1h 12m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4215a508126c4a0ba2f2470c0ab3f5a7 : SUCCESS in 26m 52s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a56b9b36384540e799163ba214700fc4 : SUCCESS in 15m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1ce7b6d1f16f494d87e3b038cb57f511 : SUCCESS in 49m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b4d857faa4784bcea6d1be0afa5ac504 : SUCCESS in 1h 04m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fc874b8749344fc960912b3e98bb064 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a10d620ed06405d8337f1a6d00b3964 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48b4a11a831a44d5b2094d0289d1038e : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d29b15ffe54c39a720c998c8bee5c1 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/15b41487be7e48878fd23a1129e3ec10 : SUCCESS in 16m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a5df83a927ee410797de33ccc2cd68dc : SUCCESS in 23m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/185681df69404cc2a121fc0fe1ff47fe : SUCCESS in 45m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b21b7da674a4e9e85783e6f7185b6a8 : FAILURE in 40m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/13c3331684f44895b6bf9802efe5d853 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":4},{"id":"8d9d179983c20823161c52bb93585bd15dc36c05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 17:42:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8cabd2cf90ef4b62a82936b7dfcb92c9 : SUCCESS in 1h 31m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f75d5caa0db4aaebece240cbc88bf26 : SUCCESS in 1h 07m 51s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1667c9e43cfb4e5cb2a6590b6fb9b68c : SUCCESS in 1h 41m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a138609da17452d8e4feb6b7e961dae : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab07d7d6f49a47d5b935a1ae8232159e : SUCCESS in 6m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7608147abf3418ea0a109c4061f35e7 : SUCCESS in 1h 11m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"30f0394726156265b478a54f58f5b88c1fe64212","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-05 07:41:27.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"47ae4092a496a0487d44d8718afd3e41d9361fdb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:13:02.000000000","message":"Abandoned\n\nalready done","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dac58441f5dc186db56c7c2a551d2d48f79954da","revisions":{"65a9d0f522f4e675660681a07aac48200a9e5fa5":{"kind":"REWORK","_number":1,"created":"2021-02-23 14:57:30.000000000","uploader":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"ref":"refs/changes/42/777142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/777142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/777142/1"}}},"commit":{"parents":[{"commit":"e9c11009f5f91f12711605d30e6562419fce1e06","subject":"Merge \"doc: describe \u0027allowed-to-fail\u0027 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9c11009f5f91f12711605d30e6562419fce1e06"}]}],"author":{"name":"cyrille.babon","email":"cyrille.babon@free.fr","date":"2021-02-23 14:56:39.000000000","tz":0},"committer":{"name":"cyrille.babon","email":"cyrille.babon@free.fr","date":"2021-02-23 14:56:39.000000000","tz":0},"subject":"add lsscsi due to change in os-brick","message":"add lsscsi due to change in os-brick\n\nChange-Id: I20f50d236d74b32e9a0454d74376c9f70919accb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a9d0f522f4e675660681a07aac48200a9e5fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a9d0f522f4e675660681a07aac48200a9e5fa5"}]},"branch":"refs/heads/master"},"30e0d04c4e2e042378182c6b2448737f47e80135":{"kind":"NO_CHANGE","_number":2,"created":"2021-02-24 12:57:36.000000000","uploader":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"ref":"refs/changes/42/777142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/777142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/777142/2"}}},"commit":{"parents":[{"commit":"e9c11009f5f91f12711605d30e6562419fce1e06","subject":"Merge \"doc: describe \u0027allowed-to-fail\u0027 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9c11009f5f91f12711605d30e6562419fce1e06"}]}],"author":{"name":"cyrille.babon","email":"cyrille.babon@free.fr","date":"2021-02-23 14:56:39.000000000","tz":0},"committer":{"name":"cyrille.babon","email":"cyrille.babon@free.fr","date":"2021-02-24 12:56:53.000000000","tz":0},"subject":"add lsscsi due to change in os-brick","message":"add lsscsi due to change in os-brick\n\nChange-Id: I20f50d236d74b32e9a0454d74376c9f70919accb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e0d04c4e2e042378182c6b2448737f47e80135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e0d04c4e2e042378182c6b2448737f47e80135"}]},"branch":"refs/heads/master"},"18e670025f1017db6229f49672551c7dfe30c12f":{"kind":"REWORK","_number":3,"created":"2021-02-24 13:02:49.000000000","uploader":{"_account_id":32186,"name":"cyrille babon","email":"cyrille.babon@free.fr","username":"cyrille.babon"},"ref":"refs/changes/42/777142/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/777142/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/777142/3"}}},"commit":{"parents":[{"commit":"e9c11009f5f91f12711605d30e6562419fce1e06","subject":"Merge \"doc: describe \u0027allowed-to-fail\u0027 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9c11009f5f91f12711605d30e6562419fce1e06"}]}],"author":{"name":"cyrille.babon","email":"cyrille.babon@free.fr","date":"2021-02-23 14:56:39.000000000","tz":0},"committer":{"name":"cyrille.babon","email":"cyrille.babon@free.fr","date":"2021-02-24 13:02:30.000000000","tz":0},"subject":"add lsscsi due to change in os-brick","message":"add lsscsi due to change in os-brick\n\nChange-Id: I20f50d236d74b32e9a0454d74376c9f70919accb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18e670025f1017db6229f49672551c7dfe30c12f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18e670025f1017db6229f49672551c7dfe30c12f"}]},"branch":"refs/heads/master"},"dac58441f5dc186db56c7c2a551d2d48f79954da":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-04-04 15:58:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/42/777142/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/777142/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/777142/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/777142/4"}}},"commit":{"parents":[{"commit":"a8fcb02f40c94c2101edff9ec4a3101d1d14fe97","subject":"cloudkitty-api: make sure that we install packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8fcb02f40c94c2101edff9ec4a3101d1d14fe97"}]}],"author":{"name":"cyrille.babon","email":"cyrille.babon@free.fr","date":"2021-02-23 14:56:39.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 15:58:03.000000000","tz":0},"subject":"add lsscsi due to change in os-brick","message":"add lsscsi due to change in os-brick\n\nChange-Id: I20f50d236d74b32e9a0454d74376c9f70919accb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dac58441f5dc186db56c7c2a551d2d48f79954da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dac58441f5dc186db56c7c2a551d2d48f79954da"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
