)]}'
{"id":"openstack%2Fkolla~777992","triplet_id":"openstack%2Fkolla~master~I6ef344d463333e5cdc3a0850c8e068218779fdbd","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I6ef344d463333e5cdc3a0850c8e068218779fdbd","subject":"neutron-server: make it build again","status":"MERGED","created":"2021-03-01 13:14:05.000000000","updated":"2021-03-02 10:31:04.000000000","submitted":"2021-03-02 10:27:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"777992","meta_rev_id":"191db8e4b843fcbeee176ae63aad1e30778be53f","_number":777992,"virtual_id_number":777992,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 10:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-03-01 18:02:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-03-02 09:14:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-02 09:10:10.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-03-02 09:14:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-02 09:10:10.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-03-01 14:55:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-01 13:34:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-03-01 14:53:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-01 14:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 18:02:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-02 09:10:10.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-03-02 09:14:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"c9a478fa285b86e314e70dffd4dd6737082cbca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-01 13:14:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"561ad9d4f88c5eec2f652c96c92693362ec50d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-01 13:31:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d094e1bb111eeaa2c38ddba34620ffba58f0bd7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 13:34:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c003427bc3c1f089c76a9f060288950af79d036a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-01 13:48:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"723edc9a853ab4424b2412e8ebc467b46b73eccc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:53:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/43d229627d394461a63915fcbf90cf06 : SUCCESS in 1h 14m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/781be6f984be4c4a82cabf1b251c1e07 : SUCCESS in 1h 09m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1b4d2f47b614ac1a95326181b0a3185 : FAILURE in 55m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8dab50daa9fb4bbe86ec31cf4d07b5b2 : SUCCESS in 1h 06m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f4bdbb25be3465614d5cfa46c58f8fe8db41467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:54:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/873b1378c0b74327ba3a2547c50f92b5 : SUCCESS in 46m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fd44122ca81048c9bdcafdc2b0193de6 : SUCCESS in 1h 11m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a571d3f8aa98471f9fef1993e8be3c89 : SUCCESS in 1h 15m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c575dbd04f16497b8f698525b0c1e58c : SUCCESS in 1h 01m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/52c27b3f7e454fe6b1295821cc4efbb7 : SUCCESS in 1h 13m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/25a3311e9d244ed692126dfd515df81f : SUCCESS in 1h 07m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2da875ecba4b43a68147300fa1d03c29 : SUCCESS in 33m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2562d12b5d9c4de6b8cd7cdd422b74b8 : SUCCESS in 59m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/615296a0a6d34b01ba5350c07734a4d6 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/384dfaf82a014910b66d28ba0e4d8c4b : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/977abbf0347c4d1c9c80dca02e4adf81 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/918fba35a4934414ba019eb78103ad10 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef105dc5a17a4c3081cae650e4338b29 : SUCCESS in 6m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c14ede9a27f490ab27b030e7bb774e5 : SUCCESS in 43m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d4c3a52019a43e0ad4c6dca18749a23 : SUCCESS in 48m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e96049e4e8642cab86c60b1de6f9ccb : SUCCESS in 53m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e8e495341ff4c70b584d2f07848d922 : SUCCESS in 52m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5adc74ce29ad424ab6257e83459e3c80 : SUCCESS in 1h 06m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d8d9178590f44bd3fe9c69ca173d7c3a1744a936","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-01 14:55:21.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"1ff53d2e6624649df344c63b42f20354ea3c92de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 15:04:15.000000000","message":"Patch Set 2:\n\nShould we add vmware-nsxlib to neutron_server_pip_packages?","accounts_in_message":[],"_revision_number":2},{"id":"8ce7a3ae8c4e4bfd3acc13abc595aed1b10c7cc8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-01 16:43:59.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Should we add vmware-nsxlib to neutron_server_pip_packages?\n\nNo need for it. It is listed as dependency of vmware-nsx so gets fetched by pip already.","accounts_in_message":[],"_revision_number":2},{"id":"f0cd23633b20ce739055265e8d801c661d79e726","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 18:02:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"41f5ec4a998ad112002b7d5c13d9963826e5a444","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-03-02 09:10:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"200bfffb2eb7e1d6c23e0aae93fcd5e8755bb245","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 09:10:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6143c8765beea5b9bc8bc826cb35bbb93178b6e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-02 09:14:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eddf88728b0101b098f3f0b3090b866572313f3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:27:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cbf6b4a53e2c44b3ae36cce65adcb5c9 : SUCCESS in 45m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2fc5a42c2dcd4d7996f5ebf90bc9f111 : SUCCESS in 1h 15m 07s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a25f85034d724fbc9bfaae42d79ad7c3 : SUCCESS in 1h 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccef6064d0fb427ea4f30b26ede11f45 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3482e1bfae24573957d03faf1f15c88 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/389b5361710744109d5dfcc0aa38547f : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/618642b132744c499a1c095bbb2d27e0 : SUCCESS in 8m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b5403f067764b39a82ad497681d417b : SUCCESS in 43m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fcd56b284e9e4c47ae2eb552997ecb02 : SUCCESS in 41m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b006fd83fa8c467581120e65d9a069f6 : SUCCESS in 49m 11s","accounts_in_message":[],"_revision_number":2},{"id":"2b724f28a926b8154f5f8d92258976bd15165a20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:27:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"191db8e4b843fcbeee176ae63aad1e30778be53f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:31:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96738e9675e94051b2db879b211d0895 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d70e3f0cc3681c6f786ccbe076cd6a028998857","revisions":{"b48958d8c289ae70af31c6d5d2e7c0cad1a34670":{"kind":"REWORK","_number":1,"created":"2021-03-01 13:14:05.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/92/777992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/777992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/777992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/777992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/777992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/777992/1"}}},"commit":{"parents":[{"commit":"e9c11009f5f91f12711605d30e6562419fce1e06","subject":"Merge \"doc: describe \u0027allowed-to-fail\u0027 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9c11009f5f91f12711605d30e6562419fce1e06"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-01 13:11:16.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-01 13:13:56.000000000","tz":60},"subject":"neutron-server: stop install vmware-nsxlib as a plugin","message":"neutron-server: stop install vmware-nsxlib as a plugin\n\nvmware-nsx is a plugin and we install it already\nvmware-nsxlib can be fetched from Pypi\n\nSolves:\n\nChange-Id: I6ef344d463333e5cdc3a0850c8e068218779fdbd\nINFO:kolla.common.utils.neutron-server:The conflict is caused by:\nINFO:kolla.common.utils.neutron-server:    The user requested vmware-nsxlib 15.0.8.dev129 (from /plugins/vmware-nsxlib-15.0.8.dev129)\nINFO:kolla.common.utils.neutron-server:    vmware-nsx 15.0.1.dev424 depends on vmware-nsxlib\u003e\u003d17.0.0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b48958d8c289ae70af31c6d5d2e7c0cad1a34670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b48958d8c289ae70af31c6d5d2e7c0cad1a34670"}]},"branch":"refs/heads/master"},"8d70e3f0cc3681c6f786ccbe076cd6a028998857":{"kind":"REWORK","_number":2,"created":"2021-03-01 13:31:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/92/777992/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/777992/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/777992/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/777992/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/777992/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/777992/2"}}},"commit":{"parents":[{"commit":"e9c11009f5f91f12711605d30e6562419fce1e06","subject":"Merge \"doc: describe \u0027allowed-to-fail\u0027 option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9c11009f5f91f12711605d30e6562419fce1e06"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-01 13:14:59.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-01 13:30:59.000000000","tz":60},"subject":"neutron-server: make it build again","message":"neutron-server: make it build again\n\n1. stop pinning networking-hyperv\n\nPike release was long time ago so maybe it is time to unpin it?\n\ncommit 6dc8e978462fd08122133e273ae956721997d3af\nAuthor: Dan Ardelean \u003cdardelean@cloudbasesolutions.com\u003e\nDate:   Mon Aug 28 12:56:25 2017 +0300\n\n    Change networking-hyperv version for Pike\n\n2. stop install vmware-nsxlib as a plugin\n\nvmware-nsx is a plugin and we install it already\nvmware-nsxlib can be fetched from Pypi\n\nSolves:\n\nINFO:kolla.common.utils.neutron-server:The conflict is caused by:\nINFO:kolla.common.utils.neutron-server:    The user requested vmware-nsxlib 15.0.8.dev129 (from /plugins/vmware-nsxlib-15.0.8.dev129)\nINFO:kolla.common.utils.neutron-server:    vmware-nsx 15.0.1.dev424 depends on vmware-nsxlib\u003e\u003d17.0.0\n\nChange-Id: I6ef344d463333e5cdc3a0850c8e068218779fdbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d70e3f0cc3681c6f786ccbe076cd6a028998857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d70e3f0cc3681c6f786ccbe076cd6a028998857"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
