)]}'
{"id":"openstack%2Fkolla~778195","triplet_id":"openstack%2Fkolla~master~I095ca6f34cc7c7bca485e9ec019cc52aeb8e3ff4","project":"openstack/kolla","branch":"master","topic":"fix_curl_302","hashtags":[],"change_id":"I095ca6f34cc7c7bca485e9ec019cc52aeb8e3ff4","subject":"Fix error when github return 302","status":"MERGED","created":"2021-03-02 14:29:05.000000000","updated":"2021-03-04 14:51:08.000000000","submitted":"2021-03-04 14:16:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778195-fix_curl_302","meta_rev_id":"a914989fb1a71dea0b31b42a48de444159fdd4a9","_number":778195,"virtual_id_number":778195,"owner":{"_account_id":32690,"name":"Sergey Chekanov","email":"chekanovsm@gmail.com","username":"s-chekanov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 14:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-03-04 11:58:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-03-03 10:52:44.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-03-04 11:58:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 15:52:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-02 16:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 10:52:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-04 11:58:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"a2bedb2bc072e2e259808c819371820c0513d9f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32690,"name":"Sergey Chekanov","email":"chekanovsm@gmail.com","username":"s-chekanov"},"date":"2021-03-02 14:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9be30e5d4bd744f7c75bbcc73d9a29fc1c1ee96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 15:52:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f4441cfad11b40599e3378d332da46b4 : SUCCESS in 1h 18m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2fa59b0318454af79d56d61a6355f385 : SUCCESS in 1h 15m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7165d22b0b5430fb739aa16fbc6b6cb : FAILURE in 1h 03m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/22d070c715eb4bb7989e0993169f3f89 : SUCCESS in 1h 07m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd54d2a42986897401a10c9572173e1b6bdb06e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:27:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/85a15fdd7c3340e386e1dd179f57d533 : SUCCESS in 41m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3ee6cc9113e0435abdfc76d45ed90349 : SUCCESS in 56m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/805f2a3c03d4438383e33b5d74403878 : SUCCESS in 1h 56m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/586bf2b48de74ad4a3d6e02fb4359e94 : SUCCESS in 1h 03m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2f06a92030324abab7c43384c857cced : SUCCESS in 1h 35m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/efffa9605c9a4b26a5d0f9d675f0bb0f : SUCCESS in 1h 03m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/408a482357bd48e08cdbbdfdc2cd234a : SUCCESS in 1h 08m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/35c71e54941b4d479da4910ad6f21ea1 : SUCCESS in 57m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4a9bbead034467aedd4af36297b3f4 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/960ecd08547f46e6bf30cb828da64d98 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6aaeb3ef4ca4d40a72d63856626ddd1 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf38217f73c54a1eaf3b77a1bef1ede7 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0af003eed24c5db2c302711720e67a : SUCCESS in 5m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/52ceb648c0e14a05b899ce6ca681de48 : SUCCESS in 38m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a3bcaea55f84cfcb8d9290e07d95767 : SUCCESS in 1h 02m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ae6a9c8f93f47c091db0a063c460873 : SUCCESS in 1h 09m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e82efabbb871498c891ec3ab5b41c958 : SUCCESS in 1h 07m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fab096d518f149cf99c23d5c77d79318 : SUCCESS in 1h 07m 35s","accounts_in_message":[],"_revision_number":1},{"id":"92e381dd23fb5b536ede4641a991d6a167914eaf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-03 10:52:44.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"6404105e9111aca34ba86e6004dfb06f0e379c2f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-04 11:58:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"59bc9785401d3c05d20234bfcdac5f904386de57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 11:58:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"081680db80bf5f45f966843498bd772f0cfe4839","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:16:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/23b2e9a8ffce4944965679636a127c06 : SUCCESS in 43m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/77a953ad22004675b515c4a14bec68a0 : SUCCESS in 1h 12m 22s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bf5fa7824c5a4a7c96efe42dda739d7f : SUCCESS in 2h 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea5428595971403dba52c1c28914e241 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/890fa81c59634c5f89079919ffb09c7f : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f81e4e049a1f426398e6b90af143c32f : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89aa8bc272ba41dc97e3e867ef05a6d8 : SUCCESS in 9m 19s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32574c0ae7ec45dfb7fbacfd898cbe69 : SUCCESS in 37m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5db50813e2340369006b76d021b3518 : SUCCESS in 56m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b9aea67b55db472b86b36092f5155087 : SUCCESS in 1h 18m 22s","accounts_in_message":[],"_revision_number":1},{"id":"137095aeec666bfdc44d860687b02d6a6503acde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:16:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"df58c8d4792d521d4de3cc908697df32e3d0b945","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:19:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3988d40a5b044f0b1df4dd0d706e812 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a914989fb1a71dea0b31b42a48de444159fdd4a9","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-04 14:51:08.000000000","message":"Created a revert of this change as I066d747400ede30037ff324ad55430af7c35dc18","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61def281d077f88051d1b1a14dfdbfabac036c8c","revisions":{"61def281d077f88051d1b1a14dfdbfabac036c8c":{"kind":"REWORK","_number":1,"created":"2021-03-02 14:29:05.000000000","uploader":{"_account_id":32690,"name":"Sergey Chekanov","email":"chekanovsm@gmail.com","username":"s-chekanov"},"ref":"refs/changes/95/778195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/778195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/778195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/778195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/778195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/778195/1"}}},"commit":{"parents":[{"commit":"8d70e3f0cc3681c6f786ccbe076cd6a028998857","subject":"neutron-server: make it build again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d70e3f0cc3681c6f786ccbe076cd6a028998857"}]}],"author":{"name":"Sergei Chekanov","email":"chekanovsm@gmail.com","date":"2021-03-02 14:23:23.000000000","tz":180},"committer":{"name":"Sergei Chekanov","email":"chekanovsm@gmail.com","date":"2021-03-02 14:28:59.000000000","tz":180},"subject":"Fix error when github return 302","message":"Fix error when github return 302\n\n-L for curl: If the server reports that the requested page has moved to a different location (indicated with a Location: header and a 3XX response code), this option will make curl redo the request on the new place\n\nIt fixes build error for me:\nINFO:kolla.common.utils.prometheus-v2-server: ---\u003e Running in 693fd4095c77\nINFO:kolla.common.utils.prometheus-v2-server:curl (https://github.com/prometheus/prometheus/releases/download/v2.23.0/prometheus-2.23.0.linux-amd64.tar.gz): response: 302, time: 0.166583, size: 640\nINFO:kolla.common.utils.prometheus-v2-server:\nINFO:kolla.common.utils.prometheus-v2-server:gzip: stdin: not in gzip format\nINFO:kolla.common.utils.prometheus-v2-server:\nINFO:kolla.common.utils.prometheus-v2-server:tar: Child returned status 1\nINFO:kolla.common.utils.prometheus-v2-server:tar: Error is not recoverable: exiting now\n\nChange-Id: I095ca6f34cc7c7bca485e9ec019cc52aeb8e3ff4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61def281d077f88051d1b1a14dfdbfabac036c8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61def281d077f88051d1b1a14dfdbfabac036c8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
