)]}'
{"id":"openstack%2Fkolla~780537","triplet_id":"openstack%2Fkolla~master~I3fac63b95a026b951f0571e3cafcfeba0038bf93","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I3fac63b95a026b951f0571e3cafcfeba0038bf93","subject":"WIP: Fix Ussuri builds due to RDO Virtualization GPG key","status":"ABANDONED","created":"2021-03-15 10:33:04.000000000","updated":"2021-03-15 11:28:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"986d1d2439314207f9167b678b2c0417aae5eb00","_number":780537,"virtual_id_number":780537,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-15 11:25:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 11:25:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1481da6e673b700266e4ee5e77cbfe06314ab1bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-15 10:33:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17f7c0369ee32c1cbeb18fde0b244ce8abd14945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:25:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b1d174a52cca40398ce8592caf5385bb : FAILURE in 33m 32s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/be391661ae234e1cab96c62f5d45d1e0 : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":1},{"id":"986d1d2439314207f9167b678b2c0417aae5eb00","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-15 11:28:16.000000000","message":"Abandoned\n\nargh wrong branch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91d92a360ba90c16216878aed72901402ac4a753","revisions":{"91d92a360ba90c16216878aed72901402ac4a753":{"kind":"REWORK","_number":1,"created":"2021-03-15 10:33:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/780537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/780537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/780537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/780537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/780537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/780537/1"}}},"commit":{"parents":[{"commit":"92da935331b4ca9ca29f392455a71afd0ee7f198","subject":"Merge \"Upgrade from ELK6 to ELK7 FOSS release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92da935331b4ca9ca29f392455a71afd0ee7f198"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-15 10:32:34.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-15 10:32:53.000000000","tz":0},"subject":"WIP: Fix Ussuri builds due to RDO Virtualization GPG key","message":"WIP: Fix Ussuri builds due to RDO Virtualization GPG key\n\nChange-Id: I3fac63b95a026b951f0571e3cafcfeba0038bf93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d92a360ba90c16216878aed72901402ac4a753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d92a360ba90c16216878aed72901402ac4a753"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
