)]}'
{"id":"openstack%2Fkolla~781130","triplet_id":"openstack%2Fkolla~master~I7c1f0c5fc643678b29f963b6769d8f900a03b726","project":"openstack/kolla","branch":"master","topic":"kolla-quay.io","hashtags":[],"change_id":"I7c1f0c5fc643678b29f963b6769d8f900a03b726","subject":"[CI] Publish images to quay.io on a daily basis","status":"MERGED","created":"2021-03-17 17:05:36.000000000","updated":"2021-04-12 03:52:42.000000000","submitted":"2021-04-08 13:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"781130-kolla-quay.io","meta_rev_id":"8879edc931123667c34b9ef026ae1be56962dd80","_number":781130,"virtual_id_number":781130,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 13:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-06 19:27:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-08 12:45:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-08 12:45:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-04-12 03:52:42.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-04-06 11:03:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-17 17:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-02 17:28:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-02 17:28:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-04-02 17:28:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-04-02 17:28:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-02 19:10:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"33a2ccb06b7948cc7327db35c0d789c25e554d26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 17:05:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c056d588653bdc528e3ed97f0397c66f4203328d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 17:06:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nworry not, it has to fail so will not merge","accounts_in_message":[],"_revision_number":1},{"id":"668554d4d16786103e10ee258030e8c8ba255cfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 17:07:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee7120c82681039c881ff8b3799e596123ed302b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 17:08:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c7015ffb7b2bf897a30358284e6f56af908f843b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:18:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49fdd9411fa3414ba960d50967f0f243 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4304982443754c9687c852ccb7865de8 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11819229a88d40b9a401c5ede4170fe1 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f6ee3af452a43539e34d2efb28e955f : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d3606a3ae6408e9ec45cbe58f126b6 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":2},{"id":"5e9c2efb8302565c678c0bf21ce6fd109e8cab9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:19:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3867eb4ccaf31528a434ce28b6b53a96233d87f1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 17:20:52.000000000","message":"Patch Set 2: -Code-Review Workflow-1\n\nsafety switch in case it fails to fail","accounts_in_message":[],"_revision_number":2},{"id":"9337362b3352a277169c238e4abe87ba12e8795e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 17:46:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8e6b69e66cd0ff8f0858e549073fa2a1830cef3f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 17:46:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c7f0ad515b170508b6a2fcc205817b17018196e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:57:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aebbc7b2cd947a68db88be9766d7f03 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0deada2fe5e148d9923c11102ded1acb : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/738ce3437c4545de8966154e6ff12c34 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/961a2fb5784b4cf8bd2f0caf66ad3309 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c19ed8c956e495baee96106ebc4a650 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":3},{"id":"0573e307684be5f462cce45a7bb0c08c7b6f2551","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:57:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"58bea998abbb8f04359bf9848886f73202164064","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 18:15:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0f1e4006f68b27d0f4b47202a36db8bd13eaf6ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 18:20:15.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Configuration file is not a list.  Each zuul.yaml configuration\n  file must be a list of items, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item starts with \"- \" so that it is parsed as a\n  YAML list.\n  \n  The incorrect values are around:\n  \n  null\n  ...","accounts_in_message":[],"_revision_number":4},{"id":"dcc544ebcb019cac0959ad97329a7a3c55b0ae35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 18:20:40.000000000","message":"Patch Set 4:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Configuration file is not a list.  Each zuul.yaml configuration\n  file must be a list of items, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item starts with \"- \" so that it is parsed as a\n  YAML list.\n  \n  The incorrect values are around:\n  \n  null\n  ...","accounts_in_message":[],"_revision_number":4},{"id":"191487798216de70bd471cbf565b0a895aa49df1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 18:33:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cc78bb9b284ea8286db6cecbbf777d740b857516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 18:43:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85c529de918b4ae89d322cc96f14cfef : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":5},{"id":"e84b311d718aa9289c12105f4e14784b33d2de19","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-17 18:44:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d745bab02c8047cff5131e327eef5602bf3af02e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 18:45:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"937931e019ed47e5cef3552d8919eea2287c1f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 19:20:40.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-publish-ubuntu-source https://zuul.opendev.org/t/openstack/build/c716a47192c54d9d85e5f4034a25c3f2 : POST_FAILURE in 6m 14s\n- kolla-publish-ubuntu-binary https://zuul.opendev.org/t/openstack/build/da32259c637643b1bd260a872d9024e4 : POST_FAILURE in 6m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"935221e4132cd81217d0f6a1b97abcd90d79f54b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-18 10:55:05.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6eedcb3ac85af548f9fe584fc4427dba53929dc9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-18 10:55:28.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0e6a166d492d2e5752c1a528c9f28c6e441f07f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 11:45:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a8b796dc35b484185b1a6f804128e81 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":6},{"id":"1cf9f46c745a3bbe25b9704dd583ddf4738af3a9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-18 12:07:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"97d4a97c338ec202b1de389b33c19b0208151c68","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-18 12:09:10.000000000","message":"Patch Set 6: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":6},{"id":"441d5b459945001aea71d3bdbd29b89e188ce1d3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-18 12:09:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ece334502ad7d7f295ab8626d10d33d6c8f50789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:18:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"20c494383fc74dfaca00481d54d682f6c2b60bb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:41:08.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-publish-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9dca9856c7e4a1c9d085f6a7dee1d38 : POST_FAILURE in 11m 35s\n- kolla-publish-ubuntu-binary https://zuul.opendev.org/t/openstack/build/737aacc6bdf3494aa5844edc769078f4 : POST_FAILURE in 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dc1aa4542836f9c8915bd5154b2bc3c14c69e20a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-18 12:50:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b82298062f8b8cf9013f5e180b13a0423bbb7116","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-18 12:52:58.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"8eaed3b824439ae74012cffa854d47f09171504b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 13:03:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3a758ffa84a4e9f958108c7025967e4 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":7},{"id":"ad2b51f48582ad64e294beb35565f943e118578e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 13:07:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d12b74deb9d1d3e53992e65a27bdfdec92563ac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 15:40:01.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-publish-ubuntu-source https://zuul.opendev.org/t/openstack/build/8fd3fd0f199d48a599bb361feaa49b95 : POST_FAILURE in 2h 28m 39s\n- kolla-publish-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f07161df4324101bc7b69a5fdd9fcc4 : POST_FAILURE in 1h 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5e1124decaffd51aea5c23db4d432b706319028a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-18 16:31:14.000000000","message":"Patch Set 7: -Workflow","accounts_in_message":[],"_revision_number":7},{"id":"29b8721527ae839036c739538f63f27220e085e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-19 16:23:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"14e6f9ec687e67975e49a8efc6bff7f9ac61db2f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-19 16:23:49.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"fc5a8329831e84006a361b95fbe2ffea67d24334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:35:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b845896149a74992bde2874b7e01b241 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":8},{"id":"a65729b59918767c8d96e0759147523b190ee50c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:41:50.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"fbf5da4cdb0f1e58e0e9f3fdf4eab6cb0374ab30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 19:20:05.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-publish-centos8s-source https://zuul.opendev.org/t/openstack/build/1dd1c22038014a45a38a8cd67d26bf79 : POST_FAILURE in 2h 30m 36s\n- kolla-publish-centos8s-binary https://zuul.opendev.org/t/openstack/build/ee8485f9b21e4826997025dac9429503 : POST_FAILURE in 1h 47m 55s (non-voting)\n- kolla-publish-debian-source https://zuul.opendev.org/t/openstack/build/bdb330b96eba41928031c46b2569abb5 : POST_FAILURE in 2h 04m 45s (non-voting)\n- kolla-publish-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/63cfb67c473f44478f0078ea6cd485e8 : POST_FAILURE in 2h 36m 59s (non-voting)\n- kolla-publish-debian-binary https://zuul.opendev.org/t/openstack/build/b287f95e31384704916a5cc6573420e0 : POST_FAILURE in 2h 17m 37s (non-voting)\n- kolla-publish-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce41620fc66c4545868c64413fb3940f : POST_FAILURE in 2h 20m 36s\n- kolla-publish-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ee4b7edef1f4925bfc5c0e88eb0210d : POST_FAILURE in 1h 26m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b16f23712cf8e250d690e73719e0ea1c1f9f9759","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-01 18:51:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dcdd7f7780278bfbb6094ce3ddc009258b211de1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-01 18:52:00.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"a7cafb017a1ed39affde04f76ad1b4be9b624c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 19:03:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02e0538c07bc4260b9e3514c413743ab : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":9},{"id":"1418c5f15c3c5131e9ac8c35a20d2747bddf2bd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 19:04:35.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"3be37f038b480c7abfeb4721937836b6290691d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 00:50:18.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-publish-centos8s-source https://zuul.opendev.org/t/openstack/build/7037728a0d4348e9bd846cb1faaf76d7 : POST_FAILURE in 3h 34m 06s\n- kolla-publish-centos8s-binary https://zuul.opendev.org/t/openstack/build/1677644f7b7645878a87dc08077099ec : POST_FAILURE in 1h 40m 28s (non-voting)\n- kolla-publish-debian-source https://zuul.opendev.org/t/openstack/build/85295dfe283f407f8b7cb32389ac0c99 : POST_FAILURE in 2h 07m 07s (non-voting)\n- kolla-publish-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/455672c75aff461d9d84b4188a3fc065 : POST_FAILURE in 2h 07m 24s (non-voting)\n- kolla-publish-debian-binary https://zuul.opendev.org/t/openstack/build/538488e2bdfc406a9d3f2891ea0c150b : POST_FAILURE in 1h 17m 53s (non-voting)\n- kolla-publish-ubuntu-source https://zuul.opendev.org/t/openstack/build/57338b24230b46b4bae2ded53dcba47c : POST_FAILURE in 2h 16m 49s\n- kolla-publish-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ab56da74e594a9fa04e99119ff4254b : POST_FAILURE in 2h 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"58b226ad25103f40d51935ddc85c11fa70e0e494","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 14:03:18.000000000","message":"Patch Set 9:\n\nwe have current images for master/Wallaby; switching this patch to run publishing to quay on a daily basis","accounts_in_message":[],"_revision_number":9},{"id":"d5ac0db41512fad9e401d881013ad037875440d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 14:08:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51cda9f1e0514f72a56888470abbfbcb : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":9},{"id":"5831970e342688fa3881e6c59682e705a6be6496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 14:08:58.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d6da6521e2208f39f5dadd8b9df556ac8ff84369","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 14:12:07.000000000","message":"Topic quay.io removed","accounts_in_message":[],"_revision_number":9},{"id":"86be80bf48f1bb8f6356a31948223682b79a5375","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 14:12:26.000000000","message":"Topic set to kolla-quay.io","accounts_in_message":[],"_revision_number":9},{"id":"b486dd4df3682e46aa7580065f8575b311cd6b4c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 14:21:19.000000000","message":"Patch Set 9:\n\noddly, just commenting here made this change go to check and gate again 😮","accounts_in_message":[],"_revision_number":9},{"id":"6977300d1741f4b109ce87d5f26d5b32e146f61b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 16:02:30.000000000","message":"Patch Set 9:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"f7b7cda933940a41b58965462cb8c5415f065785","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 17:21:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9e6eeda1d1b93cebc0ec85e217d08400a9be34c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 17:24:02.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e52fc6451ff9b9acb3b9b22267310b6d9c0dcecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 17:25:28.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d466e713a61f240c8b043d3862d39433ff0910b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 17:28:04.000000000","message":"Patch Set 12: Backport-Candidate+2 Review-Priority+1","accounts_in_message":[],"_revision_number":12},{"id":"de9cbc7e463339fb21b22f37f9767baa5a370d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 18:49:29.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e6a6cf3883e64c45aaefa17673b2cfd6 : FAILURE in 25m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/001240b62122483fbc1f239e91633849 : FAILURE in 30m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/af91cb09f25a439aa44be0bbcd6e2347 : SUCCESS in 1h 11m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/608aaa2cf4b34d0990d1f18f3108c304 : SUCCESS in 57m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6d7fc0afc8da4d9aadbed6fc821eafd6 : SUCCESS in 1h 18m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/247114ccbace49f8a41e0331230b8863 : SUCCESS in 26m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0d3183cbb1ef4f89aa50c0ef93024e90 : SUCCESS in 23m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/368394ed15b44f53b9ee3ea8c2d354f6 : SUCCESS in 56m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c909ed599242dda9224d58d36a7428 : FAILURE in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be61831dab0c43409be3c464ebca1392 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f67eb251fdec4bce831b53975af65a6c : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5aec451cc92d4d51a3ccafb3403ce923 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff8175913d814072a8dba68d3039fee0 : SUCCESS in 6m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/25b2b8d371fa48f1b0c5186291b977fd : SUCCESS in 21m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d986528fa754e2cae198b6c4e9b3063 : SUCCESS in 34m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a8a2b2cde124ce383b00383a616cde5 : SUCCESS in 1h 02m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aee5690f3ef249b2a618e99101e2e200 : SUCCESS in 53m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f3a42242b82e48d28593587add8d9770 : SUCCESS in 51m 59s","accounts_in_message":[],"_revision_number":12},{"id":"59ae830e50e49fc735ee7e51b89b7ee94f270183","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 19:04:18.000000000","message":"Patch Set 12:\n\nmonasca-grafana is failing on centos\n\nINFO:kolla.common.utils.monasca-grafana:go install -v ./pkg/cmd/grafana-server\nINFO:kolla.common.utils.monasca-grafana:\u001b[91mgo: cannot find main module, but found vendor/vendor.json in /grafana/src/github.com/grafana/grafana\nINFO:kolla.common.utils.monasca-grafana:\tto create a module there, run:\nINFO:kolla.common.utils.monasca-grafana:\tgo mod init","accounts_in_message":[],"_revision_number":12},{"id":"c03e3a65e65bf6015ebc247e17172c1644e50149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 19:10:13.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8bac4c94ba3964967347b8debbc6f57c6b925af3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 19:10:21.000000000","message":"Patch Set 13: Review-Priority+1","accounts_in_message":[],"_revision_number":13},{"id":"a63ef3b415a671633c4c34553ac54221ea3e92e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 20:53:53.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/be79396eb4d04292a1fb838ef3badf89 : FAILURE in 21m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0d13ebe0d9fa45aea50ab6467b96b950 : FAILURE in 52m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7053a130ef5c465c9991d456cd9be9f1 : SUCCESS in 1h 23m 15s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9b5427a8ae8e4eafa3cacddae1ba4c44 : SUCCESS in 1h 25m 04s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/22c6c995e42f4cd6b9a67493721894b4 : SUCCESS in 1h 05m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/484d8238c1d34252a8334553f8be4b1b : SUCCESS in 43m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/74ad887581024e5690c78e74fd97ed7a : SUCCESS in 21m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/392477e3a3f0444f942d51e3e935e6ff : SUCCESS in 1h 15m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c07d3e1bb84e90bf0c416f8e625098 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/735d42a6957a4bf18f91dcbe42af4893 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33969324a1314aedabd590a025e1f8e9 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7282d06faaf6423e9b205bd9d4242d42 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2d985aa07d345aab62334d2a7827abd : SUCCESS in 6m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec2a9858d554416395b2825617edb2ee : SUCCESS in 17m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/83004f6a6bb24635bd1b3dc21d576361 : SUCCESS in 33m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b25d21be22414985bf5a42c2587f4cce : SUCCESS in 43m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5161bdbe123c454c9bb3783bcb1baf14 : SUCCESS in 55m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4c1af45d1bf44414bbdba462009f59fe : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":13},{"id":"60ce0a8d9cf75ce272595cd6733e7071e479c6c6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-06 09:10:01.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b00ebcefe75d739df87c3b3f2697606d524ebc5f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 10:10:26.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"910520739979f70a0c7bc215e404e6658de9296a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 10:26:51.000000000","message":"Patch Set 13: Workflow-1","accounts_in_message":[],"_revision_number":13},{"id":"b670c1480d1eaddaed3c14c4544c4000aab3bb0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 11:01:22.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1dd1052154162b58c9c31e4ea10da39a38309642","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 11:03:46.000000000","message":"Patch Set 14: Review-Priority+1","accounts_in_message":[],"_revision_number":14},{"id":"24546ddeb9cfb275690aa1268322f78de00c1974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 12:22:54.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2178e77eafcd40da83483bee79d5d2ff : SUCCESS in 25m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/60405a9e70da40f1a9c5734acc53166c : SUCCESS in 39m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/eee19184cab542dc9c3b1f8f6fd493ed : SUCCESS in 54m 29s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/331bd13a68d148889f1735e0b7cca8d6 : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5fe8981fe0ea470788b5d270bbd200a1 : SUCCESS in 1h 10m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c451e37dbd8049fda89fd2bdc98a7b85 : SUCCESS in 26m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d909ec542e9541b4b13a0bc32c14b273 : SUCCESS in 22m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2d7549c151eb43d8a194366c4618e445 : SUCCESS in 58m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d8c68f05dc466eb7ffaeb765d293de : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a36ddc568b64305baf89f102a798010 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa4b9cf39a5c46bdb1cb8350f67b05b6 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5e07cf5b4884182b175f45b5038fa6a : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59eb616ab2c644648798a6b1af61cbf4 : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/621bf0fbf5c042ee86bd152172d017cb : SUCCESS in 21m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d459ec5fc3f249c3a6a22c26745e4aa4 : SUCCESS in 47m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4fe269b2fda4514b25fd5441698a8ae : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/51abb7ccf96342cdacf8c00175ef6538 : SUCCESS in 59m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/97e6507fa85f40eb8d3454937b8338df : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":14},{"id":"3806e776857c94848ef255d46b1d1f2143ff1f01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-06 16:18:35.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6e2ed0aacf1e20e5fbf2f0de0425f3b854a49c99","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 16:20:57.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"846a4889ca8fd1191d838815e2902da153bf5afb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 19:27:39.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"e5e47fcb0f43f5b5141d4e1f79d70960ea55d85b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-08 12:45:05.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d9b2c684f176ee4f6e9b52263297917c94b46fe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:45:40.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"ba99fc5e4e288e30f4d9f1e1f3420ba0240f9c5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 13:59:23.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4da25f693839499e83ae29d455be26fb : SUCCESS in 53m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb885ccdd064667859f5a10d6ecdf25 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18eef69653c8437880821cb5bc6747d5 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/222673340e4f4aa78efd4b41e7f50651 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a12b1870cbb3486b9245cfdc2be9cfef : SUCCESS in 6m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c7bcb1d93b04c9a96037fd01e7c3152 : SUCCESS in 38m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d946d17ca69f4188940d4700f4f1f17b : SUCCESS in 57m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e7830763af5e4306844b0fe1d4cece7b : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":14},{"id":"2d1b8cb55b1b855d37d1c1b7c854d9922bd5a352","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 13:59:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"d51650229f062e519c3c1e6977eacf6cbcca8246","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 14:11:49.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d5675c50104d22beffb6a4dcc0a143 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":14},{"id":"b88f5a87524289bbe1cc46b3cd50832b201c88d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:52:42.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"10d8f03c4e6fe2e78f06e9c59f44210c8c2e73eb","revisions":{"f047abe5994f2f6cc6da6f540f9c81318de68523":{"kind":"REWORK","_number":1,"created":"2021-03-17 17:05:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/1"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f047abe5994f2f6cc6da6f540f9c81318de68523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f047abe5994f2f6cc6da6f540f9c81318de68523"}]},"branch":"refs/heads/master"},"820c5b3065ecebc3b005d24897bacb4105bbfee5":{"kind":"REWORK","_number":2,"created":"2021-03-17 17:07:51.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/2"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:07:43.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/820c5b3065ecebc3b005d24897bacb4105bbfee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/820c5b3065ecebc3b005d24897bacb4105bbfee5"}]},"branch":"refs/heads/master"},"815ff72d551d041654a229a32518f472620fce87":{"kind":"REWORK","_number":3,"created":"2021-03-17 17:46:23.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/3"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:46:14.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/815ff72d551d041654a229a32518f472620fce87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/815ff72d551d041654a229a32518f472620fce87"}]},"branch":"refs/heads/master"},"77f1d58ca9103706b477c8bd06de8f536f54240a":{"kind":"REWORK","_number":4,"created":"2021-03-17 18:15:21.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/4"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 18:15:12.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77f1d58ca9103706b477c8bd06de8f536f54240a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77f1d58ca9103706b477c8bd06de8f536f54240a"}]},"branch":"refs/heads/master"},"17180f642d40fce0da3b5220772c3022a4bbdb80":{"kind":"REWORK","_number":5,"created":"2021-03-17 18:33:38.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/5"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 18:33:28.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17180f642d40fce0da3b5220772c3022a4bbdb80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17180f642d40fce0da3b5220772c3022a4bbdb80"}]},"branch":"refs/heads/master"},"d0ed6131e414ee733ca0a5ef28eeb43571d65d84":{"kind":"REWORK","_number":6,"created":"2021-03-18 10:55:05.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/6"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-18 10:55:00.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0ed6131e414ee733ca0a5ef28eeb43571d65d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0ed6131e414ee733ca0a5ef28eeb43571d65d84"}]},"branch":"refs/heads/master"},"726af005679bd33b81e78bdabcc801da39931d33":{"kind":"REWORK","_number":7,"created":"2021-03-18 12:50:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/7"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-18 12:50:15.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/726af005679bd33b81e78bdabcc801da39931d33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/726af005679bd33b81e78bdabcc801da39931d33"}]},"branch":"refs/heads/master"},"5fd235065d1dd90a251e69456b76455d41a78773":{"kind":"REWORK","_number":8,"created":"2021-03-19 16:23:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/8"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-19 16:22:22.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5fd235065d1dd90a251e69456b76455d41a78773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5fd235065d1dd90a251e69456b76455d41a78773"}]},"branch":"refs/heads/master"},"81dbb9ea067ccedb3a1ba96d9ef0842d419f13b9":{"kind":"REWORK","_number":9,"created":"2021-04-01 18:51:35.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/9"}}},"commit":{"parents":[{"commit":"823b029adb9bbe3e03e1ac9e467fed8ea749333d","subject":"Merge \"Debian: use Wallaby specific repo for libvirt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/823b029adb9bbe3e03e1ac9e467fed8ea749333d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-01 18:47:14.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81dbb9ea067ccedb3a1ba96d9ef0842d419f13b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81dbb9ea067ccedb3a1ba96d9ef0842d419f13b9"}]},"branch":"refs/heads/master"},"6a7ff01bc1919f7c47cd5877634dad374e0b5f42":{"kind":"REWORK","_number":10,"created":"2021-04-02 17:21:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/10"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-02 17:21:00.000000000","tz":0},"subject":"[CI] Publish images to quay.io on a daily basis","message":"[CI] Publish images to quay.io on a daily basis\n\nBased on the discussions from the past 3 Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nWe still publish weekly to the DockerHub.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a7ff01bc1919f7c47cd5877634dad374e0b5f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a7ff01bc1919f7c47cd5877634dad374e0b5f42"}]},"branch":"refs/heads/master"},"bdb59e09663126a270a535e5bdcbdfb89fa0918c":{"kind":"REWORK","_number":11,"created":"2021-04-02 17:24:02.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/11"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-02 17:23:51.000000000","tz":0},"subject":"[CI] Publish images to quay.io on a daily basis","message":"[CI] Publish images to quay.io on a daily basis\n\nBased on the discussions from the past 3 Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nWe still publish weekly to the DockerHub.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bdb59e09663126a270a535e5bdcbdfb89fa0918c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bdb59e09663126a270a535e5bdcbdfb89fa0918c"}]},"branch":"refs/heads/master"},"8596a0b9e57670d9adbe886c2692206056b03680":{"kind":"REWORK","_number":12,"created":"2021-04-02 17:25:28.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/12"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-02 17:25:20.000000000","tz":0},"subject":"[CI] Publish images to quay.io on a daily basis","message":"[CI] Publish images to quay.io on a daily basis\n\nBased on the discussions from the past 3 Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nWe still publish weekly to the DockerHub.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8596a0b9e57670d9adbe886c2692206056b03680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8596a0b9e57670d9adbe886c2692206056b03680"}]},"branch":"refs/heads/master"},"e340100c04ce1a7133b0dd90d1f336688f9825da":{"kind":"REWORK","_number":13,"created":"2021-04-02 19:10:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/13"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-02 19:10:05.000000000","tz":0},"subject":"[CI] Publish images to quay.io on a daily basis","message":"[CI] Publish images to quay.io on a daily basis\n\nBased on the discussions from the past 3 Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nWe still publish weekly to the DockerHub.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e340100c04ce1a7133b0dd90d1f336688f9825da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e340100c04ce1a7133b0dd90d1f336688f9825da"}]},"branch":"refs/heads/master"},"10d8f03c4e6fe2e78f06e9c59f44210c8c2e73eb":{"kind":"REWORK","_number":14,"created":"2021-04-06 11:01:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/781130/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/781130/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/781130/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/781130/14"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-06 11:00:53.000000000","tz":0},"subject":"[CI] Publish images to quay.io on a daily basis","message":"[CI] Publish images to quay.io on a daily basis\n\nBased on the discussions from the past 3 Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nWe still publish weekly to the DockerHub.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\nDepends-On: Id113a53c083d105ac9200474ff20faacef6decaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10d8f03c4e6fe2e78f06e9c59f44210c8c2e73eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10d8f03c4e6fe2e78f06e9c59f44210c8c2e73eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
