)]}'
{"id":"openstack%2Fkolla~781494","triplet_id":"openstack%2Fkolla~master~Ia7768caabfe214a629339d50973d0d9873fc65d5","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ia7768caabfe214a629339d50973d0d9873fc65d5","subject":"centos: enable PowerTools repo in images which needs it","status":"MERGED","created":"2021-03-18 14:41:31.000000000","updated":"2021-03-23 12:47:52.000000000","submitted":"2021-03-23 12:45:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781494","meta_rev_id":"93e6bb19e7130ff288c523a607721e952be96fcf","_number":781494,"virtual_id_number":781494,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-23 12:45:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-03-22 16:27:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-03-22 17:45:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-03-22 17:45:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-03-23 12:47:52.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 16:06:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 16:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 10:15:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-03-22 13:35:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-22 17:45:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"31bd12e376a00e48bea63f07e4e964c27d9be1e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-18 14:41:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8707f2973a3d8c6f375ec1d732b0dd398da4972a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-18 14:43:02.000000000","message":"Patch Set 1: Review-Priority+2\n\nFixes CI failures for CentOS","accounts_in_message":[],"_revision_number":1},{"id":"b68df4da7b1ed0eb04b7b9383e6115e724f36428","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:06:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/21188ec3cc0f44d09cd735b46463088b : SUCCESS in 1h 07m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef32cba8571e45e0893cf5a6b5701aca : FAILURE in 41m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d9f5588b3bc94b8686208fd39d80cdc7 : FAILURE in 1h 01m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86fe2fed42dd89abd28638218cdcf1ab34c05b9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:07:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9d9e8925cd614de5851a44986984843b : SUCCESS in 29m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/66f1917a04aa48f2aa462eb09984fb8a : SUCCESS in 41m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/463803d6c6b9454686f89e9ff415497c : FAILURE in 17m 10s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0cc986b23fb74aeba681361f7c060825 : FAILURE in 16m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/49916372a3b1412d80a2252b227bf5bc : SUCCESS in 1h 13m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f02d5bd0447c46fc85452aff91901a21 : SUCCESS in 41m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b3b3e3b5c36f44b786449244762738b4 : SUCCESS in 24m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/774663c76a6a48a3b7a494757f83e891 : FAILURE in 16m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2febbaa99a465e913e9a24b5534d7c : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d35cc7337004cc687823bf556cc84e5 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ec59ad032b04e08a9effe3df922f454 : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1be345b097804347a4d8e22456815d80 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b740bc7fda14ed3be9791cc8d9267c1 : SUCCESS in 7m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c1c4837bede44f99a11e66b6b5588f8c : SUCCESS in 22m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f718cecb116247409adc7b7db2323b00 : SUCCESS in 32m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e98495a78c34d268e7468eedbec6497 : FAILURE in 17m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b923b3acdfa44cafac9ba859b79ec32e : FAILURE in 18m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39eb576d376d4c52afbea6eeff8e5bd5 : SUCCESS in 1h 09m 40s","accounts_in_message":[],"_revision_number":1},{"id":"90bc521becb337bdfb0cd7df837e384884da3476","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-22 10:12:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"420f8220ce0b5d97e8f226bc1d4a5ebdc26eb88f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 10:15:20.000000000","message":"Patch Set 1:\n\nIf you update the patch, please add to the commit message why this is now required.","accounts_in_message":[],"_revision_number":1},{"id":"7a0f9a301c06688bce3518d83934179dae9eae16","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-22 10:25:03.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e If you update the patch, please add to the commit message why this is now required.\n\nIs \"Horizon, Rally, Tempest and Vitrage depend on packages from PowerTools repository.\" not precise enough?\n\nRDO binary packages are built with PowerTools enabled and then some packages depend on packages from that repo.","accounts_in_message":[],"_revision_number":1},{"id":"bb0d11d151b2786d9370cf643eece315bc952743","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 10:30:20.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e If you update the patch, please add to the commit message why this is now required.\n\u003e \n\u003e Is \"Horizon, Rally, Tempest and Vitrage depend on packages from PowerTools repository.\" not precise enough?\n\u003e \n\u003e RDO binary packages are built with PowerTools enabled and then some packages depend on packages from that repo.\n\nWell, you\u0027re saying that without this patch these images fail to build. They were building fine a few days ago, so what changed?","accounts_in_message":[],"_revision_number":1},{"id":"ad122f6910249d689fefe24a416e6dbb5d2bee96","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-22 10:42:10.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e If you update the patch, please add to the commit message why this is now required.\n\u003e \u003e \n\u003e \u003e Is \"Horizon, Rally, Tempest and Vitrage depend on packages from PowerTools repository.\" not precise enough?\n\u003e \u003e \n\u003e \u003e RDO binary packages are built with PowerTools enabled and then some packages depend on packages from that repo.\n\u003e \n\u003e Well, you\u0027re saying that without this patch these images fail to build. They were building fine a few days ago, so what changed?\n\nWe moved to CentOS Stream 8.","accounts_in_message":[],"_revision_number":1},{"id":"b75471ddea743fb9d6ec522d09765026d2d1e1da","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 10:47:37.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e If you update the patch, please add to the commit message why this is now required.\n\u003e \u003e \u003e \n\u003e \u003e \u003e Is \"Horizon, Rally, Tempest and Vitrage depend on packages from PowerTools repository.\" not precise enough?\n\u003e \u003e \u003e \n\u003e \u003e \u003e RDO binary packages are built with PowerTools enabled and then some packages depend on packages from that repo.\n\u003e \u003e \n\u003e \u003e Well, you\u0027re saying that without this patch these images fail to build. They were building fine a few days ago, so what changed?\n\u003e \n\u003e We moved to CentOS Stream 8.\n\nOk, but why did that patch merge if these images fail to build?","accounts_in_message":[],"_revision_number":1},{"id":"715ccaca1ab4714cf4ead76d65a9d4080568176f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:34:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/47923fde9fb04fe09db46190ca922e74 : FAILURE in 35m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1c168292e87e4f3a82e6e399e1e4d413 : SUCCESS in 59m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/84722c47a19b491085799f2f76904738 : SUCCESS in 1h 02m 07s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/27927c734fd543d59d85f0d350ef512e : TIMED_OUT in 2h 01m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/43b2e0538045415b83031223102d199c : SUCCESS in 1h 10m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a5ae35f56bc4f2daea94c90fbc31f10 : SUCCESS in 35m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/04b07d7fe5214134acc4100d93e728e7 : SUCCESS in 19m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9032fc3c2f2a4ddeaf3adcb82d757321 : FAILURE in 1h 10m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a866fafcd1471f9fc0845dfa981634 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18d158c9c5684f19aa5a0326a3ccd1b3 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f906db3f46b7448fa39862e7480ac44d : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e639df9fa89b4978b59600026df6ebb8 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dadc516ea3a4395980afb9081417864 : SUCCESS in 8m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a7403ce6b804567bada24d95b5e9293 : SUCCESS in 26m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10aa64f9da084934a980f1c645abbfcd : SUCCESS in 44m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d1ff9b0cbdc54494b698c426d0445f04 : SUCCESS in 51m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47eb831afc0e4554a44893f90d8bf98f : SUCCESS in 57m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0be373e00e14425b92408b5f04558711 : SUCCESS in 1h 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"31cf8624a6850411e16c0aa8e10be797713df2cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:36:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7950621040fd4d48861da031415c7215 : SUCCESS in 59m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7af297860807461996339b3ff1f38076 : FAILURE in 44m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5e03fe51d7441bb8a5a6e16343cd339 : FAILURE in 55m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe8edfa0a56172c8414cba88903de0c61aed5761","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 13:35:56.000000000","message":"Patch Set 1: Code-Review-1\n\nCentOS binary job failed","accounts_in_message":[],"_revision_number":1},{"id":"f8ce31a7e9de8774adcc89a9b247ab1b47edbdc4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 13:36:44.000000000","message":"Patch Set 1:\n\nPlease also include commit message to include explanation of root cause.","accounts_in_message":[],"_revision_number":1},{"id":"3907a0acd28041b0790e24dafdcbfa37d65c10ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-22 13:47:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"55757a7cffc3a9d23d67edff4dd48fd183cb4af7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 15:56:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a3100559162428098dc342b76a997e1 : FAILURE in 58m 09s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/29ee7abc72ac4fd39f0961b0dde335c1 : FAILURE in 28m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fec85b9a429e44ecb1e8e66a047bebb5 : FAILURE in 54m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b312e4019873fe2228cf13139896ac08d399662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 16:07:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4b43ba8a16eb48bbb7c54bcbeb96ec3c : SUCCESS in 29m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7c911db89630483d8b7722162b199059 : SUCCESS in 47m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6c491880996743238623d27651111a37 : SUCCESS in 1h 07m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/94157e7032a74c72b8ee7bc7d88cfce2 : SUCCESS in 1h 09m 13s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/41de497e398342c1b037be48779b6283 : SUCCESS in 1h 11m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/19b5c21a625445ff8b700a0e62a86bce : SUCCESS in 29m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/edb44a4e8aca48d4ac2a0b6158521341 : SUCCESS in 21m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7276f3b9f8304c8c962cb0776ffc99ca : FAILURE in 41m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93dd5e7dc92d46d79e8b225fd649b5e5 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/071618dd87b840f19fc4aa80cf63b472 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2e721c50bb74900b830620555d4bff7 : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85640567d5444114b922769371804e1e : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b87170cbe84b4e03952c2e8549e59946 : SUCCESS in 10m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b97df8430ead466bb1efc3e1a864e473 : SUCCESS in 23m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e6d37c2ab2140c5a3ca0a5303146e8b : SUCCESS in 51m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a5f7c8af93534465b9a510f43b1c1da2 : SUCCESS in 56m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/724678e11bb4443bb8ca7602caf15ba2 : SUCCESS in 59m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dd1826276b8149cb81cc903607db3038 : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":2},{"id":"99d29c4c6710880cd5988b48439b8ddbb98396d1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 16:27:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5973ffee1c1d7baf9ae043aa1dd025840f3fd67a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-22 17:45:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f4d86d1ae3d8c15f14e95ee28141331d836c0286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:47:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dd78b070d1bfaf8f1426ba1908fbc50d512f2f6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 19:53:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/68d16e0de75f404094cb9434b751b118 : SUCCESS in 56m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a3ef0a568f1d43fd8aa666be3fd422ad : SUCCESS in 1h 21m 59s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/81602e03c45242dd800e734ee717e40c : SUCCESS in 1h 27m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be669188ecce4bd1a2fdff3cdb6669f4 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9fe588f402748af8efa20e32e21e44b : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec789604320842dc9b835460ef5d7343 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62aa6ec5a9084b65b560266c03c35b65 : SUCCESS in 10m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ef1bfc7dcd54a34ae6c6d4878499d4c : SUCCESS in 48m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/366d6755a0074616973a638dd226be44 : TIMED_OUT in 2h 01m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7c47f4b3592842deba97a55f6edefcd9 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":2},{"id":"07b3e7d9cf6e454e3895a7386139fbdf621ee248","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 09:40:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"00527426401dba52780da35beb7c514477c15b23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 10:39:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2202c540138c4be495e6687521f0c80c : FAILURE in 54m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e0ff30b8789a43d0921aca8b141b950c : FAILURE in 44m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/749889d7a8894c37a2391be31838b288 : FAILURE in 53m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc94c4232725c8941dbe7e61026f9d081bcaf073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:24:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/01f5c96ff5fb4f7fbd738223893db288 : SUCCESS in 19m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fabb2d5183d44928bd35f28dccd249c6 : SUCCESS in 42m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8094831c44af4ecd9572422a82f83725 : SUCCESS in 1h 04m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1ff2a0c7c6784519a928506146e8c7cd : SUCCESS in 1h 07m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bbc34651ded048ae83cc2535053663e0 : SUCCESS in 1h 18m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/877a978991614c3788de7e4cefc2b9d4 : SUCCESS in 31m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e4db34a38e4e4298b41edffeaf301a07 : SUCCESS in 19m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/412e3af4b7ce481799c3562e27698af6 : FAILURE in 39m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c314cc41d14e31a450af1f8be88d8f : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50b6513292fa491eb120c19a47255934 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fff67bb3f2d4742bb44b9908aa2b934 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18f057d51b2a420c8e41ff582c305228 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3f345bae2b4af698277b069a0f8b95 : SUCCESS in 7m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/893f836dc62641c69fe7a1f82e800963 : SUCCESS in 40m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec2376d0201542ae88d3848566706fcf : SUCCESS in 36m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9fc3312bfd44ce6849b1e489074d1c2 : SUCCESS in 1h 19m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f29b3b6a924145f58703a0a9da41c826 : SUCCESS in 50m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4019003100154a6bbe8419e5792216f1 : SUCCESS in 1h 34m 04s","accounts_in_message":[],"_revision_number":2},{"id":"17e46d95e1dc32ef9cc8f0205fe39457febb0945","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:25:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f0b7cfec1b66c1fc04f8e753c83d3c64004e0b24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 12:45:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ff23f1d640f6478e9eb71b9ae964e109 : SUCCESS in 44m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b3b324fb7689458c9b8017d0480b72e9 : SUCCESS in 1h 01m 30s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f986e1deeff74ba484fc71f27c57d7d7 : SUCCESS in 1h 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d7f8c1f895640149fe6febdf1f1b916 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64632c8a672541cdb0c0934c2966bd19 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0935584b2a94533890c90c938806d6f : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68dfe201e1bf487aa9aaed78b4950edb : SUCCESS in 6m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/53e05b878b07496fb91838753a9eb2e6 : SUCCESS in 26m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aac7dbd56f6641279f621671e064581d : SUCCESS in 54m 08s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fcd69b42e759450eb184d33657146d89 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":2},{"id":"66b5ebb41148ee2d4479f7c83d0d5408c9bbc871","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 12:45:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dc399b382052d08fbad4eba27f46a66df01da83e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 12:47:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef4b25035f48459297528476da4a5c1d : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c0c2d2448f88a82bcc29b6771f90202011cbb1b","revisions":{"1880545f327c8dc5cb1c6390c033b72032479441":{"kind":"REWORK","_number":1,"created":"2021-03-18 14:41:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/781494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/781494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/781494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/781494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/781494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/781494/1"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-18 14:40:47.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-18 14:40:47.000000000","tz":60},"subject":"centos: enable PowerTools repo in images which needs it","message":"centos: enable PowerTools repo in images which needs it\n\nHorizon, Rally, Tempest and Vitrage depend on packages from PowerTools\nrepository.\n\nChange-Id: Ia7768caabfe214a629339d50973d0d9873fc65d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1880545f327c8dc5cb1c6390c033b72032479441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1880545f327c8dc5cb1c6390c033b72032479441"}]},"branch":"refs/heads/master"},"5c0c2d2448f88a82bcc29b6771f90202011cbb1b":{"kind":"REWORK","_number":2,"created":"2021-03-22 13:47:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/781494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/781494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/781494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/781494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/781494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/781494/2"}}},"commit":{"parents":[{"commit":"7ec4a4c8aecff1c5e3b04c2ad2145fcd6bdd63bd","subject":"Merge \"Build Masakari for Debian binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ec4a4c8aecff1c5e3b04c2ad2145fcd6bdd63bd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-18 14:40:47.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-22 13:47:38.000000000","tz":60},"subject":"centos: enable PowerTools repo in images which needs it","message":"centos: enable PowerTools repo in images which needs it\n\nWe moved to CentOS Stream 8. Just no one noticed that binary target was\nfailing to build.\n\nThis change sorts out all failing images.\n\nChange-Id: Ia7768caabfe214a629339d50973d0d9873fc65d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c0c2d2448f88a82bcc29b6771f90202011cbb1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c0c2d2448f88a82bcc29b6771f90202011cbb1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
