)]}'
{"id":"openstack%2Fkolla~781606","triplet_id":"openstack%2Fkolla~stable%2Fstein~I1d175b603bbc3adbd371a2d66401ed380378ea9e","project":"openstack/kolla","branch":"stable/stein","topic":"fix_pip2","hashtags":[],"change_id":"I1d175b603bbc3adbd371a2d66401ed380378ea9e","subject":"Fix url for get_pip macro","status":"MERGED","created":"2021-03-19 09:37:03.000000000","updated":"2021-04-01 13:30:58.000000000","submitted":"2021-04-01 13:27:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781606-fix_pip2","meta_rev_id":"1c6a83c4cf0a216459111dc32b5811052955d852","_number":781606,"virtual_id_number":781606,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 13:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-01 09:19:23.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-01 11:44:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-01 11:44:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 10:57:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 09:19:23.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-01 11:44:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"aae36b46896e2100b201ab84a99f1c7e47b9114a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-03-19 09:37:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74430faa7373f334588a51ef0ae689f5231f0c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 10:57:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6f6a9c7dcedd42829522f15bf040006b : SUCCESS in 45m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2a2ae6a19205487f85411ad6b7e0c1d4 : POST_FAILURE in 19m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/64e4d77afc9044879b7079bdb3adc56f : SUCCESS in 1h 14m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fa4da7f2132b46cda0a5ef25184cd739 : SUCCESS in 1h 09m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/35ae4fd179c440bf853f8007ac97e70a : FAILURE in 43m 31s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/acd2a9b8aa9c416c9fa4a0537262a6ed : SUCCESS in 53m 21s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bb3f16ab859041a6bef89c1ee8e49ba3 : SUCCESS in 49m 53s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/91abdf9244534aa089ddea2baa3a8b8c : SUCCESS in 44m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e3838059e444a1fa7729d28ee859835 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3749e41d8cc147abb315b0f9b88d6383 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/776e7f0b38284fd3a4089757afc24a8e : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d5ecb4c93404142824367dd7a47b3fa : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f992a58249af463eaedc48656fdae590 : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9cb693faff3549149363df418ff9c8cb : SUCCESS in 23m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d2f4916f38f4b6185530e27ba8f5ae3 : SUCCESS in 22m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c2ac6a2399f49389cdbc27f28475523 : SUCCESS in 57m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/961385fef2d44efd97ad636d77071f00 : FAILURE in 18m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eddaa2be2ca14234a75e7abd0a6effe9 : FAILURE in 22m 07s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"e9d8c36767f020a986e799d3ba8962e9ff211012","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-03-22 09:08:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88d30626ff6630c7ad253171434c09719ebc3917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:44:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a044bfe3be0e41d1a8141c7b2695535a : SUCCESS in 41m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e1543f889fc040eaab37a36d726153bc : FAILURE in 6m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b9dbb0439d2a489896b067b0f2dac491 : SUCCESS in 1h 24m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bacb6fbc73974ca5aface08a5df24ce4 : SUCCESS in 50m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5e1d2ad9d4504b2bbe76bb3382fe51c8 : FAILURE in 34m 59s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/61b6c940d8544901a57f96640ebc4e16 : SUCCESS in 39m 40s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/310cee8436cb47e4b6a6fca00acfbc76 : SUCCESS in 39m 13s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/adb30091a5c04210a0913a8626246a25 : SUCCESS in 42m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e358989a888c425c95e2aa5c4901f92a : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ce8d584e0d244b5a53eb349d80fb544 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f55768c47efb4942a9b054e75a92f074 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a9f027b391e49c0b126ead292838f5e : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b0a0ef0eef4d02a8b0cd8424c72061 : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65dea500981e4c82905389215492cf4b : SUCCESS in 23m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/11c72c8bfbda4723bec5916c17d4faeb : SUCCESS in 25m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92ace6d44a554adab2ef74a9695930eb : SUCCESS in 52m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/577f052fc00c489696f1f9f4bcb32b87 : FAILURE in 20m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/850167ae8c0540c19473b18518e37eaf : FAILURE in 38m 08s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"143a2dc5f1ace8f579e2f5372e0b1fda455eedb5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-01 09:19:23.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c903606469236ca7df184104c992cccf81ec7bc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:46:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e3ae685ee4cd4549ba2cd58d046e3d1a : SUCCESS in 44m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/92c0f419692d44ecb679074f8b5dd660 : SUCCESS in 52m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5d3d5d04a1ee49daa723c0a966ff3823 : SUCCESS in 1h 15m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/13c7c435009b401c871c52592e7999ca : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e5b68cdb863b472098333fbef612ca6f : FAILURE in 38m 57s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4851fc1f7ee24bada3001be91a240b51 : SUCCESS in 51m 54s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1a61a064977d441aa39c8bd4efaf4a19 : SUCCESS in 49m 30s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/349d100663ba41b889d8ee291d1406fb : SUCCESS in 48m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e02c2d5b3cf46e7b4d6b26d1102211b : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c51f36523f644619da2a512a2bc8c61 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21e0a136a4e542269094094a8a7bf95e : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b00c69be0fcf4e779ccf1b2185cf8616 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47aad1c494d4445184f08a059893d385 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b37adcfcfa8f49c484e62a8bea10b6a0 : SUCCESS in 30m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0620f9b06ee74fa6b87c2e2576c9487e : SUCCESS in 25m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/06f8b9c4b449424ca7e13aa7fbb36ebc : SUCCESS in 52m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/00835e4e6ddd4c88aeac931874a657b2 : FAILURE in 28m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2fbbe4a355654b9bb9472cc2fe225893 : FAILURE in 29m 53s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"995993fcf3944956d916c035bc61552204cf6581","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-01 11:44:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6cdfa04d83f4f7d1de97ae22a09737f22df7268","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 11:45:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22440b31c5402de3aced9de2fef37c1646173e7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:26:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/53759f01f8364a808bd912d0227cff11 : SUCCESS in 53m 15s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/bbde14d96e144afda42b4c40ff97e0fb : SUCCESS in 55m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9956942e939c41d6be18031e36a235fe : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fda59f6f95c048a5a0c8d9f7aa50b9fb : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0aead4bd44b41c3baf3cca094a8fc8a : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d7f60d74be7487ca6200bfd93ba2679 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa03810ea704345a94f3b071fb02c75 : SUCCESS in 4m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4265549a4a7d45f2a8fba1fece62f3fa : SUCCESS in 27m 01s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"d8354abb52660ceafb20da109d2c65a32fdab6ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:27:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1c6a83c4cf0a216459111dc32b5811052955d852","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:30:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8483602931ce4fec94981dca94f4b459 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bc50ebba67b52ad545123627b25a0f893f350f7","revisions":{"1bc50ebba67b52ad545123627b25a0f893f350f7":{"kind":"REWORK","_number":1,"created":"2021-03-19 09:37:03.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/06/781606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/781606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/781606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/781606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/781606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/781606/1"}}},"commit":{"parents":[{"commit":"8ebfc19590fe9e3824f2f022559d9ead50e31430","subject":"Fix Mistral source images to respect upper-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ebfc19590fe9e3824f2f022559d9ead50e31430"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2021-03-19 09:30:40.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2021-03-19 09:34:52.000000000","tz":180},"subject":"Fix url for get_pip macro","message":"Fix url for get_pip macro\n\nStein only.\n\nChange-Id: I1d175b603bbc3adbd371a2d66401ed380378ea9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bc50ebba67b52ad545123627b25a0f893f350f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bc50ebba67b52ad545123627b25a0f893f350f7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
