)]}'
{"id":"openstack%2Fkolla~781899","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I7c1f0c5fc643678b29f963b6769d8f900a03b726","project":"openstack/kolla","branch":"stable/victoria","topic":"kolla-quay.io-stable/victoria","hashtags":[],"change_id":"I7c1f0c5fc643678b29f963b6769d8f900a03b726","subject":"[CI] Publish images to quay.io on a daily basis","status":"MERGED","created":"2021-03-19 19:50:26.000000000","updated":"2021-04-12 03:50:58.000000000","submitted":"2021-04-08 16:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781899-kolla-quay.io-stable/victoria","cherry_pick_of_change":781130,"cherry_pick_of_patch_set":14,"meta_rev_id":"4e9c4634444ca8b7ab25a904dd8dd09f590787a1","_number":781899,"virtual_id_number":781899,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 16:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-08 15:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-08 13:07:06.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-04-08 15:23:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 20:12:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-08 13:07:06.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-08 14:51:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-08 14:51:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"185e80089907507352801dc66416bf000c24aa5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-19 19:50:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24e126db3c7d0a1b9158d91ff00c7ffbbf2abd4f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-19 19:51:50.000000000","message":"Topic set to quay.io","accounts_in_message":[],"_revision_number":1},{"id":"f492f957fa5b01e3616ded1ecb56b66151b37ac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 20:12:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a582cdf82a3469fb8274634f03b0a87 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"74a9bb9a5be0c4c269f55d5b80477a1361626368","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-19 20:42:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80a3d246aaab917d3ab168b97352e5e706addebb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 20:43:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"baaffb72960eff3b541aeef3b18d3b7f118b1b4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 23:21:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-publish-centos8-source https://zuul.opendev.org/t/openstack/build/72dd1cb271c343edaed24dead4d4c4a5 : POST_FAILURE in 2h 12m 06s\n- kolla-publish-centos8-binary https://zuul.opendev.org/t/openstack/build/48516a972ac44bcbadb96b4b029d7cb8 : POST_FAILURE in 2h 37m 04s (non-voting)\n- kolla-publish-debian-source https://zuul.opendev.org/t/openstack/build/8d205546bdf4435eb5b21dcecb97c31d : POST_FAILURE in 2h 00m 58s (non-voting)\n- kolla-publish-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/748cba74d79a49fe93eb028107a87caa : POST_FAILURE in 2h 09m 56s (non-voting)\n- kolla-publish-debian-binary https://zuul.opendev.org/t/openstack/build/9223a8c176cc45f7a4ddd9b0a8f41c9a : POST_FAILURE in 1h 28m 45s (non-voting)\n- kolla-publish-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d34b8a06ff44fb79d1e2b60ba3abee3 : POST_FAILURE in 2h 09m 34s\n- kolla-publish-ubuntu-binary https://zuul.opendev.org/t/openstack/build/55b7bebb5d5044989701be183539edeb : POST_FAILURE in 1h 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4b0048f1d3a5a7db828ad692f4efb106206aae7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-01 18:45:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9699411cb753819b4607eaf763708a923c6228ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:54:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9ec48f02e44a4a89bcee9fb984d59b : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7af798f3564675fbfbecf362afd8f4208d63b71d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:55:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbc1c8779d5fc1a0859ec5bc20bfb732df5a3957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 21:38:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-publish-centos8-source https://zuul.opendev.org/t/openstack/build/1a73071a809b4826a2e27bb41bfc2add : POST_FAILURE in 2h 15m 49s\n- kolla-publish-centos8-binary https://zuul.opendev.org/t/openstack/build/fcb717457a1547c5aac9fb8680ffb043 : POST_FAILURE in 2h 41m 31s (non-voting)\n- kolla-publish-debian-source https://zuul.opendev.org/t/openstack/build/97797f06cf254d2386a729c996327fd7 : POST_FAILURE in 2h 04m 22s (non-voting)\n- kolla-publish-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e09ba80a30784d08a513124f1e81df14 : POST_FAILURE in 2h 34m 14s (non-voting)\n- kolla-publish-debian-binary https://zuul.opendev.org/t/openstack/build/554584e978f24bd2b754b4cf284f42d5 : POST_FAILURE in 1h 35m 34s (non-voting)\n- kolla-publish-ubuntu-source https://zuul.opendev.org/t/openstack/build/612a80c6690e484f87a2e027159b82dd : POST_FAILURE in 2h 36m 06s\n- kolla-publish-ubuntu-binary https://zuul.opendev.org/t/openstack/build/20de3466c5334d0581baeb69df7b870e : POST_FAILURE in 1h 41m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aff1f452a30241f32459f82905562aa7b37b5920","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 14:03:27.000000000","message":"Patch Set 1:\n\nwe have current images for Victoria; switching this patch to run publishing to quay on a daily basis","accounts_in_message":[],"_revision_number":1},{"id":"5e3625bd42b1ef56c21d5badc93dc5332fced7d5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 14:04:00.000000000","message":"Patch Set 1:\n\nbut first let\u0027s backport to all supported branches to publish their images","accounts_in_message":[],"_revision_number":1},{"id":"eddba981d17722c9af8928ca5022108b7829ba8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 14:09:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/431a9e1601bb4465a4deeda192051c40 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"842f8902a66f0b0e87e7f34fbf96491d3d9ec15f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 14:09:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2adaca71f20a5f2be298d19fdbf9184c74d58e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 14:11:41.000000000","message":"Topic quay.io removed","accounts_in_message":[],"_revision_number":1},{"id":"dfa12b3003d46d67194412e30b7bcd791eca3c80","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 14:11:57.000000000","message":"Topic set to kolla-quay.io","accounts_in_message":[],"_revision_number":1},{"id":"42dee2c62e29a3f4dc06a85b3199133992e28310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 16:02:37.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e05c1f4ac8b110fbb0485f3828ec2ad553c0eda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 20:07:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-publish-centos8-source https://zuul.opendev.org/t/openstack/build/4f5cd3eb86f242978fd7fb5a3fce4a84 : POST_FAILURE in 2h 36m 48s\n- kolla-publish-centos8-binary https://zuul.opendev.org/t/openstack/build/53414afb626a4b719f57acb407fd2cac : POST_FAILURE in 2h 00m 34s (non-voting)\n- kolla-publish-debian-source https://zuul.opendev.org/t/openstack/build/9bb2a703867c4342badcb088222d924d : POST_FAILURE in 1h 46m 14s (non-voting)\n- kolla-publish-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7aee3fc57f9949a19bf9128b77623388 : POST_FAILURE in 2h 38m 34s (non-voting)\n- kolla-publish-debian-binary https://zuul.opendev.org/t/openstack/build/511ab401ed0b455bb2302dafde267e49 : POST_FAILURE in 2h 11m 19s (non-voting)\n- kolla-publish-ubuntu-source https://zuul.opendev.org/t/openstack/build/603047509a6d43aa87204c3939bd11da : POST_FAILURE in 2h 09m 26s\n- kolla-publish-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ac58006c5d647c7bf64cd57615ff51e : POST_FAILURE in 2h 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce8106257a9bbee1d8c7b5d7c23795a8fe38b4d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 12:50:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ca0544e100bd9b16f15e1ef92929765144093f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:50:59.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 19, column 1","accounts_in_message":[],"_revision_number":2},{"id":"594e38b157e13f02b5a52d4c073e8dbbb8ec94be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:51:06.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 19, column 1","accounts_in_message":[],"_revision_number":2},{"id":"06e0c5cf6f3ac22744786500175f1a8438d25be4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 12:51:54.000000000","message":"Patch Set 2:\n\none big conflict 😊","accounts_in_message":[],"_revision_number":2},{"id":"10ec2b83e6811618a757e9e6467fff1afb65553e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 12:56:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9944a1b7a03c988dc4fc59f067966acc2b5c977e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 12:58:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"edf23b54e61316c4973f115e1ffaa2588dc38214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 13:01:57.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6cb6d5451ef95c85e36cd6ed437e7111cc963d45","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-08 13:07:06.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"25e3b9d140ac7faeddef4d5949d7ec001c86d7a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 14:34:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9c2050a26d1a493086912954053a928f : SUCCESS in 28m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b593a67e09584678a0e819d5e33820c3 : SUCCESS in 1h 10m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bf6cea8873324bf2b5bde8e5e5e13e66 : SUCCESS in 1h 07m 47s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b414a0541a134737bac28896126424ee : SUCCESS in 1h 27m 04s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/326857f62b314fe6ad940e56ac820576 : SUCCESS in 1h 15m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4ed3cdd54bb047e6a8384d8e5eb9a30e : SUCCESS in 17m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/32d69107640a4e02a135fcc670d3bcaa : SUCCESS in 44m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3137c3106d934358abb4bf1e8904bcd1 : SUCCESS in 57m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb2987ce9f14826aba75bc5357feaf9 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97abbea1c01a4050a7d6b737bd2067ea : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7980b5bdf41420e847c4e7e3f860fc9 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b54609637d324121ae994efd0a3b0a66 : SUCCESS in 6m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/35d5989d2beb4406a0a09d0e0e8fac44 : SUCCESS in 23m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/82dc5debbe2343a7af8a05f467dde3d3 : SUCCESS in 24m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7a977e68e4140d8986b846e2196abb1 : SUCCESS in 53m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e89ed42e3b654962a1e9ca2015101ffd : SUCCESS in 55m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/253277d7aaab465b88b381456b415b59 : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":5},{"id":"30988c6f742fe26fd1ee15a191ddf3162cfe67fd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-08 15:23:08.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b26ef0284b425cf6c5a819a6cfcde4d403a9cb80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 15:23:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c817bcfbe91eb859466d79e1ebc0abca5910826f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 16:42:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/36a082710b8f406f8ba994b438633706 : SUCCESS in 26m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c85cd98199da424395309908909f47cc : SUCCESS in 1h 03m 43s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/80f203ae89a94909b6b8dbe49ddf0040 : SUCCESS in 1h 13m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91dc0598e4fa4fc8ab764705be6c6896 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a27cab30b97e4e8baaf156bc1cc4721b : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0cf5d3a93bf40a7a001321283380aa2 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ff18630fa0420e8047ae30c66b5d7e : SUCCESS in 6m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8854e9f33a25453ba757b94700d9ebec : SUCCESS in 20m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4569860d49fd4ead874ab985560f8cad : SUCCESS in 46m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9536ba625694f9db57587bca2d0147b : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":5},{"id":"2af0ad5064ee3a02c08d6c53b598959e8b0b9e88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 16:42:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"71e3f60a3e2addecd9cb05c598938dad78e1a053","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 16:48:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2896cd1712664edf829df2986aba195f : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":5},{"id":"4e9c4634444ca8b7ab25a904dd8dd09f590787a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:50:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a695f987756c48f588f0f94824ae53d6 : SUCCESS in 1h 39m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cc310a525c9c2f1add957abf7a10caa5715d39d2","revisions":{"03f794745bf7bbc1378344a637d6867b9b86fa3f":{"kind":"REWORK","_number":1,"created":"2021-03-19 19:50:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/781899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/781899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/781899/1"}}},"commit":{"parents":[{"commit":"f72a4b36e49cb4c6f86f2ad4069f77892124b9fd","subject":"Make kolla_set_configs open files in binary mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f72a4b36e49cb4c6f86f2ad4069f77892124b9fd"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-19 19:50:12.000000000","tz":0},"subject":"[WIP] quay.io","message":"[WIP] quay.io\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n(cherry picked from commit 5fd235065d1dd90a251e69456b76455d41a78773)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03f794745bf7bbc1378344a637d6867b9b86fa3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03f794745bf7bbc1378344a637d6867b9b86fa3f"}]},"branch":"refs/heads/stable/victoria"},"38577c29575cfb346189023b4201cf071b9b3275":{"kind":"REWORK","_number":2,"created":"2021-04-08 12:50:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/781899/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/781899/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/781899/2"}}},"commit":{"parents":[{"commit":"302f4a9cdb23ce7a1e780e63615089c6478bcdff","subject":"base: configure curl before first use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/302f4a9cdb23ce7a1e780e63615089c6478bcdff"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 12:50:13.000000000","tz":0},"subject":"[CI] Publish images to quay.io on a daily basis","message":"[CI] Publish images to quay.io on a daily basis\n\nBased on the discussions from the past 3 Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nWe still publish weekly to the DockerHub.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\nDepends-On: Id113a53c083d105ac9200474ff20faacef6decaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38577c29575cfb346189023b4201cf071b9b3275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38577c29575cfb346189023b4201cf071b9b3275"}]},"branch":"refs/heads/stable/victoria"},"1a38aa5007af01c7b27a042c30adbfa63c962baf":{"kind":"REWORK","_number":3,"created":"2021-04-08 12:56:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/781899/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/781899/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/781899/3"}}},"commit":{"parents":[{"commit":"302f4a9cdb23ce7a1e780e63615089c6478bcdff","subject":"base: configure curl before first use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/302f4a9cdb23ce7a1e780e63615089c6478bcdff"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 12:55:38.000000000","tz":0},"subject":"[CI] Publish images to quay.io on a daily basis","message":"[CI] Publish images to quay.io on a daily basis\n\nBased on the discussions from the past 3 Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nWe still publish weekly to the DockerHub.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\nDepends-On: Id113a53c083d105ac9200474ff20faacef6decaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a38aa5007af01c7b27a042c30adbfa63c962baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a38aa5007af01c7b27a042c30adbfa63c962baf"}]},"branch":"refs/heads/stable/victoria"},"2ad9ae440c349a2ea607ad33a9ca380eca75ce80":{"kind":"REWORK","_number":4,"created":"2021-04-08 12:58:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/781899/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/781899/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/781899/4"}}},"commit":{"parents":[{"commit":"302f4a9cdb23ce7a1e780e63615089c6478bcdff","subject":"base: configure curl before first use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/302f4a9cdb23ce7a1e780e63615089c6478bcdff"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 12:57:34.000000000","tz":0},"subject":"[CI] Publish images to quay.io on a daily basis","message":"[CI] Publish images to quay.io on a daily basis\n\nBased on the discussions from the past 3 Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nWe still publish weekly to the DockerHub.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\nDepends-On: Id113a53c083d105ac9200474ff20faacef6decaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ad9ae440c349a2ea607ad33a9ca380eca75ce80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ad9ae440c349a2ea607ad33a9ca380eca75ce80"}]},"branch":"refs/heads/stable/victoria"},"cc310a525c9c2f1add957abf7a10caa5715d39d2":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-04-08 13:01:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/781899/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/781899/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/781899/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/781899/5"}}},"commit":{"parents":[{"commit":"302f4a9cdb23ce7a1e780e63615089c6478bcdff","subject":"base: configure curl before first use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/302f4a9cdb23ce7a1e780e63615089c6478bcdff"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-03-17 17:05:22.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 13:00:45.000000000","tz":0},"subject":"[CI] Publish images to quay.io on a daily basis","message":"[CI] Publish images to quay.io on a daily basis\n\nBased on the discussions from the past 3 Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nWe still publish weekly to the DockerHub.\n\nCherry-picks in Victoria and below include the switch to\nDocker module for Ansible.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n\nChange-Id: I7c1f0c5fc643678b29f963b6769d8f900a03b726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc310a525c9c2f1add957abf7a10caa5715d39d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc310a525c9c2f1add957abf7a10caa5715d39d2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
