)]}'
{"id":"openstack%2Fkolla~782247","triplet_id":"openstack%2Fkolla~master~I21c3a0e423235301269d0a5b212d0ce6970ab88f","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I21c3a0e423235301269d0a5b212d0ce6970ab88f","subject":"rabbitmq: bump Erlang requirements to v23 in Debian","status":"MERGED","created":"2021-03-22 12:38:50.000000000","updated":"2021-03-25 19:40:46.000000000","submitted":"2021-03-25 19:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"782247","meta_rev_id":"667ad50617bea5fda3b1085e85bcac9c9f691cdc","_number":782247,"virtual_id_number":782247,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 19:39:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-25 13:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2021-03-25 18:11:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2021-03-25 18:11:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 14:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 17:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 20:30:14.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-03-23 10:01:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-03-23 10:07:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-25 13:08:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-03-25 18:11:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"9f68853088f8d2378734843c6e4468471f662137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-22 12:38:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"617ebeb3e8f026df30a58b8bf06530ce04ffef7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 14:11:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/06ab43dc90704ab4942b40074a97f239 : SUCCESS in 1h 07m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ef374d8a6824dcd93c3b034aba6b36b : FAILURE in 1h 32m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/85d2d80c603b4e95b4294952bc4dd447 : FAILURE in 1h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0488da3efa23253a356e5576d09adf4a635914d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:40:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/87ff80e8ddf44edda90eeb41e5469740 : FAILURE in 1h 34m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1907b4450ace4d4e9249e46992e96dd0 : SUCCESS in 1h 47m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3c1bf83bb2314ce2bb7ada8ba6fe0434 : SUCCESS in 1h 14m 07s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/30adc689130e4e23840beb266192964c : FAILURE in 1h 06m 06s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5d2553437b304b779d8b30816d3db79d : SUCCESS in 1h 42m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6fc6e8e769a84b829d46ca64dd2bab78 : SUCCESS in 2h 31m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/693b0a346a7a4048be228f702c6e9a86 : SUCCESS in 1h 46m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dbbea7d39def4e528bcd091b2d652fed : FAILURE in 53m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d36479716543d1a04621a620165269 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e79f92075f654cd59e7f3ea9463cc218 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aa503b17df4447fa3086f50f14bdae6 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4039b1317a8d4a67953fc3e336268994 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b01a3bc3254b451e9b859cb1ba61e799 : SUCCESS in 10m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c5e4e0ec79ce4c478742183fa16eb464 : SUCCESS in 22m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c53212c99e344afdbf0b524375cd87b8 : SUCCESS in 3h 01m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92736cf9e92e4036b84f791ceac3bdd5 : SUCCESS in 50m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f9a973e8bfa540d4872c5acc9f093b4c : SUCCESS in 39m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d4ecf87e580e40a392075689e1edd9c1 : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8b7ff8a376d1fc8891f01301910c894ef5021b7f","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-03-22 20:30:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78bfc84e9916cfe17accd7bbe4c249fb7f02f83f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 10:01:46.000000000","message":"Patch Set 1:\n\nhttps://www.rabbitmq.com/which-erlang.html seems to agree 23.x is supported.","accounts_in_message":[],"_revision_number":1},{"id":"f9a6dc97c87ab01ddee9a9dc1eb4bff917e17cf2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 10:07:16.000000000","message":"Patch Set 1: Code-Review-1\n\nDebian job failing.\n\nAttributeError: module \u0027libvirt\u0027 has no attribute \u0027VIR_CONNECT_LIST_NODE_DEVICES_CAP_VDPA\u0027\n\nhttps://zuul.opendev.org/t/openstack/build/dbbea7d39def4e528bcd091b2d652fed/log/primary/logs/kolla/nova/nova-compute.txt","accounts_in_message":[],"_revision_number":1},{"id":"7c6cd60e7bc1cf101c5803268085a2c9a5ac7d00","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-23 11:06:26.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Debian job failing.\n\u003e \n\u003e AttributeError: module \u0027libvirt\u0027 has no attribute \u0027VIR_CONNECT_LIST_NODE_DEVICES_CAP_VDPA\u0027\n\u003e \n\u003e https://zuul.opendev.org/t/openstack/build/dbbea7d39def4e528bcd091b2d652fed/log/primary/logs/kolla/nova/nova-compute.txt\n\nThis is not related to this change. I am working on fix.","accounts_in_message":[],"_revision_number":1},{"id":"cfd7e052b58e5f5592e46d19a6ca11aa5b074d06","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-23 11:37:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"000079968174fd26c985a6272a10cf437099d8aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 12:59:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb24f7ba21e448848246ceede0678866 : SUCCESS in 1h 09m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a350fce767834715a397dfd688dd9de4 : FAILURE in 1h 12m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/27e503a82d17421b853a0f1f4fa47693 : FAILURE in 1h 08m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6914b795e5d2c801ea0d872e2444b8796ec346cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:10:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1b0ab6335a0945bcb462e57375a2e01c : FAILURE in 25m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d65e71ff6ca04c0f86ddbcd9261d96b8 : SUCCESS in 46m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/51986cd978ad47f083d0004259c60d8a : SUCCESS in 1h 07m 46s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0829dcb06dc04976a3212105db382193 : FAILURE in 24m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b0d199c760054d34a92726736e6646db : SUCCESS in 1h 23m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d4acd7a43d274344aeb4e0235acd6cb3 : SUCCESS in 36m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7bbe1ce489494f11be2a3c4b66f947a4 : SUCCESS in 22m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0003f10e5db94f0db10626af43acd2da : FAILURE in 47m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad712ed67a840c9a21bff179c334486 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52b48564939a437fb69d12ad293797ee : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d38acc0a85944963aa352079f7b1686b : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c37cc75b5c94b2a8548f49740ae3557 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc5ffd8cc1a4684a6acde0b96f6a364 : SUCCESS in 6m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a0d7bff078f7456192367d2fdced07d7 : SUCCESS in 23m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b35c89dcf79840f8a7c19e0956add5c2 : SUCCESS in 37m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f4f1ac93aad4e4a873e951d07b0906a : SUCCESS in 1h 01m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d53b0b203fcf44d780cfee3f8aa3169a : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da5d2b4d88374f8c8d3ee1de56adcfa1 : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7be57049ac681b4884e31720a0ae859ad3c08a2f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 09:13:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ac785f6a79f4c77585fb244f582a7b4eb2e4d64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-24 09:28:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d87f31e5a1b2ce3e203d3e0a30daeda85e1e6b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 10:42:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5ef611cc035a4beda25785ed5e30f107 : SUCCESS in 21m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3db82c4699ef40a38110f3e59ebc0746 : SUCCESS in 44m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7bb114b39cf6454c9ef074c085d5dc5b : SUCCESS in 1h 01m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c1917a44ddeb47d38b7bf4aecb02edf8 : SUCCESS in 1h 04m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/44603728cee04e92a49879900246943e : SUCCESS in 1h 10m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/af4f6bae8a0e47fab9309cd2ff38358a : SUCCESS in 37m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/59aa7acb8add4e738e4d98c83a252421 : SUCCESS in 18m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f6a5f69cacc74f50930895a4bf19761f : FAILURE in 41m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d1bdd27a5c428a9f66f8a33f374ac7 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8268145f2d574e6db85ba1fddeb86309 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5692ac931d9f461a98d2ad373ddc1f3e : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3489320ffa924fdcbdc8334bfabe9f35 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11a7ad9dd834ec6bf0271b905d3d81e : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2e8bc797e80049079e713430a31af5ab : SUCCESS in 22m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/01b017010b2a4926848e96ca78955662 : SUCCESS in 29m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3cad147a1264415955c645b67dd1428 : SUCCESS in 49m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85640fb9a77d47c0a9958ed78c9f1c94 : SUCCESS in 48m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e928d71cfcca4904b181c5d093f48831 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2b5a0365a65ff2d6eb9c56c3ab8e0fc400a4d90f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 11:15:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/688df22d539b48ff8f0ab05f6293b1bc : FAILURE in 1h 20m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c6e6ce5ecbbd4eb4947101fd11bbd748 : FAILURE in 1h 31m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c7120312a634fb697970388dbe09d62 : FAILURE in 1h 09m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6eabe63b42f738b89c3ef3aa1582a1c1e03e68d1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 13:08:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"13ac32e0fa355836c985d7750b07ebe52ca1989a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-25 18:11:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"be20187aa54f09871b847033d2d6e22fdb2c1c53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 18:11:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2e60035a3201a885df09d4bf45c4149ffa20ca98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 19:39:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/26e6a96e2c0943e48e8b1891e38735e0 : SUCCESS in 39m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a200067a149e468bb523e46042ad34d7 : SUCCESS in 1h 09m 13s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8bb76f5186a4463ea1a741ec9d8ab8d1 : SUCCESS in 1h 22m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d278b460c484ff182666052137e4cc9 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be8c3e15957042f8aa69d0624414fd73 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb1e51583a9e4f95af2097acebf03e4e : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86232fa7fd2d4f8ba4bdbc93e48206ce : SUCCESS in 5m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/263a5f6c782d4c8f9b7686c9a357a195 : SUCCESS in 43m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f31d2b5504fc4179bd92d84b41d1290a : SUCCESS in 57m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/49c86d689bab4842821cadec6b7cda13 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":2},{"id":"973504699b4992b9b288daf0b6c4d5673b1fd24d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 19:39:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"667ad50617bea5fda3b1085e85bcac9c9f691cdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 19:40:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7bdc6584d864b768e77b815b8878805 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c49148864b0769f8537ae909421b4923277d944","revisions":{"59225fea5f3cba456e94555ed0a031d61c67c8ab":{"kind":"REWORK","_number":1,"created":"2021-03-22 12:38:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/47/782247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/782247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/782247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/782247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/782247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/782247/1"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-22 12:36:52.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-22 12:38:37.000000000","tz":60},"subject":"rabbitmq: bump Erlang requirements to v23","message":"rabbitmq: bump Erlang requirements to v23\n\nCurrent rabbitmq requires erlang v22.3+ while we had v22.1 on aarch64.\n\nOn x86-64 we use upstream repo so have both v22 and v23.\n\nOn AArch64 we use Linaro backports repo so have v23.2\n\nChange-Id: I21c3a0e423235301269d0a5b212d0ce6970ab88f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59225fea5f3cba456e94555ed0a031d61c67c8ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59225fea5f3cba456e94555ed0a031d61c67c8ab"}]},"branch":"refs/heads/master"},"8c49148864b0769f8537ae909421b4923277d944":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-24 09:28:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/47/782247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/782247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/782247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/782247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/782247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/782247/2"}}},"commit":{"parents":[{"commit":"be544f4e66a6d0dbaa0c26e4565ccfe6a2548838","subject":"switch to CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be544f4e66a6d0dbaa0c26e4565ccfe6a2548838"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-22 12:36:52.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-24 09:28:01.000000000","tz":0},"subject":"rabbitmq: bump Erlang requirements to v23 in Debian","message":"rabbitmq: bump Erlang requirements to v23 in Debian\n\nCurrent rabbitmq requires erlang v22.3+ while we had v22.1 on aarch64.\n\nOn x86-64 we use upstream repo so have both v22 and v23.\n\nOn AArch64 we use Linaro backports repo so have v23.2\n\nChange-Id: I21c3a0e423235301269d0a5b212d0ce6970ab88f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c49148864b0769f8537ae909421b4923277d944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c49148864b0769f8537ae909421b4923277d944"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
