)]}'
{"id":"openstack%2Fkolla~782386","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I1b3e879ec8752c63dc3f1ec3d8f62c3b43df55f3","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I1b3e879ec8752c63dc3f1ec3d8f62c3b43df55f3","subject":"Debian: use QEMU 5 from buster-backports","status":"MERGED","created":"2021-03-23 09:45:26.000000000","updated":"2021-03-31 10:35:50.000000000","submitted":"2021-03-31 10:33:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782386","meta_rev_id":"159954926e51c4bc68ff541277d0b8cac62a3a0f","_number":782386,"virtual_id_number":782386,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 10:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-26 09:37:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-03-26 17:44:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2021-03-31 08:05:51.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-03-30 15:07:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-03-24 15:19:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-23 11:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 11:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 15:19:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-26 09:37:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e22850c7805da1fbd2af3ed1bfa5d60f1f5d8df1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-23 09:45:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f02d5a0b5d6eaf3c7c80d0ebf6e9ea8964e993cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:06:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e5e3af0a9a0b4cc9bd404fd92b0c4e60 : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5eb5b4a884120ea45d682b3ede166ce9158224c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:47:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e7e0218f940d418d81f6afe47a8e96c5 : SUCCESS in 26m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/06ebd5c79ed84520a77f3c21948a3503 : SUCCESS in 1h 59m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/db1c04f4054a492fa006f44fd10430d9 : SUCCESS in 57m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8ca4e8755f854659b855a990513cdd29 : SUCCESS in 59m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/64b5816239d8426b81153acfe98f0efe : SUCCESS in 1h 08m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a6bc5b9e453f4150a7ed158a32a8475a : SUCCESS in 22m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dafdfc6d9e8e4c1d829f2f718f7048f1 : SUCCESS in 20m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c6c477cbf2684c6687b5be0fb69f7367 : SUCCESS in 58m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2921e5b8eef49eabfd4d848ef6d06eb : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de64a24e0444b139950cc4c9f094b4c : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fd17ebd10fe419495665ad99315c315 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29da812ec8ad4c7cbd71a9bd239f692b : SUCCESS in 8m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d44fae29b7bc4d96b1c120e3f82971b9 : SUCCESS in 22m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e77f27afebb4dbabf7a11dde847a50d : SUCCESS in 35m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/686c6eef5d0a401fb8d7a21f7f4ddae0 : SUCCESS in 55m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6f8defb5551452d8560b5279f52b1a6 : SUCCESS in 58m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/825c3eea2ebb4a0f9954a7bc70687a93 : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":1},{"id":"920671a1ee0ce4928872488512db10d7ec97a77b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 15:19:47.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"261f46aecdf6e38417b561f290b14f06f35db087","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-26 09:37:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"97685d1c83d67b9d634018fa89a7b9188a3e5fa4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-26 17:44:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b40b368753fd0e4e3e5436b98b19b61e20f9f411","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 17:44:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db4e6038303da512f2bfdb3b542fff3c67d05361","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:39:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7f22a9712c5b493a981f51993babe2ec : SUCCESS in 38m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/dfe61d5537a6493c834e17aa7d3c25af : FAILURE in 9m 03s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8d9fab7f7e3a4f4b9c10d6e8d0fba1d8 : FAILURE in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e4309f9443548e784db22155772dde7 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/489f115826644386a9168ff3a02f7ee2 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fd294e72e294e328b00132870e05daa : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8f2783957048c8b45d1c2a569f2faf : SUCCESS in 5m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3212ff6aae584251b8fd6ce15cc81cb1 : SUCCESS in 37m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28d6d2f36dbc4f2ebd823f7f43fdb567 : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3ac6c8a916bc4c5eb4b4f66491fd351c : FAILURE in 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e3d0fcdd91b9fbf6b2250dbe6be2f437578d62ec","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-26 22:44:52.000000000","message":"Patch Set 1: Workflow-1\n\nneed more testing","accounts_in_message":[],"_revision_number":1},{"id":"f241091712fa1f8b553b9d097b946cffc5db9946","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-26 22:44:55.000000000","message":"Removed Workflow+1 by Mark Goddard \u003cmark@stackhpc.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"9bb5ecbbcdbc022ff8348fc9dfa2f7ce2880aaf4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-30 09:44:51.000000000","message":"Patch Set 1: -Workflow\n\n(kolla) 11:44 (s) hrw@puchatek:kolla-ansible$ openstack server list\n+--------------------------------------+-----------+--------+--------------------+--------+---------+\n| ID                                   | Name      | Status | Networks           | Image  | Flavor  |\n+--------------------------------------+-----------+--------+--------------------+--------+---------+\n| 6725dd18-2df1-4a0d-aa13-128c0664cebb | demo-Ng1o | ACTIVE | demo-net\u003d10.0.0.23 | cirros | m1.tiny |\n+--------------------------------------+-----------+--------+--------------------+--------+---------+\n\nworks.","accounts_in_message":[],"_revision_number":1},{"id":"76b7c77fb945e0aa7f2966ab832510d7a683a7fd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-30 09:45:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2531c248f635370eaf08bd96a33cdf6e4ecdf329","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 10:51:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/119d1d8a6de944f89052bdb46b7d67e1 : SUCCESS in 1h 02m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cdf3eea2e8c479aa0686d777e6ab741 : FAILURE in 58m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b47bd3739582408f8690d36439a83a1f : FAILURE in 36m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3371a019401c4b1193b9620ea84a17e2 : SUCCESS in 57m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48aa56ba4702537f6de05b0b7566eb86398b105c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 10:55:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3f9c60e87262443180ac25c0932ac11e : SUCCESS in 20m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/857817aabd5848f68f8ac04a9c74fc25 : SUCCESS in 37m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/984778563bfb483f813f7722b47bd3e0 : SUCCESS in 1h 01m 44s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6c53919f40834edc9ca4986a89cf17b9 : SUCCESS in 1h 06m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d3ac70a4aab64364b1df1c286c821efc : SUCCESS in 53m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/75707b242ea04da090d434171f1558c2 : SUCCESS in 29m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/33e78a1b570a49bebdf7e6f533421555 : SUCCESS in 24m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1a082d3841ad4ee3acfe2dbeda2703d2 : SUCCESS in 55m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c1ded913cec4724905c99aab666e966 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/558d6effdf1046669f00cbd062223dc9 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8245e212e57648c9aa62d334093ce73f : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d0f16727904128bd82486d7b9d44d9 : SUCCESS in 6m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f89f8abe1ed46f884ddcce6390f920d : SUCCESS in 32m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c93afbda8dbd43ec87d6bd0255925a69 : SUCCESS in 25m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fbce2f8b6a18434f9b9fac0132ee871e : SUCCESS in 1h 02m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ddae137736b8482e91b2def9ed683d7e : SUCCESS in 57m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8458edda44e34e719d3ea6af351dd514 : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6ea842e3a530287e955bbf204c2623b3bb158ecb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-30 15:07:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1a000778c1ef37c07021359b537b9de4929b750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 15:11:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21a3a3473abc41685165c0dcd4e1178003b99311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 17:50:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/03f41a01faf342acaaf93a2945d02447 : FAILURE in 27m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e7033a6369934fafa4edc18470361ea9 : SUCCESS in 1h 00m 17s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8e2061a04b324cb1a07eaee545636e05 : SUCCESS in 57m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5f3f5510d048c1874623d04ccea8d4 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96f8bf0db86d42f6a80cc5a3a5fc9f2d : POST_FAILURE in 24m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d31c043683ae4764a1970963dcedfe73 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/622c55cbf15c4444a3d6f0473349a0ff : SUCCESS in 7m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e87102f5291470ca7724a4b124fc6d0 : SUCCESS in 28m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23c9a86f53b84d0e96964fd96cffb3a1 : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7db08cf6c15243238019086b371ed813 : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b4d0e2f4290cedc3eca645a5c42083c0fe509631","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-31 08:05:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce179dc8256889c014fc1e82c312d9678474d035","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:01:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/46547ec95f9040768300029dbd24ead9 : SUCCESS in 47m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c7ed6d04cf5145c7b52adc58ab53fc5a : FAILURE in 45m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e429568b38704e3388661b1f6643ebcf : FAILURE in 32m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/018c46816d0f42ef8464e00c6e271658 : SUCCESS in 45m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"342a5d39e9216501a41403461862ea98f0d3586b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:15:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/45567fe4dc3e4723befd4520a4acfb06 : SUCCESS in 34m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4418c05c166240d791032f480b7eb45e : SUCCESS in 37m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/02c5a10e41524b5abe3fde724c19cd4b : SUCCESS in 1h 05m 20s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/662f9172505d4be794f419295abce5e7 : SUCCESS in 58m 20s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a572220b1ebb4ee48994bf21a30594b5 : SUCCESS in 1h 07m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/daab53c0a3594b34a1556b1c788370f5 : SUCCESS in 24m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0a3ba91b32d14e8e8d1db487fabdadb1 : SUCCESS in 21m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3390f898d5db4f988fd97889052e56a7 : SUCCESS in 58m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99dc5fc6bef8451099d167a9874cd294 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a1cab67ea8a4d0db913cf021a1bce75 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef4aa59be38a492e8c0094c93376df2d : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d100a5cf7b4a688493033cd2af68f2 : SUCCESS in 7m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73c2b144355c4d9681218c91d5fbb95c : SUCCESS in 28m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7663ca944b4f40ffbe10f0543a970a31 : SUCCESS in 36m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/beb83660e1d04072ac90272fa3543bbd : SUCCESS in 59m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d2b37b090f4045f18bc1a3c6503752b7 : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/12df0de2ee4246e7a340311bb094502f : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e5b3dbe516c46bba59bf4d3c81b49af12d8f0c41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:16:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8dc33250914781efa61d174476d52fc450feb8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 10:33:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3503ba96599746668f3ce6ee7e0c5a25 : SUCCESS in 27m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/75f239e01b8744288c0b34591e7bfa01 : SUCCESS in 53m 42s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/dca84358730f4eb687fd3713c74c3cba : SUCCESS in 1h 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c500ac4aa740379897befd452c6416 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8ac3cb5eb414cd79634843042605f41 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/891a17ce01dc440e983c0c39c0d5befc : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b558402dc3894c9c9aa2266b53a8c092 : SUCCESS in 6m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3564c88a2d4c4fbb9fcb62692cf0450e : SUCCESS in 34m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f66e16fdca345e189f299f3e6b161c7 : SUCCESS in 59m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e907374ef0184a61acffe7bfa3360e8f : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":1},{"id":"51380d5a514a231e9335784b0ccd91f1a6303f4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 10:33:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"159954926e51c4bc68ff541277d0b8cac62a3a0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 10:35:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1250af5b6634dc88cc59dabdcb348bb : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3de3fea92f63ca45145297b034025f543383bf66","revisions":{"3de3fea92f63ca45145297b034025f543383bf66":{"kind":"REWORK","_number":1,"created":"2021-03-23 09:45:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/86/782386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/782386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/782386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/782386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/782386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/782386/1"}}},"commit":{"parents":[{"commit":"8a8c7bbffaddef0fa678aeabd0bd1012466dbe8f","subject":"Fix Ussuri builds due to RDO Virtualization GPG key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a8c7bbffaddef0fa678aeabd0bd1012466dbe8f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-23 09:35:44.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-03-23 09:35:44.000000000","tz":60},"subject":"Debian: use QEMU 5 from buster-backports","message":"Debian: use QEMU 5 from buster-backports\n\nThis will allow us to make sure of all updates.\n\n\u0027qemu-kvm\u0027 got dropped in Debian\n\nThree changes from master are merged into one:\nChange Id: If09a8188030baa284dd353b1c3f830d78c5091b7\nChange Id: I4371daf7b28c8a057364d17e2e47f6e22665447c\nChange Id: I2c82c6269b2c357b0815ea29bc6a7c322f84dac9\n\nChange-Id: I1b3e879ec8752c63dc3f1ec3d8f62c3b43df55f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3de3fea92f63ca45145297b034025f543383bf66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3de3fea92f63ca45145297b034025f543383bf66"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
