)]}'
{"id":"openstack%2Fkolla~782606","triplet_id":"openstack%2Fkolla~master~I1c91d27f2578f5ca7c83c4747725b1d9371880b0","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I1c91d27f2578f5ca7c83c4747725b1d9371880b0","subject":"enable \u0027libvirt\u0027 repo for images with python*-libvirt","status":"MERGED","created":"2021-03-23 21:08:26.000000000","updated":"2021-03-25 20:21:14.000000000","submitted":"2021-03-25 20:18:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782606","meta_rev_id":"1b71e2c7aac441734becb0c8c4374c407d4b111e","_number":782606,"virtual_id_number":782606,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 20:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-25 13:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-25 10:45:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-25 13:07:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-23 22:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 23:08:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 09:12:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-25 13:07:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2826ee5310cb5e604fae8b38c01102d9f68569a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-23 21:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0d77cb16a09390fd2fb501713f5960e2b917332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 22:45:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/48980155e5214219aba84fc47d356439 : FAILURE in 57m 56s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1dcda8528a814eaebb901e3f703e1c69 : FAILURE in 36m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a38606de1b24d1196741ac36a6547c1 : FAILURE in 51m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"469458e2a232eb5669f831f9a90a19095a253233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 23:08:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a92aee175a434345b71d8685202b9334 : SUCCESS in 31m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/25410e61059b4583a1bdcde3e5a35672 : SUCCESS in 47m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ca21831490364a0eb950cb045de5c535 : SUCCESS in 1h 17m 24s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fc62f8eea4fa41d3aa297139ce64336e : SUCCESS in 1h 04m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/da992f8f72534af4a2b39e1689ec389e : SUCCESS in 1h 22m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/778978e82b65442dbe7e2567ff302beb : SUCCESS in 33m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c9d3b17cc34a4928870bd171a207e7aa : SUCCESS in 18m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4d9f5a29af8b4c4eb876edcf5c129b50 : SUCCESS in 58m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d0f1206d14448d9c471d0b556c133a : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/833067e3964346338ee10a5527058e93 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2344ad46501946ab871eb1945df60ff9 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc8e8793202440a0a8fed1069a987883 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d6272fc3b5d48458f263235b3df00c4 : SUCCESS in 6m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/60e0202dd80d4d4e9e3e8d83eca7477b : SUCCESS in 21m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5684d6898e8f40bb9ec2589e1be99346 : SUCCESS in 35m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/24eb3d1666594490894a6ecc3ccec775 : SUCCESS in 53m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1a2602eedab04f9295c51f1b1e3d55bf : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/419b52bbc41f42c2ad032175a20f8172 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a995fd8f3fb12d4f12454d8d199cdec4ce585249","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 09:12:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"903d8fbb7f5665dd413dadab5f89ec7174bd4b51","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-24 09:26:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b75e8cc28aa2be67fc5ab5ac06bfdd05d98f3fa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 10:45:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3dcfb9dc4eb57d2ee0448dac59c5881868de5b55","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 13:07:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"973f97f00ef9f7a1964934c8fdd31d961a99aaab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 13:07:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04fd32433a6e7b470903594fb5a971184e88f8e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 15:08:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d51e99df99884734ab03c2fc4d05232d : FAILURE in 35m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/73c95e5e912d41dab79f0254b2b39a85 : SUCCESS in 1h 05m 53s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/376c204fb4b9432b91273432a19e1070 : SUCCESS in 1h 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a53d93ec9cf48088881eee2b62ece61 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b54fcb8d49fc4c6cb96c2d7948825f75 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4cd143be8b84e08ae29ec632e49d1ff : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153a3c695b604c63a51014ae785f95f3 : SUCCESS in 8m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/07f8f9b84fd342fba3cecf59751a7d7b : SUCCESS in 35m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/91c1ad66e46b4508a3cb090f3f572e4f : SUCCESS in 51m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/24b5e5fd147348e7814008b75498f1bc : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d8902a7f6d4ebf93bbbe599ec894be87a5252b40","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-25 17:18:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8318798298cac5e357460ac36a4523745c0ac4ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 18:46:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e490059152aa49919ef96e3989296cc8 : SUCCESS in 29m 36s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/176b8fb5de5a483dadf8cbcee02becf8 : SUCCESS in 42m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5af603a7bda847c4af62d53e43f30e3c : SUCCESS in 1h 07m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b0dcf437fb7c42fa8d672f8c28bcc602 : SUCCESS in 1h 04m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/da1f7db4ce1247fbaa4ff43ce134a6e6 : SUCCESS in 1h 21m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/844adc08bb234486a6e20aa82cb72e09 : SUCCESS in 33m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4731b0386a5b45819841ddd3f01c1272 : SUCCESS in 22m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/452735022c9942ed848764aced55052a : SUCCESS in 57m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0beb2c39f4f488fa001b0439091dabc : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f6868459f754c9fb3fd9f140586a60b : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1bb85a3ca2b4aed888e36e2a5753dab : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2365e4c6fdbb4107b7a32d53e6bd0557 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8045101936e4bf6a26eb1ff91b1b524 : SUCCESS in 7m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e6a55612bb54c5f9a5937d1a027f2ca : SUCCESS in 25m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fddade44787f4e01b357e0bb0c77c7d5 : SUCCESS in 37m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/48884854d71b471b92ffa75a84b4d73d : SUCCESS in 1h 00m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d102d443c23493aa85beb433be24b3f : SUCCESS in 55m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b378db9ff6f491e92fd818e4906b176 : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b2c3ca18b453aec683225f6df0b3266ed850b9d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 18:47:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba4b8dfa3b3981c12cc49caac8747fef0f4279ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:10:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b9bd050619ec412a80af60907984f9f1 : FAILURE in 53m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/019a7aa9dedc442ea1aab64ce3354738 : FAILURE in 38m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/550a3b115f0643deb6e1edf6565041d6 : FAILURE in 59m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3eccdbe6b098ea318cbff3df626bc94267e0b63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:18:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0dfa5639ad0b45e3b78182a1261f0d02 : SUCCESS in 50m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/44143ce4b4294325845d1b1831be09fd : SUCCESS in 1h 01m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/abaebb3b20c247388f52a3e0a92ff42a : SUCCESS in 1h 24m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6972ec7eb24d2fa0f89c9419d54eba : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4adaec46917f4887bbd5cae724278fd8 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35c9b0f637ad421fb4be233cc7c8cdb3 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d34be7e7f85c42178197516afc6f532c : SUCCESS in 5m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/954094093ff24639a95c473acdf09932 : SUCCESS in 28m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0f21b3d967a432ba672f0e4ed0bd67d : SUCCESS in 57m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8bb2623b719e465fa5958ace108883b1 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f85b7f7ea8575314ebad225e9410f46f2ebbfd89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:18:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1b71e2c7aac441734becb0c8c4374c407d4b111e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:21:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b4f61daf6142999de915a4f6d1c139 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb61ba7daccfe8b13eff3b773cc09d69aa0039e7","revisions":{"fb61ba7daccfe8b13eff3b773cc09d69aa0039e7":{"kind":"REWORK","_number":1,"created":"2021-03-23 21:08:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/06/782606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/782606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/782606/1"}}},"commit":{"parents":[{"commit":"5c0c2d2448f88a82bcc29b6771f90202011cbb1b","subject":"centos: enable PowerTools repo in images which needs it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c0c2d2448f88a82bcc29b6771f90202011cbb1b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-03-23 21:05:10.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-03-23 21:08:11.000000000","tz":60},"subject":"enable \u0027libvirt\u0027 repo for images with python*-libvirt","message":"enable \u0027libvirt\u0027 repo for images with python*-libvirt\n\nThere are several images installing \u0027python3-libvirt\u0027 package. Which for\nDebian reside in \u0027libvirt\u0027 repo. So let\u0027s enable it where needed.\n\nChange-Id: I1c91d27f2578f5ca7c83c4747725b1d9371880b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb61ba7daccfe8b13eff3b773cc09d69aa0039e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb61ba7daccfe8b13eff3b773cc09d69aa0039e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
