)]}'
{"id":"openstack%2Fkolla~782619","triplet_id":"openstack%2Fkolla~master~I546bcf910e886eb75ece62f6be382fe1187efb32","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I546bcf910e886eb75ece62f6be382fe1187efb32","subject":"Debian: install newer firmware for qemu/aarch64","status":"MERGED","created":"2021-03-23 22:12:10.000000000","updated":"2021-03-25 16:14:24.000000000","submitted":"2021-03-25 16:09:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782619","meta_rev_id":"2017a8ea25d6af475052f4746fdfd990bd09049c","_number":782619,"virtual_id_number":782619,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 16:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-25 13:17:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-25 13:58:33.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-03-25 13:58:33.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-23 23:05:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 23:30:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 09:07:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-03-25 13:17:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-03-25 13:58:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"7c39e5c78b0c047764c19bb0a8de896998089d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-23 22:12:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19b87f65ead55c8922415c0ebfc33b8f643bbfa3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 23:05:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/363c8bc0468c4768b0a3a85f06f70746 : FAILURE in 47m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0de3410691c142d291a77898625ba6cf : FAILURE in 32m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/32107f224202443b980de63187ce111b : FAILURE in 45m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b445b00b0a6d7af5fe67ea4ea13685e460e510a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 23:30:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eacb1f4b53da44f4a4d8c502d617261c : SUCCESS in 27m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3fa8792677174adea74569955f7c2158 : SUCCESS in 45m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b509dbdf7da24b0698b1d0606fd9f0f2 : SUCCESS in 1h 08m 13s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f9c519fb00c644c4976a72a29dadbf3a : SUCCESS in 58m 06s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/379c654754b94ea59f6aa528ad09d092 : SUCCESS in 1h 15m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c89979384dae4f3dbcd7e7c907c9924a : SUCCESS in 36m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9073b16d639a4034a5ce01b6435db6b5 : SUCCESS in 23m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/03ddd3e594884c11814bb9ac1982fc00 : FAILURE in 40m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c78f5b3adf494545b491a7edef5296c4 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9848083c0d1499699d687c4c1dfb029 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8670c582a2a94fe9a559535b6f6b50fb : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/079e402d961145f793ba0bf8150b5d64 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f46a3fd18a439cb6652b195a76361f : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f03f720a35874af0bdb3b912be292c37 : SUCCESS in 23m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a830453a6aa744b8b01facf54a7b7e6c : SUCCESS in 35m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f7cc6c171ae41379682f60ba122739c : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e7bb6855aa84edf97d0514250dc2b6e : POST_FAILURE in 25m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2615f509abf544659c577fbe9a935f2d : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b67335ed80c6b48b01f67705db897930a738a596","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 09:07:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c6065846833e9be6fd2139f6b215f52ffe46c5f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-24 09:24:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f03c5e15ece5576a5082a9d7a61179567eb15d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 13:17:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6492bb45f0690ecb99fbd3ab575237a70c9fa8f6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 13:58:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"55904fe739c9e15b3d34ac527885d9de8e3bbb23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 13:59:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34a0b590f8b81d3656fb7812486072c7d1dd9941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:09:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cb02aeb3396b45518005539127a029a4 : SUCCESS in 48m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3e52683540364188af784f10dee5f299 : SUCCESS in 1h 02m 57s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8282646361ef4c609583de632ffb2ba9 : SUCCESS in 1h 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cd13e110c1045ef98f59680eb5dd371 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc9c60cf81634dfa990ce6102af7f5ee : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05f2c672ef4c4858819ae5f2627b7f7e : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862e097a532448edb7cd91fac5edbfdc : SUCCESS in 7m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/062d5c628be74bb1be16bcfeec1a3811 : SUCCESS in 40m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7259ab8ad1214a80885a7089afd6091c : SUCCESS in 1h 04m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c4361ec66b17465a92f7ccdc2e86e389 : SUCCESS in 1h 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a1ef6951e9440578c5b7103008d9f92706233323","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:09:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2017a8ea25d6af475052f4746fdfd990bd09049c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:14:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dd2e0d8ab3f4050b9354c823a7ad492 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc5c164eb85b09ff44e67764ca8e8fe30394e536","revisions":{"cc5c164eb85b09ff44e67764ca8e8fe30394e536":{"kind":"REWORK","_number":1,"created":"2021-03-23 22:12:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/19/782619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/782619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/782619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/782619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/782619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/782619/1"}}},"commit":{"parents":[{"commit":"5c0c2d2448f88a82bcc29b6771f90202011cbb1b","subject":"centos: enable PowerTools repo in images which needs it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c0c2d2448f88a82bcc29b6771f90202011cbb1b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-03-23 22:10:14.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-03-23 22:12:00.000000000","tz":60},"subject":"Debian: install newer firmware for qemu/aarch64","message":"Debian: install newer firmware for qemu/aarch64\n\nNova looks into /usr/share/qemu/firmware/ directory. We need 2020.05\nversion of firmware package for it.\n\nERROR: nova.exception.InternalError: Failed to locate firmware descriptor files\nChange-Id: I546bcf910e886eb75ece62f6be382fe1187efb32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc5c164eb85b09ff44e67764ca8e8fe30394e536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc5c164eb85b09ff44e67764ca8e8fe30394e536"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
