)]}'
{"id":"openstack%2Fkolla~782906","triplet_id":"openstack%2Fkolla~master~I92dab4394b7eebdcff7f90765cde078c1292312b","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I92dab4394b7eebdcff7f90765cde078c1292312b","subject":"kolla-toolbox: Upgrade openvswitch collection to newer version","status":"MERGED","created":"2021-03-25 06:34:50.000000000","updated":"2021-04-12 03:54:17.000000000","submitted":"2021-04-08 09:53:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"782906","meta_rev_id":"6572b1893df32da65edd85ea3eb636d222ddad83","_number":782906,"virtual_id_number":782906,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 09:53:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-12 03:54:17.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-12 03:54:17.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-04-06 16:21:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-12 03:54:17.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-25 08:00:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 09:16:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-03-25 13:59:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-04-02 19:18:18.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"dd776033112f6d3665bc602344a8429d53320518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 06:34:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"268b78795d7be4dfdc478a9f00e506f8bb21ce26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:00:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/16ee4076755b490797357601d4da615f : SUCCESS in 22m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c1eae050a5c34bb1b315af34bf2030de : SUCCESS in 47m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/078c2430cc6f45a48d341d7082ec3075 : SUCCESS in 1h 10m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/dec3659bd3b0417fb0c203d1d976547c : SUCCESS in 1h 14m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cb4465fbeec545ec94fef3f7419d2dfb : SUCCESS in 1h 19m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/18b30a9b4ea84fc69acbaa318200c0ca : SUCCESS in 31m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1030749e483b4e04b48b9990300da20d : SUCCESS in 20m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/08d783df2b544262b1ab9d0365b08185 : FAILURE in 41m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda5374db06a4e76a8117377b7e82395 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d3b0c93dc0047cd9fe1a279e85cbbcd : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b71877096de648e1b457655e2b2eb959 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2d0cf34f6834224abb95473ad70d6f8 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e931e815b6e443597f3b90e581a50cb : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1613caee2d724b04b7a44d0984eae0eb : SUCCESS in 26m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bef09a805f4b4118bdb51f19758ec6b0 : SUCCESS in 33m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/038c742738c44381a9196fcc77d794f7 : SUCCESS in 55m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f1c40b51abba4da1bb17063b81a1e375 : SUCCESS in 54m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac4d446fdd764a1aaa7816b17706aa45 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"bc20757f22fa59c7239aa74339d62d7b24cab7a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 09:16:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0dddc357cef7fcdf8df30ff62c4c69d2260dfa9c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 09:21:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c8c3f7492bc6ba0b23d092107e6a1cc3864f10e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 09:29:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03f7547230714f152f01898ea40af1a832780020","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 12:27:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5221a5db6284add8e7bd2e6ff039cb0d3136e0cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 12:28:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c4033db980f84077bb2868e8050291071f3257e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 13:42:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4096ea3647e14aa1a100a51dce180c22 : SUCCESS in 26m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9ae983f6b9844a7e82f3fb2898c6cf44 : SUCCESS in 49m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fc2311c4a7a946f78f1a075813d75cd7 : SUCCESS in 1h 01m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/911610e27a2a43cca9e562a4592c2d5d : SUCCESS in 59m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/381ced53e1dd46efac7e2ce39f5ea34d : SUCCESS in 1h 10m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0a7278780e114cf2a1a333b36996b047 : SUCCESS in 31m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/62a12be38bcd475f9ac383ae33c438e5 : SUCCESS in 18m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fe23b8d5fd5f429284a8afe81bb90224 : FAILURE in 41m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c1fed3febc4ae288b2b23b2ca991ce : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/307a6f7469384f6caaab71d58d26b60f : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a28a993c22d746208eb0b80cee58b944 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2222e7b11d3c47c7b49e7e171063cf54 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede9ae27567342f39c2891537b82a503 : SUCCESS in 7m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f338b380044a4043a9a56b5af9aa3cf0 : SUCCESS in 39m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/52b3a0b66fcf4dd4805e06b529447683 : SUCCESS in 58m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d635b0fbb22440368895121bc50556d8 : SUCCESS in 1h 01m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8eeaf57ba9184ef59ecf2216cbd179b2 : SUCCESS in 57m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/24454b8a88bf43bc9933b7365066f169 : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":2},{"id":"0d496df074e1fe5b16e6aacdc321260c9d9e1382","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 13:59:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ea60181bcb4135924b5a551e00b9a9fc94d0beb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 18:10:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/47d75c81747e4d4aa08a0d42e1c8542a : FAILURE in 1h 14m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/30df746ece48480493fe50b1e62a2792 : FAILURE in 43m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d753a24c9dc14fdf90ee17f571b3a1b0 : FAILURE in 50m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b13c9cebc96590d3bf29b67cd9514b1b4a7772b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 19:36:18.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"43e0bc027c97a2e3459f817f5c341cc3814a3d17","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 19:18:18.000000000","message":"Patch Set 2: Code-Review+2\n\ngate is fried, hence not approving","accounts_in_message":[],"_revision_number":2},{"id":"b2c52b0c264b3e5e9232860f21abdd51052aa4de","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-06 07:31:24.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"784a958785922b4b71598f32a7fa0e18e4e54f8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 07:31:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c6bfee91b023d0553e24a9961c5d3a0219c26962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 09:37:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/28debda0c34f459ba7ec21e68bbe20f9 : FAILURE in 45m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d762a793da4848e3a90735393219ab9c : TIMED_OUT in 2h 01m 50s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/edc84793d5ad4c57b8bd4535ede7896e : SUCCESS in 1h 31m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc31b028049141308d040ffde34ff1b7 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bc957c5ec29481abcf30ec25590a38f : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dff8b341c025466789a75ef77742bdd6 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af0345a2448476cbab9b32b59540ae6 : SUCCESS in 5m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6f2140a078c46d5bd7fb815a4fc5279 : SUCCESS in 1h 10m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/763df81f6a92476e99d4cc7e6c50d1e0 : SUCCESS in 1h 47m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fdf3609ab964286be6d92b9c5739a24 : SUCCESS in 1h 03m 57s","accounts_in_message":[],"_revision_number":2},{"id":"2b811e81fa74b61ebb15af125ae717bf15fe6f0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-06 12:12:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9a8b41ddbf5de9fa1d6660475ef4d36eb23ab6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:40:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b476c52e97854ce3aa04b4dd0c08349b : FAILURE in 24m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/88f5c64a78ce4d63b3720f27cf086035 : FAILURE in 42m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/451d8b720ce246739cc7c17cec6b0a69 : SUCCESS in 1h 13m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3c9ccc307b124edd9bab0ca88bd11fe8 : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b018ee5f75ea450fb9289566461f23a5 : SUCCESS in 1h 14m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2bd3ce6c4f754d4199d6386418944a8d : SUCCESS in 26m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a488ceb39af24c72a53b9f9dba928074 : SUCCESS in 23m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7cabcb57422e4873bf9ab4507c7f5161 : SUCCESS in 57m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b605a063345400d9de2629e704a402a : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db190e7417da4142a283c05f956fad1d : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/683ae37191b34395b436afb6ad8efff9 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d995b486258418f8969ab1d13b24a1f : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6bf70abd834518ba07c99bdc31ae65 : SUCCESS in 7m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b6f12b5bfc924c2596543417a3bf255f : SUCCESS in 21m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3520ec0d24b3407fb39a45c321ff4819 : SUCCESS in 41m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5840f19b31f4948a22f08b234bc8824 : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d9208d7abef94d1d9a0d9d4b6748dea4 : SUCCESS in 54m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fbd58af3f7be4f33a6df074f35c33146 : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":3},{"id":"fba9512c9404059ec7c7868af6721c46d5894c3e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-06 16:15:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8299850c2677d6a5cff9295fdd0fe34154071b82","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 16:21:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4cc342744e0331209cbd06e087474503a7f9747f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 17:35:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d4a23ab5c52b4cafb0dd84ba8f40b626 : SUCCESS in 28m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/83aaebc4c04f4ad7bd968f07fd6109e5 : SUCCESS in 1h 14m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1c48aa7ac7d242b68ebca7f8329d4078 : SUCCESS in 57m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/762b14d57e064e04a7fd79edf8112776 : SUCCESS in 58m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1b930b1baa904a9c95e28b1e0e551a91 : POST_FAILURE in 25m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0a589a1995d54bca91d4b541781920b7 : SUCCESS in 45m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/534de98bafd449d3aea13b4de9cc25eb : SUCCESS in 20m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ae32bacf250546508c2ceb553fcd185d : FAILURE in 20m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f8ce48f48c4482a8e05b8b3b2a2172a : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eaa480386c745d19380d00c2bbab7d7 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dcc5c5f1387441e992b548b51e01f2a : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c985dacb87684ae182c21c6ab5edf4ce : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b449311c0d9344b69f918d30e28b234d : SUCCESS in 6m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc8f73ecb8394342abe0e29f5efc8e83 : SUCCESS in 23m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/396ab289f8a346dc889a1be727308e6c : SUCCESS in 45m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8095a6d21804a75932f61c1539ba505 : SUCCESS in 50m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f753a329ba53435b8aad6542e5f48583 : SUCCESS in 56m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/06aa21906aa0443d8ff347affa0d50bf : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":3},{"id":"d4a87e35d28c7f3cef59fca5488dd84cba3cad1f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 18:04:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6407f34bf8a39c68cc13e11dd4035faa8e0609d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 20:38:35.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/71294883996344cca680f59a8c0dc738 : SUCCESS in 24m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/82360e44cb5b42dab2e13c61a4a3afbd : SUCCESS in 49m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a7fc723ec44049e4b755dacd8d70f7ec : SUCCESS in 1h 14m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/aa2e1488c59c488ba27f441db0f16fde : SUCCESS in 1h 13m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fef7160a6fd84b089523251126d1ade3 : TIMED_OUT in 2h 31m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d11035c388a54e159fa9d01d9a58bed4 : SUCCESS in 39m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0a985ccc5b624501b488a23f1e7b8afe : SUCCESS in 2h 10m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d38dde811f9744299a968b8d3c9402c0 : FAILURE in 38m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588fb8d0979048cea8737f26495f541a : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c9dc091030746fa9a43e566d55e58cc : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c5a3e3db5ee46d9b0e6381dde48a180 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d2e7b21c0b54956a3a179e24396e090 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c849ac310fa04787b823ebc0dac7b124 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/136286fbcc5841af9a6168270c42501f : SUCCESS in 22m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/07366bf6aff34b8b9a77795a5a056fc7 : SUCCESS in 35m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2b22f971e1540d1b5a1385f36cd40c8 : SUCCESS in 48m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9c95064a26da4fdfa9558e7f69e15ae8 : SUCCESS in 49m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/210f58d539844221b9a6fce8c134e838 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":3},{"id":"65a96464016154a30bca1e7d602db0bd92791913","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 08:30:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7c11e1bcef421147e637eb008f25966b1be31e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:51:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/67a7345770ba4e27b2654e78c9fef841 : SUCCESS in 1h 17m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5d7cc4ee88414410af6c670aae0ef753 : FAILURE in 53m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0efc8ea878114ff08e9b4a456b552dd5 : SUCCESS in 48m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3e9134b63679448384980e2ef79b3ca4 : SUCCESS in 28m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b6c6983bbe54e338df02ac252de8522 : FAILURE in 22m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9066574f16942cba3aeb084a794253e : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42b8955067fc475f81ad7579de12aeda : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45e9781cf7f146519cbaabca8215ecbc : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/841daba40cdd4a5698efaab69eb72f7e : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9257535df14a5bae42f29400dde647 : SUCCESS in 7m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/536b6e50b2164be79125c09d9d0e8609 : SUCCESS in 24m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/389b0c390cb643cfb797f8ab80d38960 : SUCCESS in 41m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/51ea0ee9eb76414c9773e839cb4551d6 : SUCCESS in 55m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/876adcd66cb5479b812ad75b330b5be7 : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1d6218602a234041b05ed15650477d15 : SUCCESS in 1h 04m 35s","accounts_in_message":[],"_revision_number":3},{"id":"e46bf01b3e7caca1aa80b45f09214ddcc4f53916","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-07 11:46:56.000000000","message":"Patch Set 3:\n\nrecheck (opendev cert problems?)","accounts_in_message":[],"_revision_number":3},{"id":"822c7b1a5ac5fbe21187b0021aceb104a3da8e8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:51:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6153ba6db85d4c608b24d73a691150fa : FAILURE in 21m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/56513167089b4eff8a6cfce5c58b1f66 : FAILURE in 38m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d10b15c67be74d29a5aaa688571d247c : SUCCESS in 37m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9cd54d7eaec543a686ece17be407466a : SUCCESS in 18m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/446b5c9ae62d462f942c3a8c9f1ceaf1 : FAILURE in 22m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/326f48b3bda54b549aaee4e0eaeff7d9 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04a99fd8cd364afc87ac6377ab1f8606 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/710fcabfa42445ae988bdd1a6e6ce188 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff3966fe21a24b9fbeece0445bf0f990 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01a41bfe3ac471dbedda576e301f321 : SUCCESS in 6m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/29a5b9dbf1154cedb5bd88168c9545da : SUCCESS in 19m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/087853fba76845e8b70be4b050228119 : SUCCESS in 35m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/93d6f28f96ca4f9dbfa5822a3eff38f2 : FAILURE in 18m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a499077151d404c9fda5e5e64478dba : FAILURE in 19m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b9704adbc0ab49259af7e6e0b2e6847c : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e7a3e3d598b89b1e748978d8accaf8936175f8fb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-08 07:48:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1339066ec834d4f4447689baa76403a763298de1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 08:50:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/86c1d71d4813439ab8fc4dee39d62aef : SUCCESS in 23m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f8be5c2de2114166a9d511bcc87a2f4a : SUCCESS in 45m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d76e1e50df384cb0bc68e3ffb1eed8cf : SUCCESS in 36m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0c0117fc54da415498fd16acfe040a7a : SUCCESS in 20m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4a89f402e57b405396062ffb8b8b07d3 : FAILURE in 19m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5085bdb10fc43a2b32d188179500909 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a02c4186bb443ed9d398267964549b8 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c40ef7fc4754498bbdb0c190e7579d94 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2192c8bb8c494bebafd59596965f0141 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f75b2fcc03848a18fa21a0f44d1ac5a : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0692ad4ad83e4eeab19feaeb7e93c6d7 : SUCCESS in 29m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e22dc08cc98401f8a33d285d9a0ce77 : SUCCESS in 39m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f5331c8f1634e198498dd0c8b1c3c31 : SUCCESS in 1h 00m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dec1e09ae2a146868c9755c4a80bb486 : SUCCESS in 55m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d68b983750c8429f96205a9740d598f0 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":3},{"id":"f3b5c1fb577e20563b7eedeecc6d1d8e9f8e5c7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 08:50:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fec46333266a8527823b6b3e503167949ba59c74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 09:53:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dbcab44584734619bc959ee89f0551e4 : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79836721ec7b473c86c069e98381fc38 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7314806c926478e8ba2d23206572155 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/509135ac4f854319b776f92435582f10 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4059f81bf5f04b2faf198111234d186d : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84c8c0370c8c450ba4cb2193ef4fcdc6 : SUCCESS in 46m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74330575df75452ba481eb4be390307f : SUCCESS in 37m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/97ac4e5f2c0f4116890e40ce48e6f7d4 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":3},{"id":"6876890535977eb64f41b9e135828388cf0166d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 09:53:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cd5ee9838590796dbcb3b1f58ede8de27e0334f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 09:55:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd61ce2e5314f7cba2a0815e556ee40 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3},{"id":"23042fa87d420ff91833153fe47862211b4783ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:54:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"910153f342ee7b3968c6f6ae8a2c427aa0f41f94","revisions":{"30f29cecef0a6cad07fcdf9cf37b4a9733cea3be":{"kind":"REWORK","_number":1,"created":"2021-03-25 06:34:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/782906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/782906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/782906/1"}}},"commit":{"parents":[{"commit":"5c0c2d2448f88a82bcc29b6771f90202011cbb1b","subject":"centos: enable PowerTools repo in images which needs it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c0c2d2448f88a82bcc29b6771f90202011cbb1b"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-25 06:31:58.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-25 06:34:45.000000000","tz":60},"subject":"kolla-toolbox: Upgrade openvswitch collection to newer version","message":"kolla-toolbox: Upgrade openvswitch collection to newer version\n\nAnsible 2.10 includes openvswitch.openvswitch collection in lower version than\n2.0.0 - which fixes the bug that state: absent requires value: to be set.\n\nRelated change: https://review.opendev.org/c/openstack/kolla-ansible/+/750631\n\nChange-Id: I92dab4394b7eebdcff7f90765cde078c1292312b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30f29cecef0a6cad07fcdf9cf37b4a9733cea3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30f29cecef0a6cad07fcdf9cf37b4a9733cea3be"}]},"branch":"refs/heads/master"},"c5fc80d8f33585e6f71ed30b7788b276db90a6e7":{"kind":"REWORK","_number":2,"created":"2021-03-25 12:28:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/782906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/782906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/782906/2"}}},"commit":{"parents":[{"commit":"5c0c2d2448f88a82bcc29b6771f90202011cbb1b","subject":"centos: enable PowerTools repo in images which needs it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c0c2d2448f88a82bcc29b6771f90202011cbb1b"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-25 06:31:58.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-25 12:27:56.000000000","tz":0},"subject":"kolla-toolbox: Upgrade openvswitch collection to newer version","message":"kolla-toolbox: Upgrade openvswitch collection to newer version\n\nAnsible 2.10 includes openvswitch.openvswitch collection in lower version than\n2.0.0 - which fixes the bug that state: absent requires value: to be set.\n\nRelated change: https://review.opendev.org/c/openstack/kolla-ansible/+/750631\n\nChange-Id: I92dab4394b7eebdcff7f90765cde078c1292312b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5fc80d8f33585e6f71ed30b7788b276db90a6e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5fc80d8f33585e6f71ed30b7788b276db90a6e7"}]},"branch":"refs/heads/master"},"910153f342ee7b3968c6f6ae8a2c427aa0f41f94":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-06 12:12:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/782906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/782906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/782906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/782906/3"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-25 06:31:58.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-06 12:12:05.000000000","tz":0},"subject":"kolla-toolbox: Upgrade openvswitch collection to newer version","message":"kolla-toolbox: Upgrade openvswitch collection to newer version\n\nAnsible 2.10 includes openvswitch.openvswitch collection in lower version than\n2.0.0 - which fixes the bug that state: absent requires value: to be set.\n\nRelated change: https://review.opendev.org/c/openstack/kolla-ansible/+/750631\n\nChange-Id: I92dab4394b7eebdcff7f90765cde078c1292312b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/910153f342ee7b3968c6f6ae8a2c427aa0f41f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/910153f342ee7b3968c6f6ae8a2c427aa0f41f94"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
