)]}'
{"id":"openstack%2Fkolla~782957","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421","subject":"Add octavia-driver-agent","status":"MERGED","created":"2021-03-25 11:15:17.000000000","updated":"2021-04-01 13:38:28.000000000","submitted":"2021-04-01 13:28:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782957","cherry_pick_of_change":771191,"cherry_pick_of_patch_set":7,"meta_rev_id":"3c216dbb1063934ec08e39353e50ca0c56aaaf40","_number":782957,"virtual_id_number":782957,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 13:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-01 09:25:29.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-01 11:45:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-01 11:45:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-25 12:19:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-25 12:40:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 09:25:29.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-01 11:45:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"9d3e33b1288decd07a444958a43d762a519e5e74","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 11:15:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/common/config.py","accounts_in_message":[],"_revision_number":1},{"id":"6a0696188e11b00702b2ff38746de9a3c0b60978","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 11:16:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"98c059c127d0ed89c167cd9686b61ff04a1b7562","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 11:22:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"750e810c6bc7bc51cd57904c1723b3db1b2064d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 12:19:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7781446f50fb47379011d3d234de683d : SUCCESS in 47m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/420db4a74dae4079bc00e2f9ed2c7f8d : FAILURE in 38m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/369a3e4d89a242d7bdffdb27eaf80775 : FAILURE in 27m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b550cce9f10a4e34a1c98921b3bf5951 : FAILURE in 42m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44ecc79c4fabd0074eb81c982b4da78c0ab41dc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 12:40:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/12ea284353464eb697f1c9966fd948e2 : SUCCESS in 29m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/79a41f2985824e55987f1e00e46d8c0c : FAILURE in 30m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e798e7cc4fb94f05bd49681a6f05dcfb : SUCCESS in 1h 17m 23s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0b013275e69d413bba3f47449091bc81 : SUCCESS in 1h 16m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0e0108581ec44b589e161d610e70e369 : SUCCESS in 56m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1a3a0437616d4d8e81ec782bab4caff0 : FAILURE in 19m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b4a105ef64904e89b275e6b53ca8803f : SUCCESS in 23m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/65abce322f624c22987d908aab6cf790 : SUCCESS in 56m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f95b07d45b44ee9c708cd476ddfd2d : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0baecfa30094c64961e7f369d2e225c : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f5d3b43e5d1473abfa67477b6691122 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fab734e557014ce2b73a317c9107cff4 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f97c8f2361ee4c8bb16aa4dce62a2be8 : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afbc1bf1de2e443383af4617dbd76c67 : SUCCESS in 39m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ca0d70548354a2fb52a2b8c7a9d8d05 : SUCCESS in 27m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bfe87383004d4375839c7416d5f37b6e : SUCCESS in 55m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e5f39b44816848bea798f7bc52320c18 : FAILURE in 23m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/09b7873d56bc4a9ab7acb9f20a5971af : FAILURE in 23m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f410a94c56886f6a7861a2f55cb569eda44bca27","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-25 12:56:40.000000000","message":"Patch Set 2:\n\nrecheck (opendev.org connection refused)","accounts_in_message":[],"_revision_number":2},{"id":"a823467225c385cc708226dae076ed436db90ae7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 13:58:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5f33f17ed604de3ac645345389ee3f7 : SUCCESS in 1h 00m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6792c48fe284382972b2b1d067884e4 : FAILURE in 1h 00m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/22228bf9045841ecb07086c8d3fb6984 : FAILURE in 35m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4be404c30c1444cba21f30463fb1b268 : FAILURE in 51m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"546b6e3d7c061e7294fea73af69526c976825a71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:17:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8779d4d8ad004a0b9cbd102036f0130a : SUCCESS in 27m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/386de363522149148429600f561924d4 : SUCCESS in 26m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/265b7f942fa34100a5f3dca0ef7522d3 : SUCCESS in 1h 00m 19s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/aa136759291e4d2898b52197c38cc517 : SUCCESS in 1h 07m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2bd850d91a8a48e48100ea6de7e20314 : SUCCESS in 1h 13m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6d5bb9808cc849dbb1856f241a6f8ff8 : SUCCESS in 28m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b0c5c9b22df74f0f97e81bcc2a98d240 : SUCCESS in 28m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ab32025ab1d141e58135e8ae0ac85dcf : SUCCESS in 1h 02m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f08d31c89954e00bacbb1ebc57e1f04 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9032b7111f334b6a8a0088455bd7e209 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a672285248b4a26bf459f269cc9c1f4 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a880c71fb7403e83d6b4c62f7bb0f3 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d9b2c720fc64b90920d3d3c73dae7a9 : SUCCESS in 4m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e9dd1fbb524d47dfa218155b30b78f60 : SUCCESS in 23m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/76bf685b55544275a98ec2ba319eb1b8 : SUCCESS in 23m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1e5f1a4d36247a7b495557b431a49bb : SUCCESS in 52m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f12153275f67461880ea65ed1beda4cb : SUCCESS in 50m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2650294961734ba9bea123decdd11ec1 : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":2},{"id":"0050c5379dfcce5e96d227cd8242af3f50a418b2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-01 09:25:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"807b2050521c4c492ba7eb562de866174be56a19","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-01 11:45:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"633fb398a30f9d668f731bfb56e6454dcd14a586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 11:46:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dc5718d07c066a43d68ad726bc787b6ee7992593","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:28:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/553ee58934374474a8487d203fc90e85 : SUCCESS in 26m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/32bea502f08a4eb299ecae58e2f7190a : SUCCESS in 1h 02m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/389bffa3e96642f0a230697e991dab63 : SUCCESS in 1h 39m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/773efba6ba4e4a58b34abc43d520fd5a : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e584f9bebbd4547897cf886dd55c2cd : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7c01b35283f43168ef70a26a58c14af : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9595c664ddc4f68957fe626ffd849d3 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa9a2c6200ca416986345d16c7964926 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aea94d7144034d94a3549d1df6982bdf : SUCCESS in 21m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aaffda848cf0459a90afc9a14f7b47a6 : SUCCESS in 1h 13m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b9259115060497c922511379759c619 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f3e6ae0f94632d1138cabaa308b4662848bd8fbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:28:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3c216dbb1063934ec08e39353e50ca0c56aaaf40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:38:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21162536ce14464a3675d1c6d799599 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5dc68ca7efc422d81aa5b27da928fb8 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"619d017dc4ffde1448fab659fd5691650b2324ab","revisions":{"50c0b7c1f4e7dbf930983817827eb1163d93c7e5":{"kind":"REWORK","_number":1,"created":"2021-03-25 11:15:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/782957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/782957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/782957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/782957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/782957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/782957/1"}}},"commit":{"parents":[{"commit":"f72a4b36e49cb4c6f86f2ad4069f77892124b9fd","subject":"Make kolla_set_configs open files in binary mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f72a4b36e49cb4c6f86f2ad4069f77892124b9fd"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:08:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-25 11:15:17.000000000","tz":0},"subject":"Add octavia-driver-agent","message":"Add octavia-driver-agent\n\nChange-Id: Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421\n(cherry picked from commit 785fe5641de3e46b17e52d16e0c8a9e02b8a5633)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50c0b7c1f4e7dbf930983817827eb1163d93c7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50c0b7c1f4e7dbf930983817827eb1163d93c7e5"}]},"branch":"refs/heads/stable/victoria"},"619d017dc4ffde1448fab659fd5691650b2324ab":{"kind":"REWORK","_number":2,"created":"2021-03-25 11:16:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/782957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/782957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/782957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/782957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/782957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/782957/2"}}},"commit":{"parents":[{"commit":"f72a4b36e49cb4c6f86f2ad4069f77892124b9fd","subject":"Make kolla_set_configs open files in binary mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f72a4b36e49cb4c6f86f2ad4069f77892124b9fd"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-18 12:08:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-25 11:16:08.000000000","tz":0},"subject":"Add octavia-driver-agent","message":"Add octavia-driver-agent\n\nChange-Id: Iea2c02d59d2906141ca37ce9b48a0f7f4fc14421\n(cherry picked from commit 785fe5641de3e46b17e52d16e0c8a9e02b8a5633)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/619d017dc4ffde1448fab659fd5691650b2324ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/619d017dc4ffde1448fab659fd5691650b2324ab"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
