)]}'
{"id":"openstack%2Fkolla~784891","triplet_id":"openstack%2Fkolla~master~Id113a53c083d105ac9200474ff20faacef6decaf","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Id113a53c083d105ac9200474ff20faacef6decaf","subject":"CI: add monasca-grafana to allowed-to-fail images","status":"MERGED","created":"2021-04-06 08:42:30.000000000","updated":"2021-05-20 15:53:53.000000000","submitted":"2021-04-06 13:13:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784891","meta_rev_id":"02924c7c395b98cb4e82471c95dbf884cc7f8065","_number":784891,"virtual_id_number":784891,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 13:13:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-06 08:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-06 08:45:36.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-06 10:11:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-04-06 10:11:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-06 08:43:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 08:45:36.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-06 08:58:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-06 10:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 10:11:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-04-06 10:11:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"6d26d5384af3f9f669afc96fd3289b915411e420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-06 08:42:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6d8110787656a1cb55b84e342efdad4457ccc0e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-06 08:43:32.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"237355b3040157f7cb3b1ee4deaa5cdf899d35f6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 08:45:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"13f290578ecd93215825d4fe17a639d5691b6c11","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-06 08:58:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e720a44bcf544e26cfe0d3ba4d7fe801403deae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 10:03:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/83bf567f15164287b640fae1668e64c5 : SUCCESS in 23m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/917569fa385042d7b79440816ab2417c : SUCCESS in 32m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/83787f0fb9744dd4924029a35c803ae9 : SUCCESS in 1h 15m 44s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ecba00a9524d4aa9b257add1ac84d3ac : SUCCESS in 1h 02m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2b788b3db54946c381dc12cfe775ca45 : SUCCESS in 1h 10m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1169840017514e8c8ec853f43707d2f9 : SUCCESS in 33m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/af8a395e6fae4fcd8fb2d61804610a98 : SUCCESS in 25m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1fb14229ab554a5eb52ce225ca16dada : SUCCESS in 53m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d172a1c334a4e199c4c9143f9a86f9c : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93faa93ccd254fce953b3458c89113d4 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af0478a611e5455891ab78d4c145848e : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/475c5b33651d409488e6a86abc1b5d41 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aeac42bf9ae4a53a37048ef0536badb : SUCCESS in 6m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e4ecb4f8849e42ae84fb6d7e79c3a44a : SUCCESS in 21m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4360212297724417b92184288834b5ac : SUCCESS in 40m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dab1f867fd584c58bc205cfad7cd97b3 : FAILURE in 9m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/22442c10a4a64137bccbd18dafebd43c : SUCCESS in 52m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/84f92a852690410aa005209a2849363f : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d5fd335e65c977b06af3ee279856c8c984a13c7b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 10:11:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cdbe2db54d6a37d68364c2a9fc911ed34dfa8e26","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 10:11:45.000000000","message":"Patch Set 1:\n\nrc -13\nffs lol","accounts_in_message":[],"_revision_number":1},{"id":"d906ea9370331bd5793ba04da28129d2189b6072","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 10:11:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1de467a17605720c4fc6a4783246d4300667f620","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 10:11:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa297b07ef2fbaaf02e99861c899e568204cdee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 11:58:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b65e3a446cdb4dfca2c4ee93dfbb05d3 : SUCCESS in 32m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ee75029b876c4690ac2a1bab6594f796 : SUCCESS in 46m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f271aa9543574edea2991757f793e955 : SUCCESS in 1h 45m 13s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f2a35b7b3480499687c982e2d18e4bd4 : SUCCESS in 1h 00m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/229cebd48a9a4f03b5a185a8bb093484 : SUCCESS in 59m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6ebf7d54db1d4fa092de3d52e222738f : SUCCESS in 27m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/28795c2c2e2d455292f2eaa820472894 : SUCCESS in 18m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/35a74d0b84fb41738cb942b56849583d : SUCCESS in 48m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dafab1ba66da43d3b6ff16dd0242a87c : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a506c6de42c45c1ae78678f53cf5ce5 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/860170ac81f34c87a7ba9bc7a3f1b059 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89cac9fdf9f24ebaa5498c0eebcd2c00 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37a08fc3ae224a499e39df4d88d82804 : SUCCESS in 5m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/69bd007a074d4ed3b61b427abafc56aa : SUCCESS in 36m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/56dc3acc6209479e84ee02242f97d61e : SUCCESS in 34m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e85693f4a7a8418a8c7123849aba1b7c : SUCCESS in 53m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e0c8cc4e0b984a87b019392eb2adcd8c : SUCCESS in 1h 07m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/12f75b0ed56444cf845a11f1a8dba1ae : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"587369acf8ce465705a0302747e55719613cd3e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 11:58:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7c0061822102704bb873b6f6e3bec10b138696f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:13:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7587f223e65c480ba3eac29ed60af827 : SUCCESS in 43m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f4d10e6b0edc4c759b0a7d40b98596e8 : SUCCESS in 1h 08m 41s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6c71e87e70ab4ff882433641fc00a5cd : SUCCESS in 1h 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c83cf393db4f84b72e3b2e8bed31ad : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/943f6161cc5d457cac2b2b8eb595a529 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2156e4c257734a01bf1e7c82cb19a162 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88bce7724a91492b9f81bef789a4179f : SUCCESS in 9m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/29134323d4a44d5eaaac1fac9a8cab91 : SUCCESS in 27m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/146589eea40c431aa2eb50059faa90eb : SUCCESS in 45m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/462b83561726483d92af4bb33807fa96 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0cd90e946825db2179f1ea97dcc4445c987ecd99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:13:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0981840705c81813ded166dd2f7998315233e864","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:15:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a837183c9f34c88bcbc25d58d94ea4f : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"60703ef863285916c888c59e6d4261c5ec8c7472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 04:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/64f3c8e9ba544c38a4fa8026a5f76acf : FAILURE in 1h 02m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43b01ea33cd04f0e98b75d413105bb27 : SUCCESS in 1h 45m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c82d3b1b024040b78f0c72a54ad928fc : SUCCESS in 2h 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02924c7c395b98cb4e82471c95dbf884cc7f8065","tag":"autogenerated:gerrit:revert","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-20 15:53:53.000000000","message":"Created a revert of this change as I483db6d6eddd84b376251b38509e74d4a5c2a10b","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"931d401887252d8830ebfb4d0b5d352f77094e9a","revisions":{"931d401887252d8830ebfb4d0b5d352f77094e9a":{"kind":"REWORK","_number":1,"created":"2021-04-06 08:42:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/91/784891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/784891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/784891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/784891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/784891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/784891/1"}}},"commit":{"parents":[{"commit":"ce1cb062689ff681b55d5b74a105d9da32e184f7","subject":"Merge \"base: Add lsof for healthcheck_socket\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce1cb062689ff681b55d5b74a105d9da32e184f7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 08:41:30.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 08:41:32.000000000","tz":60},"subject":"CI: add monasca-grafana to allowed-to-fail images","message":"CI: add monasca-grafana to allowed-to-fail images\n\nThe monasca-grafana image is currently failing to build.\n\nINFO:kolla.common.utils.monasca-grafana:go install -v ./pkg/cmd/grafana-server\nINFO:kolla.common.utils.monasca-grafana:[91mgo: cannot find main module, but found vendor/vendor.json in /grafana/src/github.com/grafana/grafana\nINFO:kolla.common.utils.monasca-grafana:        to create a module there, run:\nINFO:kolla.common.utils.monasca-grafana:        go mod init\n\nLet\u0027s unblock the gate and diagnose asynchronously.\n\nChange-Id: Id113a53c083d105ac9200474ff20faacef6decaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/931d401887252d8830ebfb4d0b5d352f77094e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/931d401887252d8830ebfb4d0b5d352f77094e9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
