)]}'
{"id":"openstack%2Fkolla~784917","triplet_id":"openstack%2Fkolla~master~I01ded1ca42b3136491ee084c6ea687f45d93f60e","project":"openstack/kolla","branch":"master","hashtags":["debuntu-cleanups"],"change_id":"I01ded1ca42b3136491ee084c6ea687f45d93f60e","subject":"mariadb: use Focal repo for Ubuntu/AArch64","status":"MERGED","created":"2021-04-06 10:41:51.000000000","updated":"2021-04-12 03:54:30.000000000","submitted":"2021-04-07 09:44:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"784917","meta_rev_id":"1b9537c528630ad4dd8078d820492b2a5aefa184","_number":784917,"virtual_id_number":784917,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 09:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-04-07 08:02:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-07 07:42:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-04-07 08:02:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 12:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 07:42:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-07 08:02:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"306cff15c4bb97f75ac40981311859ff270d2021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 10:41:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12a1926a8552ab24d5e5b116f9f9b29a5def3204","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 11:30:07.000000000","message":"Hashtag added: debuntu-cleanups","accounts_in_message":[],"_revision_number":1},{"id":"96f41ec73dd310d27590108efdb4831f517670c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 12:12:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ec2ece9e44104f3e941127efa563b3be : FAILURE in 34m 44s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/60ee9130938b4c06b149b663e0a963d9 : FAILURE in 49m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3f4540f2af664b178119f3aafe72fb61 : SUCCESS in 1h 17m 46s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/21521c37502e42c1ae200aa6d08a5f9b : SUCCESS in 1h 08m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/25526903864d4b8cb392a3be10f00fed : SUCCESS in 1h 18m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0f39b41fbf57465bb06a24419f8de3d2 : SUCCESS in 37m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/61874a73b5f846408bba004f456154a2 : SUCCESS in 21m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b45311d6fb7482d8e786c5557b6aff6 : SUCCESS in 57m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40446e78ff2f48cb902e7dea42ef1c74 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19cacd29d54744d5b7996d6850e96edb : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/640b72c391bd4193a80b1cc9a0c880d7 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38a874347d504e33afab49188c12b62c : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad85cddab19a4f88b5a207ec9e860684 : SUCCESS in 6m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/62e4ebad3f0844e1957058f343f13f69 : SUCCESS in 32m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dee982765a534daebf3bb9603972eb29 : SUCCESS in 38m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d02b084cac54d58ad173cd279c263e5 : SUCCESS in 51m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/880517bb33114bff874749f860cf03c8 : SUCCESS in 51m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8af16cd6c4f94918933befa3bd380ad9 : SUCCESS in 1h 05m 29s","accounts_in_message":[],"_revision_number":1},{"id":"fb45b840fbb5c5be1cea87c8f10be08d3db2c811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 15:46:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4657409c693277cf1f9e2009094226de2caba7cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 17:19:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c46b8c6440e645a89c2b586b2127830f : SUCCESS in 23m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/efeeb47d13104e9198e6f53407ded1a9 : SUCCESS in 48m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f7aefd53f5464f5da053e863480f1fc8 : SUCCESS in 1h 02m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4ef30243baab4ffdb5fefde3fab2890c : SUCCESS in 1h 02m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/767035c1a22e45dfa580d99e08a77ad1 : SUCCESS in 1h 09m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1f937063aa4d4feaaf9c7a49b66f7234 : SUCCESS in 39m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e8df339d6e13450783ec6bda8a9abf22 : SUCCESS in 45m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bb9040880a534182ad5a28034ac606fe : FAILURE in 41m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3980c8a13ca741daada7d929ca5b1020 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/488f49aa5c63450b86d5cb0818053d86 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6c1366359a2427a88dc119c2f69ea20 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d573ed68c3214dd29185aceabb47b5ba : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2da968cef454671b07efe5f7a202264 : SUCCESS in 15m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6cc0d925fbb3469b942d202e8fcf5321 : SUCCESS in 37m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/786dbec489f94317b4924852372ea1bc : SUCCESS in 59m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/77d533541f964478b573d32194b59c99 : SUCCESS in 53m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/38e5158ff34a4d4688cd89574454155f : SUCCESS in 53m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0459cd51d5164a39a03a2723d790d115 : SUCCESS in 1h 24m 42s","accounts_in_message":[],"_revision_number":2},{"id":"4ac54e93f16b356d79a1cadd3af9f303c4059e4b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 07:42:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f861611446e7403957e66687df6b0f4d78ea0919","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 08:02:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89543174f0324b2b0c5ebba26b49cbbfb33e7c77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 08:02:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"58eb2bf1ca17412d2b808ba8ec7eab385183fc69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:44:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7d9ff9f248fb430ba3622c169d8431c4 : SUCCESS in 47m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e9027fd8ec241e7898f21b2c0ec93b2 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df4a38bab0a240b9b0d159ee3fb9ed54 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9310f8db7814865a15f557a7c63e9bf : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098a1911aeeb448fb1a2c606426a8721 : SUCCESS in 6m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b39d088409148e79500d838e5a5f92f : SUCCESS in 50m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/160db72eadd24fb794e14638d9c55046 : SUCCESS in 1h 02m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/18dd1489f9a24467ae1d6f7916f8d777 : SUCCESS in 1h 10m 17s","accounts_in_message":[],"_revision_number":2},{"id":"733a336c9c9b6aaa52fd71cf3540316992e545ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:44:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4f49298f528518e34f87cfaaed35f8d3b9dad2d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:46:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5deb0fc2fb14afcb535b55683bb91e6 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2},{"id":"1b9537c528630ad4dd8078d820492b2a5aefa184","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:54:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/22bdcd516f544cf0964607bd937001e9 : FAILURE in 1h 50m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/28da7908635d42678734ea65be1ec5df : SUCCESS in 1h 44m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f242e8d66aded9a8e3df903f0f7c466f10f00434","revisions":{"7378d4b9227f9da5218f3e2023364f6225f46b23":{"kind":"REWORK","_number":1,"created":"2021-04-06 10:41:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/17/784917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/784917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/784917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/784917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/784917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/784917/1"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 10:41:36.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 10:41:36.000000000","tz":120},"subject":"mariadb: use Focal repo for Ubuntu/AArch64","message":"mariadb: use Focal repo for Ubuntu/AArch64\n\nChange-Id: I01ded1ca42b3136491ee084c6ea687f45d93f60e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7378d4b9227f9da5218f3e2023364f6225f46b23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7378d4b9227f9da5218f3e2023364f6225f46b23"}]},"branch":"refs/heads/master"},"f242e8d66aded9a8e3df903f0f7c466f10f00434":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-06 15:46:05.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/17/784917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/784917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/784917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/784917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/784917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/784917/2"}}},"commit":{"parents":[{"commit":"0db9b010afb27d21a540aa9b15c733c5919825fe","subject":"Merge \"CI: add monasca-grafana to allowed-to-fail images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0db9b010afb27d21a540aa9b15c733c5919825fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 10:41:36.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 15:46:05.000000000","tz":0},"subject":"mariadb: use Focal repo for Ubuntu/AArch64","message":"mariadb: use Focal repo for Ubuntu/AArch64\n\nChange-Id: I01ded1ca42b3136491ee084c6ea687f45d93f60e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f242e8d66aded9a8e3df903f0f7c466f10f00434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f242e8d66aded9a8e3df903f0f7c466f10f00434"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
