)]}'
{"id":"openstack%2Fkolla~784919","triplet_id":"openstack%2Fkolla~master~I50e5355bd7dd96a5eeaa004c5fc230d1cb8e8fd2","project":"openstack/kolla","branch":"master","hashtags":["debuntu-cleanups"],"change_id":"I50e5355bd7dd96a5eeaa004c5fc230d1cb8e8fd2","subject":"base: we already install sudo for Debian/Ubuntu","status":"MERGED","created":"2021-04-06 10:52:28.000000000","updated":"2021-04-12 03:54:37.000000000","submitted":"2021-04-07 23:33:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784919","meta_rev_id":"262010b49c50ea7a436b859d92f9b33b9fd5c7dc","_number":784919,"virtual_id_number":784919,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 23:33:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":2,"date":"2021-04-06 20:14:49.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2021-04-07 07:41:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2021-04-07 07:41:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 12:12:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 20:14:49.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-04-07 07:41:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"9986d56c158fae01985211635ff934fad31c7e6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 10:52:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab4b3212c78edfcb59a20c40d4cb702dd389edb3","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 11:29:45.000000000","message":"Hashtag added: debuntu-cleanups","accounts_in_message":[],"_revision_number":1},{"id":"7692596fd450968d26c114f95366dcc1f14d10a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 12:12:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9897dd24dd0d43f888d85c4b9161c9bd : FAILURE in 21m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d3dfd57ba82d45cda879cd055cbb1f3d : FAILURE in 36m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b36a31c7f0f14efea5f337ddb4d6ffce : SUCCESS in 59m 33s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e317335c1d2c40548446c79bd10147e9 : SUCCESS in 1h 12m 32s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/94fa6d69c0a04f05bb3cb2f4ff8c675c : SUCCESS in 1h 10m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/51402729ae6f4d478fbf62d53fe62cc2 : SUCCESS in 51m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c38f436161ec4f08b881e8208d483938 : SUCCESS in 19m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a6f65d2803f14491918a14d4227e2688 : SUCCESS in 50m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb366378cb245fa98979899a9cb7198 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff49fb933b59452caf059c6641d5ac82 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d53f4104af0d48a9b284c3f07a291c48 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/082f8b696354466bac9d37336f768c8f : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3680a4ea2dd24b95b754c4bf03060830 : SUCCESS in 7m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04a3285dbf8e4432b35738f27ad33546 : SUCCESS in 36m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d92899fb26b4ca3b6fd06bb0897ba75 : SUCCESS in 48m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a8fe2b3a46e4780a2555a1d0ca10d24 : SUCCESS in 58m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de29ac18f8904472a39f35a0bffc6a6a : SUCCESS in 48m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a249d1bb075249d0aa2cc453a108496e : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":1},{"id":"00c6673539ae0a14306aa0585afc83588ed3c6bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 15:33:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ca6f9270085c7c3568d6a37b92277e769eff776d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 17:05:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/16887d6e87a641ad8230ff66f7a3e249 : SUCCESS in 41m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4ef091f9b567465499a2545fd91ff0d1 : SUCCESS in 34m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8b31687d7dd94249b8e369a83aa1b951 : SUCCESS in 1h 01m 36s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9fda33eaf4ec41d88674a61268e5dfdf : SUCCESS in 1h 17m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2d2d840765424e41995971d3ff53f422 : SUCCESS in 1h 19m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f1a0c6dbe3ff4b5380bcb73aefb077f1 : SUCCESS in 51m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ab5b14731343452f8f4423e8225a4756 : SUCCESS in 21m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0bdafe85be44416a851cf3ac95732b33 : FAILURE in 22m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9689ca5aa998411e9953a68e0916d571 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a279618edd546908a6fadfee128f963 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fd2d85744ba498da2cdb59ff2fd0af7 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e58d969fb62945cfa1188e1af6b0e256 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f90b5bf6594950939327fb0e4e49bb : SUCCESS in 6m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e82f6b76beb34654a33e1bc02456995b : SUCCESS in 24m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b88b3b396744c55ab3bf761a1fc0a2c : SUCCESS in 44m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d7ed6feeab544e8ab25674ca44d53ad : SUCCESS in 53m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8bc1a47633174282b1ec5a3e38b901d7 : SUCCESS in 1h 05m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/14386f81d3d94bf289a6916ffe3d7499 : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0ac2bdaa89d37e71a50e57d06bec05683827203b","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-04-06 20:14:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dbbb5728e08d441bf8a976733c50728a58140b0c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 07:41:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0f742bd886619fbc4f01e1a1e580878822603051","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 07:42:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4449c73aba6f79b16c773b0e13c79fdb7a5ba7f2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 07:42:11.000000000","message":"Patch Set 2:\n\nmore sudo never hurts!","accounts_in_message":[],"_revision_number":2},{"id":"423765290bae68e3e102fa37bfbf12102da78778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 08:45:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2e7c8001ecd8403fa46684d468e21852 : FAILURE in 49m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e881325fee624d43aa20c48d8930fa80 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5806302695ce42e4bfdc9a559d3307a8 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b01f3020629a4294a19b8b7cbecae236 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcad4fe525864e1c91c9053c5ad71ab5 : SUCCESS in 6m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ebd6b2b8a92242b387dc28164951a51e : SUCCESS in 54m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4a243ad665147f5ad3eb40fa342fd73 : SUCCESS in 52m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ebee749c00574a4f9db015e3d244587a : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0b1a0975542eadf12e95a89f238b3b21a586bfb8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 08:54:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"16489b96bc7486e28f7622395636f2aa881474be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:56:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/887d287b8336481ab4fa6f4ad45474e6 : SUCCESS in 32m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9b96017039fb4463a12fa2d7b085c945 : SUCCESS in 50m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/41b99589bcbf41a0a84293ad4d604599 : SUCCESS in 47m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d3d5b98f0cc4b37ba1400199f7a2864 : SUCCESS in 27m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6fc12f6cf1ca4ea7b4d5013c6b5156d1 : FAILURE in 16m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b33f9d2627f404eaca2e538f73db999 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ed0bcd42bf044e287d9ca9431797e39 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c195a5c5d394efd9a35e61ad9ddf88c : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ff09d11e638430a979b30adc244e38f : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8749a5820c4ffabcdfad1f97261399 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/34f063cd64704110a2fe5ddb5980379b : SUCCESS in 24m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3486aeb1e86e44148ec29f4b3097a822 : SUCCESS in 51m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4cb4c8e357124621a33b941d43ade319 : FAILURE in 29m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/925ec3295a7e4139bf4a9782f9462bcf : FAILURE in 30m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/54da4b8e38a64bd7a215355533ae6553 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":2},{"id":"aefd17b05776ed53884cc599c76e1c96af99fef2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-07 10:10:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9ce1b415195dabf33417074cdbf0ba080e3e41b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 11:07:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/57f4f6315c0f4b5cb8d13fc7496192bb : SUCCESS in 22m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8419be05138b447ea813cfa31833d175 : SUCCESS in 41m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ca241f1949934b818c5534594022ce53 : SUCCESS in 37m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ea33a30f428140e4bcdf1f1c71dae21f : SUCCESS in 22m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/375ea3792ea94d10883e2cdb7ab7fe41 : FAILURE in 23m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9cc65c09765487fa2708da252fce264 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeaf4053e34c4788b51d6f57ecbfedd7 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac0c9a438c314aa493cedee0437115a6 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d39d5bf139f433aafb15cde4a619c03 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5619e67ce2664b2191941d28ffa934ba : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/74f6b12691ce4aa7a6f60c1ce0d7edd8 : SUCCESS in 22m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9876b490f1dc4506b575c41f7c5a2423 : FAILURE in 36m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6fa56ab5273040898b625210a9faad11 : SUCCESS in 42m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e5e9a19fc3245d1946274493d375fcf : SUCCESS in 53m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4dbf13fc87084657bfb01b05f9fd840a : SUCCESS in 51m 27s","accounts_in_message":[],"_revision_number":2},{"id":"0f6897fbd136c8ce89b40fe137be52bf01eca27a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-07 21:38:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"82776c6a3c51308a2451831cfa3567ad5b98944b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 22:41:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d66039e9620844aebf58762a54dde7ac : SUCCESS in 32m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0a32aeb312f34358b269d64d1868a74c : SUCCESS in 39m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ca5ab4ef05e949de834f0d8188ac29cf : SUCCESS in 35m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/43aa1832ef6b4b458b6136fb24ad4ab8 : SUCCESS in 21m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f9c3cfd8cea4ee6a4c92368c71e02b0 : FAILURE in 24m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea4429d539c4090a8e6acea33509a50 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/056edb2b8b844f20bcb9beb53c807eac : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b57f7e2fdb3e47c0ad940b716604ec51 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19d179ce78984841a85d299b626a3121 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d3f9ae1b824412be627ef88557e312 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18a80073761540e89dc2ef6fb1ea11fe : SUCCESS in 24m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b9fd94548e954f64abc3a09cbd75e11f : SUCCESS in 45m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6b4ca9f615a4878936a233365be42b1 : SUCCESS in 50m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4064422f34d14390beb9ed657b454a7c : SUCCESS in 49m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/81c4bf8ce5054bd790ca21ed95adca4d : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":2},{"id":"207e27f8e7482914f14dd5d8facd24b43aee4f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 22:42:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7cc92c41aab39aa575c3163164e50b58faced231","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 23:33:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/97b2fcb4f47b47e08a47789234889229 : SUCCESS in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0762f65c19ea4ad1a4f1b5e18a7c694a : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a961ab83e6b45429f3499b7e7f6520a : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a013bc4e750e491a860b8ef0e0d80f6f : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5ae8c612004273a2a74b10462b5be8 : SUCCESS in 8m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/741535fe416b4ed79f19f024b0defcf3 : SUCCESS in 38m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5bf0be6bcd5245669d2fd210b2e0bd25 : SUCCESS in 47m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/45b5481fe6df4548ab21e795f39a9c67 : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f6746e84538ccb38389ed6308cafc45279ca5f63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 23:33:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"215791bcf2c863c1f6f64f67e0306be7e803257a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 23:35:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/461ae56062564354880e4499459122fb : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2},{"id":"262010b49c50ea7a436b859d92f9b33b9fd5c7dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:54:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0efc3ce3e4f4376761c45c6d78cbd302543840b0","revisions":{"e82a09f5a01b0caf97b1b270c8b8f96ac34b01b4":{"kind":"REWORK","_number":1,"created":"2021-04-06 10:52:28.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/19/784919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/784919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/784919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/784919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/784919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/784919/1"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 10:51:58.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 10:52:18.000000000","tz":120},"subject":"base: we already install sudo for Debian/Ubuntu","message":"base: we already install sudo for Debian/Ubuntu\n\nChange-Id: I50e5355bd7dd96a5eeaa004c5fc230d1cb8e8fd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e82a09f5a01b0caf97b1b270c8b8f96ac34b01b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e82a09f5a01b0caf97b1b270c8b8f96ac34b01b4"}]},"branch":"refs/heads/master"},"0efc3ce3e4f4376761c45c6d78cbd302543840b0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-06 15:33:01.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/19/784919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/784919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/784919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/784919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/784919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/784919/2"}}},"commit":{"parents":[{"commit":"0db9b010afb27d21a540aa9b15c733c5919825fe","subject":"Merge \"CI: add monasca-grafana to allowed-to-fail images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0db9b010afb27d21a540aa9b15c733c5919825fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 10:51:58.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 15:33:01.000000000","tz":0},"subject":"base: we already install sudo for Debian/Ubuntu","message":"base: we already install sudo for Debian/Ubuntu\n\nChange-Id: I50e5355bd7dd96a5eeaa004c5fc230d1cb8e8fd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0efc3ce3e4f4376761c45c6d78cbd302543840b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0efc3ce3e4f4376761c45c6d78cbd302543840b0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
