)]}'
{"id":"openstack%2Fkolla~784923","triplet_id":"openstack%2Fkolla~master~Ie19bf46236907d79cb42a5e94cd1067e952b8cbb","project":"openstack/kolla","branch":"master","hashtags":["cleanup","debuntu-cleanups"],"change_id":"Ie19bf46236907d79cb42a5e94cd1067e952b8cbb","subject":"Debian/Ubuntu: handle APT keys in proper way","status":"MERGED","created":"2021-04-06 11:18:20.000000000","updated":"2021-08-25 17:22:17.000000000","submitted":"2021-08-25 17:18:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":46,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"784923","meta_rev_id":"e709e37d68c0ea84d2574413bcb15e83cca418cb","_number":784923,"virtual_id_number":784923,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-25 17:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-08-25 08:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-08-25 15:51:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-08-25 15:51:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-08-25 17:22:17.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-08-25 17:22:17.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 11:28:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-04-06 11:32:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-06 13:06:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 20:18:32.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-04-07 08:17:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-04-22 07:41:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-08-06 10:15:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-08-16 02:15:16.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"5e91873b4bb23df242af9b942eac7d1b1f777ff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 11:18:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f03a1f48eca6d053ad7e93a4066731dbd1556590","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 11:28:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db1b0233d2ece42f421cde7b60ea59e0dd6289c6","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 11:29:36.000000000","message":"Hashtag added: debuntu-cleanups","accounts_in_message":[],"_revision_number":1},{"id":"fbb49d58abd7d570956ee2e6efc5edf451362b51","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 11:32:25.000000000","message":"Patch Set 1: Code-Review-1\n\nplease append those for debian only when debian","accounts_in_message":[],"_revision_number":1},{"id":"3ed7c93dc7618ee42c2a72d1da4b3e4dd1bf2ab1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-06 12:37:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e761f0068a3fa41175fad889afc507353bb90a3b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 12:41:31.000000000","message":"Patch Set 1:\n\n(2 comments)\n\n\u003e please append those for debian only when debian","accounts_in_message":[],"_revision_number":1},{"id":"90bfb8b85ee7301052a2d94f704160b32b75386c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:06:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eb4851e7d8b744f6b3862abe5a569ad7 : FAILURE in 22m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a05651065117408dbdca0fe227ad4b24 : FAILURE in 42m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6fb1dbe114af41c68761ba31345fb32f : SUCCESS in 1h 01m 12s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ad77b1ee976540c0af711b01e80227a4 : SUCCESS in 1h 06m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f29f8b4e5a7b4de393223e3a38c6b677 : SUCCESS in 1h 08m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/544d80faaa064049bab8414c39da15cc : SUCCESS in 32m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/98df59662a1942a2bf5da9ecc7d93106 : SUCCESS in 22m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/39671144d75044e8b91e2db064d37ce0 : SUCCESS in 55m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13c529cba011412ea6e490808ab02d63 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddcb9440b32047c89e17e97e715bdf12 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6009a4ed8c5d4c7da73046899a3f27c8 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b61a59bf1444417a9126eb0a87d68d3d : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58420740c0034fad837df13c4da71734 : SUCCESS in 7m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b31b0f3d9d8a4c278b8329a3d333a0b0 : SUCCESS in 20m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3d36a747eaf41f08c1b037c3fa437b3 : SUCCESS in 36m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/245d0814fda44827b4aedee846ed5b5e : SUCCESS in 1h 04m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0765842ccaff40879a816cc5a7afaf2a : SUCCESS in 58m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/df1363ea16bf41e499b6d570e3dea8f8 : SUCCESS in 1h 44m 35s","accounts_in_message":[],"_revision_number":1},{"id":"15661958861be52d7e95702387d82239ea6cc9fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 14:26:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc1852a4f797def0e1da4d9d05981b55a032e0c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 14:33:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"22ef33e0e18268524d2b614bc06199e0767d5f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 16:29:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a7cf2728d47d471dac2813fe04fa2521 : SUCCESS in 20m 41s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/efe809b8da8f4879b2edd1ba5f6743a3 : SUCCESS in 46m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8c3e2a4f34e746fe89b6c84b7d83f7a1 : SUCCESS in 1h 05m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/74fc505e220e473d8cfd55bc92496a0d : SUCCESS in 1h 09m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e53991fef77c42a089fd2b7e9b5acca1 : SUCCESS in 1h 45m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5c2ad9ae14c542adaac549b639173cd8 : POST_FAILURE in 17m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/16bd1a1fb92040a1a521d4c5f211ec54 : POST_FAILURE in 20m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/977b8caa869947c982ba626c99049db6 : FAILURE in 23m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa3f3b33bfba4a36b9b2f6fb0d6302c2 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4f9e6bdcd7e4b4e9942a5c0267caf1d : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/954ece67785d4260b5ae8749f08e3d37 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac9eadae2790443885954f9093e721c9 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c08c48dec34cbdaf3698ed84a8b085 : SUCCESS in 6m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cec130b95983409f8c80c7f6f755f5cc : SUCCESS in 26m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d77736dfa4564a8f8fe3676b46b31979 : SUCCESS in 52m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c45cd3e7d2647e5b344c6b49df41c50 : SUCCESS in 1h 09m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c88d0d1427594181ab0304ecdf02b828 : SUCCESS in 1h 07m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/819495ad6baf47d4ba0208fbd0ae2bc5 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b575b92585260ccdbdca5d0c53beae72dd98ee15","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 07:50:20.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nwill need a rebase due to the conflicting bionic/focal fix for aarch64","accounts_in_message":[],"_revision_number":3},{"id":"51a01c4686a5ef93758f095d82298b8b64eb55e0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 07:51:31.000000000","message":"Patch Set 3:\n\nalso - maybe let\u0027s do this before bullseye?","accounts_in_message":[],"_revision_number":3},{"id":"096d7f7593e2af962506d6dc187375defcdf3970","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-07 07:55:31.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8b40cdcaea9055cc8db9452f5ea7cd17197617fe","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-07 07:56:05.000000000","message":"Patch Set 3:\n\n\u003e also - maybe let\u0027s do this before bullseye?\n\nCan be done, sure.","accounts_in_message":[],"_revision_number":3},{"id":"c56cc04bcb4bd1cb144babadf7d274505c07586c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 07:59:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"57f0f596ea799ff9a32738503da8cdd63ad31459","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 08:17:53.000000000","message":"Patch Set 3:\n\nLet\u0027s not rush into this. It changes the way that repo keys are defined, which may make existing operator overrides incompatible.","accounts_in_message":[],"_revision_number":3},{"id":"5f189bc8e59371d839cf34f386af679c3056d5d9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 08:33:38.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"664142cc82c502d12bccf368fe8e92d69c15d365","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-07 08:47:49.000000000","message":"Patch Set 3:\n\n(2 comments)\n\n\u003e Let\u0027s not rush into this. It changes the way that repo keys are defined, which may make existing operator overrides incompatible.","accounts_in_message":[],"_revision_number":3},{"id":"d1e1fc349cc6d86dc0abcc4ec266c6066802f9dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-08 08:07:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4ffb2d95eb00194c4afccc6faafc03cb09ef339e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:15:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/539186ebcda449978c8f6ce2d16a1acb : SUCCESS in 32m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5d148db4e07940b082c4470abfa82f4d : SUCCESS in 2h 59m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9231f2dfd45f49ac89299fae6f6426d0 : SUCCESS in 47m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3dc35249876047ff9808f3b3fee1f5e6 : SUCCESS in 22m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7df2f18bd9374c5ebc8717deb5ad3c02 : FAILURE in 30m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a31163ea2f146198fb5af9614d0b3bd : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36150bd04f9a4a2ca77bdddc3f827ab7 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25abca2706d2485cb5891594ef907dc7 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e0d867a6e09433da5e0f5b55e02186c : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c8a0d251cf948b29082239c4f1dff9a : SUCCESS in 7m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c49f63e1456f49628e40c9faf3a54945 : SUCCESS in 31m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/da17e0b07e464ef79a76c0cbab8eae0a : SUCCESS in 40m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c518ed4300ff470ea62df1a896d76043 : SUCCESS in 44m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d8c85fac873458d901e663bfcebef1f : SUCCESS in 42m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e0e44c289f144da0bd14e76564106b4c : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":4},{"id":"4b8a31f2e0905de33da1893e1632c19e4a496f91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:50:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0bac4d9a5d2ad65f8acf7989b16b267dbf0f486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-13 10:56:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c013028af990d322bd8eb993741a62d7d6876241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 12:17:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c60e18de280945c2beb1bb640151a315 : SUCCESS in 26m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ec5014e58d26472abffc3d86492b04c8 : SUCCESS in 45m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/341560f9e3ba4765b650ce9d1c54f1e3 : SUCCESS in 45m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/076fbd40cddf4faaaf481b72d61e0ee7 : SUCCESS in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c8ade5c9d0b46d8afe425085838a54b : FAILURE in 38m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726e4decfe6941ada88b8ffb48d0c2f0 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/681dae36c0bd4ea8ba6f4739539e5d2b : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d82c11897ffd44fba82a1d718de7cbfa : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ccd45befc6c4157a83bde08b6dcf6c1 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba2710c6b50438ba9060c4a582f68c4 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ec42c0070a047348c4550f4dd2b0ea0 : SUCCESS in 38m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d19b1bf3c6824395932289f1bf106cf4 : SUCCESS in 38m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/715fbe56d07b4602a3204c065bfb16b5 : SUCCESS in 48m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f28ba177d8fc4642aca64deec293d796 : SUCCESS in 1h 02m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1054df9e175f4d8bb81d26182158e6f3 : SUCCESS in 1h 16m 22s","accounts_in_message":[],"_revision_number":5},{"id":"d61362d616c9d3d93bcaeedcd7aa9565b34d4663","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 14:10:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c049a648b174bd58527a0bb8441d793 : FAILURE in 2h 33m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/18396c0715eb4ba0a47aa61aebddbf78 : FAILURE in 1h 50m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/834e65d1d72b477cbde93a38e5bad5c5 : FAILURE in 2h 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f61324220d073d6c782bd2ef1a8325e3af676302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-14 17:02:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6dd1142f2a208998ab0ce648c4919ef14a726d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 18:29:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0875cec7d1f94c1b9777eab58bca000d : SUCCESS in 34m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e96a693a4031452fae5ea81b9ffe9507 : SUCCESS in 39m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f31010fded9e4b1e9ecddaf45dd9f49d : SUCCESS in 40m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bf72696e6f8e4de6a0e6e3367448ee0f : SUCCESS in 18m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4fa89fc50ae748d8ab5849947c452a2a : SUCCESS in 57m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/432807f6362742cc928500cb76c91079 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a811a4f35c443719d8725f845e0a0e7 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ad332ea863646d3b50bd4eb2282ea7e : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d859a7c76ea4b10ba5fb5db02947086 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c0a1072084e436c967ec1008beaaef7 : SUCCESS in 7m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/16280eb080a0434bb2ed6bf39911d9fc : SUCCESS in 22m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/35e4432ef17b406c9b1d4a9466dff108 : SUCCESS in 43m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b989fa17da6047f3a99732a78db2db9b : SUCCESS in 53m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6b8f0e994cb94a06a7189929d91d8cf1 : SUCCESS in 56m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/01abaa8d1b52412c940a14c7c50185f7 : SUCCESS in 1h 22m 21s","accounts_in_message":[],"_revision_number":6},{"id":"4c48c8a776906e91eac053e57fd6e0a537124975","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 21:14:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e108f00b0ed64c5ebc6a42ecb38da3d1 : SUCCESS in 1h 03m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/be4fa39d66fa432d964efdffc93bc438 : FAILURE in 47m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/addc9d8b68fa40ada80bed063121019f : FAILURE in 1h 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"72ed5b5213767a9da2f98438ba35b41d5bd41dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-19 17:19:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"5a3e2d078f8bae9191cc4635c44f2db1925375ec","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-19 17:21:33.000000000","message":"Topic wallaby/no-apt-overwrite removed","accounts_in_message":[],"_revision_number":7},{"id":"e5f6ccaff26ead3e782e90732682eb04868abfd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 19:24:35.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/030585b86e0b4d6380920b4e76ac3410 : SUCCESS in 57m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b24779d8065c4ecb8866eaf7dc49a00f : SUCCESS in 41m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/036c7284d7f34e88a8c8b2ca15a92eed : SUCCESS in 42m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fd0028a3275d4859be196c94193a29b5 : SUCCESS in 1h 13m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4be693ef690144f488a9e24c416f04b9 : SUCCESS in 55m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186e4db571c943378241263bb7db7d8c : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3585e16b4cb4b9ba55deabe8dab136f : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71e7e150716c45d7b6ae364c3d3c35a8 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d4c17ba342c493086c0593a650e392c : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65573f64d5954731b36887c77dec3d25 : SUCCESS in 6m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c5fc0cf60292469bafade86743eb404a : SUCCESS in 1h 51m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/783966887a8842a690f9db2c01f137c7 : SUCCESS in 40m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/851df1b635ff4e0a85d1bd7139d7d9ed : SUCCESS in 46m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ab1a82341ee44dd906549a41e09362c : SUCCESS in 57m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4e8362a272b640909b43f1937832777a : FAILURE in 56m 27s","accounts_in_message":[],"_revision_number":7},{"id":"2144bad462a38d6f385f32afa5093d2cee26c88e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 19:56:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4a2215f8490417f98a307650f644cea : RETRY_LIMIT in 1m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0e3ea9e519d4ceb8c26f962a06941c1 : FAILURE in 1h 26m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9908520d325f4e669e81cb5abfa0ff3e : FAILURE in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2e1ba7e2a0b95ebca3c2068308637bc257fd816b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-20 12:55:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"38fa2c255ba4541301c5d740994363cd116e73bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 14:09:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3539e41f15f848dd8042b0f8759fdb2b : SUCCESS in 32m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/447fc60e74db4eb299873063e82d9591 : SUCCESS in 1h 11m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/97b58ce57a9f45118042379e61836ff6 : SUCCESS in 58m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b763077fadae4b189a13dcb34e71c09f : SUCCESS in 28m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/50c2d702b9d245a2ae20425e105d27cc : SUCCESS in 43m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24797dddc85b4e5c8a7f11d6ec879f74 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c6a6cad83c437ba95178332b92d862 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40339ccc9280464da95752b235ec3f04 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5e6cbbb729b4599a2a527bea3c5c2b2 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda1df77148c47398075b135551b9384 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/325ac38e58fa4014bc02e4d0a1c44d39 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c4e08bb5f99d41c2855b8db1844c2cd9 : SUCCESS in 30m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e03392f68cc45f78b455865a44fa9dc : SUCCESS in 41m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/04bf015918ee4bb5a272685afb02e80d : SUCCESS in 55m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/89ad77552c6a4c17873b8238f53e40ec : SUCCESS in 53m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5172a5f9e0c64de0a11638df3b9c5acf : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":8},{"id":"7c21db552ea1eabdfbd31472c04dee81ccffeb5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 15:07:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/03f5ec921ad6428f9456a105b2944859 : RETRY_LIMIT in 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e91aff1b9c49441c9c0210c6eed701b4 : FAILURE in 51m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/94eacbbddb324f629f15d703d66575bd : SUCCESS in 2h 06m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5b0c1427733b4b8ab3df64bc2cd4585575f0188a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-21 09:37:48.000000000","message":"Patch Set 8:\n\ncheck arm64","accounts_in_message":[],"_revision_number":8},{"id":"a3a464ce01f4abbf56b7f8ddb6a45c1d169e97d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 11:25:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/763e5f604149438b8fb14c5ebe1d8743 : RETRY_LIMIT in 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4776db823a1b46d298a5306e05bbcaed : FAILURE in 46m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/26e9b3213e1a4da691ba65c6f70b3e1c : SUCCESS in 1h 42m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"af69449092c8de77c076e7a33e3402b608a6df1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-21 11:27:30.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"70a0dd99bb5998bf2214524e2e7e86bfe7355f7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-21 11:28:11.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"66605c19f280ed953dd24fea19c381100aa1ae16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 12:47:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cf0515eb5195486598c07b182f97d0db : SUCCESS in 27m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2998090164db40ed97a09d4c5e576910 : SUCCESS in 40m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6593000819b94436878fb7c5ff88648a : SUCCESS in 1h 13m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d91b81d11de44bd4b42550a2d3825fac : SUCCESS in 1h 04m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5500b112475d4d60ad89323bd1a0f8ad : SUCCESS in 1h 17m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/10d09128eefa47c3947bae7ea6467e27 : SUCCESS in 29m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e695ea2356dd4fe7a8cb6fd9c10f2283 : SUCCESS in 24m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1e41e291468c42af905dd99273ce6295 : SUCCESS in 55m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b044487f28ef46ba8f58808f659eb534 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d193171a9ed24cb4b373373652425df1 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e20c035d10f24f0880c9a5516561e62a : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66c0763f203d4a9dabac92a63a390b4a : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6576325e43214b94857940c3a9374756 : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e4c81206fc44fe3bdc6f53d1279dc8c : SUCCESS in 4m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/33f41b8e59ce494282fb45051c9e7175 : SUCCESS in 38m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/06e7eed3f6d94734971dc44963a1dfcb : SUCCESS in 40m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c97dba84b3a6428d9faeb19a18d53f6c : SUCCESS in 59m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ff41a5e552cf4ac18a3d835943a14426 : SUCCESS in 54m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2cdf91c7905545f392b4239ce8e26836 : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":10},{"id":"e6403aba7ad446ba5b8f01f9c7f7c59b83fe687a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 13:13:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8fc1e566a46c4541b653a7c0a88be47b : RETRY_LIMIT in 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c117bd773d674080bcbd4fe2a9ec59ba : SUCCESS in 57m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f587f469a186426c99919b6503062920 : SUCCESS in 1h 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f4ba2438f18e9d491b9d31a08c742c8826eb8b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-21 22:04:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"916ddd8468dc1ff9873460d896583d1c7f3c1f55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 23:58:43.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ae294ccb4fd4cafb0a48d4ba52de531 : SUCCESS in 1h 41m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/799001ab72ef46f8b9805035b494ed28 : SUCCESS in 1h 50m 42s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d2fad61c15e10606ea17f2f0f962f40c9d8d3e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 00:26:12.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/651c23a3972c48b391ccc4a2e389de6b : SUCCESS in 2h 19m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b995154d157e43fc82e7e2725117fb4d : SUCCESS in 29m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/04fccbca30a34f4d9aea3e81ec45f9b8 : SUCCESS in 55m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/24c84aeeccc94a3986dfd3c4d2e2d1e1 : TIMED_OUT in 2h 01m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/331bc2f1810444b29479c18a72f7479b : SUCCESS in 1h 09m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/de8972927bf542569e83c4f50338eab4 : SUCCESS in 1h 12m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6cb7035da47c4b8b93eb9b54cf2d275d : SUCCESS in 25m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc9910b24fc4417c9ae7c10847258c77 : SUCCESS in 58m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef42c5853e943f4b6d16457200ba9fe : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4ee4ee784994994a83798c4dbd3723c : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc2b798245f64db2ba23d401d2d7b3d3 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6def03aa79d2482cbda6db9fdfe5ffed : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf217158cef4d71bde24a5597bc6acc : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/070055ec861645f1b83eb123bf1c019c : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c635fb8fabc548e096565b3c9d204b0c : FAILURE in 24m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1bcfe2716acd4358b2aef8ff17043718 : FAILURE in 39m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7fbc28c0f6ec435ebe8e467f40708d38 : FAILURE in 19m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3164fdc69d84772b8e4f87f6f086935 : FAILURE in 19m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5088ffd21bd44ba9a7917afbbe1db662 : FAILURE in 19m 30s","accounts_in_message":[],"_revision_number":11},{"id":"4839c0653cac6948c88b3d1e58c8f0cce9972154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 06:35:26.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c0cee6b6605b9deea49d0a04e9cc353a791ee9c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-22 07:41:35.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"e81fcce30cf0776d4a39998631261e3322af3c50","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 07:44:58.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"f0f0323b70ac898977947e42a5e4a29583977d7b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-22 07:50:59.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fa7da409b03880631e05d457c53d4eae1c5abf60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 08:05:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d287f704e41e497da23e53b4f3695e08 : SUCCESS in 31m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1ecf7f2553cf42fc88787c5f8757726a : SUCCESS in 45m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/73688352b88d40e5b95376099ff09f6e : SUCCESS in 1h 02m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ad03fff30ba24841808b56f823a0f3f0 : SUCCESS in 1h 00m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0a62258c8bfc4347ad666dccaefb54e8 : SUCCESS in 1h 28m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ca25efe162294e769ddfadfaa6caffdb : SUCCESS in 32m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/09d0a40f52ba4d1694710995567ce552 : SUCCESS in 24m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a2f4d7b7bbba4e1a9167cbf5421fc2a2 : SUCCESS in 55m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f7447ee43684a2084e03df9bd4d2f30 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e03ab46538e747e7acf97bdbbcdf8c3f : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0abb4e423cc42f1a91a560400811d79 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51a48b02a00e435684a3b527438d0f73 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c040506bd7464d4f87e1736806abc5bc : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5e4f0e7d47740f9bfde730f0beb0de1 : SUCCESS in 8m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8c566685bfad47ad8d58d2bd5021be61 : SUCCESS in 21m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cdebb50554ac4930b6fa37c3ee0b3441 : SUCCESS in 34m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a54f2a53eda9494c827bf27efaf6b2cd : SUCCESS in 1h 03m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0e7d463608dd406aaf5f557b898c8401 : SUCCESS in 53m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a9f1d5e75653472aa94c41ddc4263a84 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":12},{"id":"27c50b77676c997b993dfaa2110b877b5dc014a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:17:10.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b17645d82c9454f9f1537dec8045bff : SUCCESS in 1h 49m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/efa08dff19f1453d8029cba390a20a9c : SUCCESS in 2h 38m 30s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"59751b000fa3203a54ea3706405d69f9229f070b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 09:47:17.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"ad1f9262b57df33398b6104d7e5d5c4d77604c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 11:53:30.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1a57357f7a0a43ff8a9a92b3e136e237 : SUCCESS in 20m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ee29dd7839e441999223ada1e4ba4acb : SUCCESS in 37m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/25d4e075b00c4c85933976e92f88dacb : SUCCESS in 1h 09m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/91c6c6973ba44cdca260f99ad239c53f : SUCCESS in 52m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1b02a77ecfe242c6bdb23f66c50e06cd : SUCCESS in 1h 57m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ba3540baf5fd459ea06ade6dd5cb583d : SUCCESS in 31m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a9bfc5f9cf0341ac8096d0b312a60190 : SUCCESS in 21m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e1235edbcd6b41328746b6e52c37e8ea : SUCCESS in 53m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/753ffaa799f24795adbea741ddfcab5e : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19759bfe46b94f16a4dc839b16c01108 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/593ea64d826a40f5a01ccd7ca21720cd : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8a0a13fabe04f318f7f0716f384a511 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ca70a6db95c40d285bada47ba4b51d9 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6c156f09b2d4cbd9c89865613c8182d : SUCCESS in 4m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3907e766c70f4b15ab8bcd4ff25436a8 : SUCCESS in 25m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0136b017af347a1b52dd681806e0d31 : SUCCESS in 1h 02m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/619b2d4d50a54289a54cb2205773c6e5 : SUCCESS in 57m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6b68f7826d034f8f8f94f8d173b7cb62 : SUCCESS in 53m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7c81399fead24ec09a1c5a49975c3fe2 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":13},{"id":"c709ff2d07d64bce5b36a7867152c00735daf08e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 12:25:24.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8d7990ee7e148208bbf2a39616ccde3 : FAILURE in 52m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/50001c96535c4c8c85698fa79fb3b77b : SUCCESS in 2h 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b4e6371474f029093f3d956d7b36dfa128516d90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 17:16:25.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"7a277d593b0e0736cc4d0ec5f6c37699cceb1f7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 18:44:21.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/652af0bc3ecc43dba7159634571fb527 : SUCCESS in 20m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/70921e2c01b544a3af487c529014bf38 : SUCCESS in 40m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9eb358d0f890492fb5b1e93fb9e5f1f8 : SUCCESS in 1h 06m 16s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/04c0a95bf00d411bbd5566147e07cfc3 : SUCCESS in 1h 05m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4a3264acdba84c67a137f2f0cd0153db : SUCCESS in 1h 23m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/80d54b4949a149bcaa0a748c8dfd5477 : SUCCESS in 37m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/63f67af06c1c4e2f89e5b84424df782f : SUCCESS in 18m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c8ac3f7d0864fb8ae88f9c4c339ec75 : SUCCESS in 57m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9211ac2e53444eab4c99beda39e8dd3 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4707981e60f846c8a008e8bdebcc3199 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da40fb3809df46cf94e7ab5ac9f77cd3 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03fed9e1e2864596812b30a7af7925dc : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62c69f44e1b4cc5a162d1ccb13716aa : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f48c8f5589ad41f0bfc701b775c2355e : SUCCESS in 4m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f2524f09421c4e2a8cd39311ae435c7c : SUCCESS in 30m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b9878cd8bec4609b72b45cae46b3ead : SUCCESS in 46m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/43dcd83889d24d359f7191571d385b87 : SUCCESS in 47m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bbcc70ddb6614a53b1bfab3511c2a2f0 : SUCCESS in 52m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fd806326a5b245c5bbbc5f5a16188032 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":14},{"id":"8ca5969748a8dbcea292981decdccb5e534625a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 19:51:18.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e48915e6f6e14b9bbf37ead70454072f : SUCCESS in 2h 01m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1644eb853ee14152a42cffcb061a9240 : SUCCESS in 2h 26m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e29fdd8271ddcf3adaf43913c6dfde14676a6f10","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-23 18:32:27.000000000","message":"Patch Set 14: Backport-Candidate+1 Review-Priority+1","accounts_in_message":[],"_revision_number":14},{"id":"8f4253407312d9c14eab391a1ecddd6f25e0f4e7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-25 08:42:23.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0499cfcf76dfb9513c57ede3e28791b836403c9d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 09:27:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7b088b24fef4e67090c4b9af4c349453d0bc87b5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-26 09:32:52.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"b8b5b19818343f7e09b3971b04af11574dfcde9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 18:03:50.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"629eb312c18d1e588b8879bae506ca316e29f773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:36:56.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0c8c1a1a8b144e67bb8e8df7b0e30c96 : SUCCESS in 24m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b192eebfac20472aaa951f565e2c3154 : SUCCESS in 37m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0871bfb16a9a4667b527ed938cbce991 : SUCCESS in 1h 11m 07s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8c9e3f53bd7146798a6534ca2485987b : SUCCESS in 1h 00m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/028af53e9b92432ba9c13cf5848b9dac : SUCCESS in 1h 17m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3144732eb66a4361b0cbdb969dd9d0bb : SUCCESS in 34m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/222592349ade4087a3ffd102312be4a8 : SUCCESS in 18m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/754b1a7b13c74d43ba51f38280839742 : SUCCESS in 1h 02m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ca78a9f6574b6bb3a0d638d6f5d9cc : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59957a5afdd049d98544a0d7342637fd : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f56176ff7e54b91a653117ca4faa9d5 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a251c129970442fa9d32aa70306aad88 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf86f63319094c33aef82dbd91707c07 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/512e7d9638c64ce0894bad57b059b974 : SUCCESS in 4m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b28a3ce0c60041208b0cd6198c1cedf1 : SUCCESS in 28m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c6dc38bdcc8040098f2162dd6311360f : FAILURE in 42m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9e98bb4c7944b188fab1c6e361ca2b5 : FAILURE in 19m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1fb64ec1aea54975b243357ae33394f3 : SUCCESS in 56m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/017c84aae68d40daa5bfc5b5250a1018 : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":15},{"id":"f925876776eb41f4e63e75ca3150dba888646cf5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:37:05.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b41821c296142f988ad95aa85c9ddac : SUCCESS in 58m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3276dc49e6f145b7b080e391aac19352 : SUCCESS in 1h 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2eca5689c7ef363a368f59a70e3b4a08aa09b9db","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-28 15:56:34.000000000","message":"Patch Set 15:\n\nNeed to rebase on https://review.opendev.org/c/openstack/kolla/+/788519","accounts_in_message":[],"_revision_number":15},{"id":"c349f793224474f4126d85341c3f9f83cc01b274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-03 18:14:00.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"81b52e3c855b7ec05e786636623f30f8b7a5eb1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 19:51:06.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ba144ce41ff04020840451b008db5909 : FAILURE in 41m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a55b4d88a47048b2a81b38abdf26af34 : FAILURE in 56m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bb877846942b44ffbb9e7e8055e0f384 : SUCCESS in 1h 24m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bfd2e8e36d24423394b01fd3016bbb3b : SUCCESS in 1h 17m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c6a08382bafb4eae9604001d979ee255 : SUCCESS in 1h 32m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9f958ae7cb1e49579220e99a8a836d2b : FAILURE in 53m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4c9be66d74f24e3ba91e119e79a34e9e : FAILURE in 42m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9613a0e334ba47bd8d7c8607d10ffd4b : SUCCESS in 1h 18m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8a3c2c6a164a0f82ec154d892ed536 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bf393568ca745c6ac26383d0de15a52 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f9ad97174d34361aae36217af0dd9ce : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9be2237d1ddc4706bd3710f55e02c8fb : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06bf41daaa2452084d076b862809dd2 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c776c33fbb504455876f6ce7573eabc9 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/20cf99009b9744edb93ad57acc77b0bc : FAILURE in 40m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/99234c0f8cb54c11ac2bd6ff4ba21241 : FAILURE in 48m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5745adabd363462196b30dfa66157e4f : SUCCESS in 1h 04m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e5a19b3c8a8c42fc94dc3a225de454ea : SUCCESS in 51m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/afb15433fc564a6a9dc39c3afddbca69 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":16},{"id":"d5caa8286aecc96a7c2392f274600b72d9544b51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 20:26:25.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed1e71e5f415413581e7120e71747248 : SUCCESS in 51m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/23f75d526e634481884b20abf2866be1 : SUCCESS in 2h 09m 59s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e75f7fca7eb92e0dd9c71eee0fdad09562ccb50b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-05 14:43:43.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"a971415a964f0dece66c05463964dd120e98ece4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:59:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/071e80505c7f4711a083658ed91e0600 : SUCCESS in 24m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4a26a64a8c7f4c85abacae01b66f974b : SUCCESS in 38m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/47951d9260664a8e92bc11bc7b7ece2b : SUCCESS in 51m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/220ed45a7c9c43dc983e0426dbca5913 : SUCCESS in 1h 05m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/12247793c3ea4a278da2b5bebf18939f : SUCCESS in 1h 12m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d1d77c3aeaaf47de924d6453d027a9d1 : SUCCESS in 37m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1f5679b738bf42a0a28a450581a3b606 : SUCCESS in 17m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/297bf432ef264d5b97b6c9ca51d51d80 : SUCCESS in 47m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f8cea8395a47a48b98dcb849940588 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0992f8464a164359bdb7727cac941083 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aa060c3160d4928be4ba486accfaeac : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cc1062e0a7949f9ac8980686d9307f8 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcdde7c0656d4bac8099b4870964545a : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/728385133e93460eab8586e44e6e2eea : SUCCESS in 4m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1b1656ca119946cea3abd6ea8137b5d4 : SUCCESS in 21m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/766055ca7d6143b0a6e69f79d582c6de : SUCCESS in 27m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d17ee6d14de64ff5a7fd6b5e72b1efa3 : SUCCESS in 53m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7ac07b92ea8846549c733d1cf5d71e77 : SUCCESS in 59m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/90b5cff73fcc4fb28ee024bff89646ed : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":17},{"id":"018bfb22f47855907a6d7e0149e87a2c01729c16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 17:45:54.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d3a54b1af0f4f938f08b6d9de4e0317 : SUCCESS in 2h 39m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/175dff22ef394d8b9d857510ce06d2f8 : FAILURE in 50m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/14e3e4939da24b7aa30e83be76694325 : SUCCESS in 2h 50m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/048dcdb1cace4070bb21c3a42f0062bd : SUCCESS in 2h 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"82d19d00e4d322a2dc01aaa9ea01cadc59cc3696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-06 12:05:46.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"2ac0c0210544515662e94791a4f7949547b477d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:31:28.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b6516141db03488e9563c2f824f5d66b : SUCCESS in 22m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ff6d6876c6ff465986e06c1d5910cadf : SUCCESS in 30m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9a928d2e686b4d7495c0f48e198438ab : SUCCESS in 1h 10m 25s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4e1a7a07fb334ccba3224b1f8dc752b4 : SUCCESS in 58m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d41bf829bb4b42c5957d3e509bd65d7f : SUCCESS in 1h 08m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3b8d475293c047a4ac9d1e2e1d022bed : SUCCESS in 1h 13m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6984c980ef404202b63423855993375a : SUCCESS in 21m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b9e38fac6f9144968f0b1f75f649745f : SUCCESS in 52m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e02667bc3972432eab41a8b1ee8cc956 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64f3a840fd6c44cf91e3375f5cd5a7ec : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/213dfa737cf74c97b18366299c364bfe : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ba345b1b71c4f70baf2ba5d982f4458 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e924e3e1f78427ba452efb19f5b9e52 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49a47594f8e14f879514263dd0ab4a27 : SUCCESS in 4m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c6de0e0edb414fda9601a1ecbb0248f6 : SUCCESS in 23m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f5d0ef893e8457ca3cf2a0ac84460b3 : SUCCESS in 38m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8cd2bd45d74949f8ae409c861075019b : SUCCESS in 1h 08m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8bfe8e055fca488899d32cc19e93bb08 : SUCCESS in 53m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bbf8403c3d5641c8b07a9824b9835c34 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":18},{"id":"cc42710f3aaf6e5bae732fababeedd22a3152af4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 14:13:02.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6fc2fa01dfc4718844f653187834d2f : SUCCESS in 2h 04m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0f95e7bf1ad48aead0d769f5d6c6168 : SUCCESS in 1h 07m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f8a8fc844d8b483988893e6dbd6b7362 : SUCCESS in 2h 05m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/40518512eabd4c879b2622a6566553cb : SUCCESS in 1h 06m 32s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3181d2dcff8761f84041e113e66a5a94a95fae5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 07:00:15.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"23ec9b2947d29894b5e3af66f4bd6fdaeb1ff0a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 08:25:00.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/685bee913d4f440c85b9676dcbaefd89 : SUCCESS in 27m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/18c0dafd38f345ad98dd081d007a2ac7 : SUCCESS in 31m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d26cd41fc9aa41a4bcacf9023c1fedc0 : SUCCESS in 53m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e1c8b9e2730e4c208fa41545723659ae : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/378fc3f2741c4db8874b5bb189c29cf4 : SUCCESS in 1h 21m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e47607cd9963429ba150da720625181e : SUCCESS in 35m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0e8ae55b1f514e94aa7840c7b1e99eca : FAILURE in 32m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/95762a96b440489e81678a0237ad1236 : SUCCESS in 1h 06m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0faec7bca75c471689da6ad7b0e6adae : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f1dca06e0754090ba9d556edacf3a10 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ec94824b50340aa8aa858525d400b90 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f7f83bb05f34a5d91229677c0f1f2f3 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3a9b54e80d4d34ae8327c48f159379 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55daf0d4129e4a6b83d2d15597690a12 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c585b0a28938484caccb7ed2685f86b0 : SUCCESS in 25m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e901f42938f4bd28546d3ba981753c3 : SUCCESS in 24m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17658a623cbb49ddb2ea9259996326b8 : SUCCESS in 43m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5a8cb061a5854617ae7c9b2b8cf9201a : SUCCESS in 58m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/54cd17fa35944a5c82547b24b6deb2f3 : SUCCESS in 53m 32s","accounts_in_message":[],"_revision_number":19},{"id":"d5879b192ecad34b16ff762617b1d4b92a4f802d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:24:54.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/295b4d3d83d84bb3a1bd2268515e7c27 : SUCCESS in 1h 59m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a97a49bb688d4c418ba9c5ca434825d2 : SUCCESS in 1h 46m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/715462f7834f4c119acbf5fda12c20a3 : SUCCESS in 2h 19m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/04655e36f9774f92ac9eeb0eef37d36c : SUCCESS in 1h 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"cf854a31e4751afee5e0148d685dd2446dbd6c98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 12:39:55.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"e29928b4bdb7979e6d95e0b86e664e2e9c8699a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-17 13:35:08.000000000","message":"Patch Set 20:\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"cfb44ca040692dadbe135d675d40e24931f89d21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 13:59:04.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2f36b2c7a47d47a18107294e57d8c839 : SUCCESS in 22m 36s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3c10c6763de640fe9071346967b4ffe5 : SUCCESS in 34m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7dd87944a46547c6910bd8967d7a4df0 : SUCCESS in 59m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a4e4091f494b4532b00c174f42a0a3ee : SUCCESS in 59m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7f5a839713d54fd5a04c526d5e31f792 : SUCCESS in 1h 01m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c99954e8420743fd95172ad242d6e70a : SUCCESS in 41m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5141ce47faed494b92c419318755adaa : FAILURE in 17m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f25325ace08a45a5b9d2f35d325ba0a1 : SUCCESS in 51m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94294146364481380f978de5adea4d1 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17f11c330eac4f0ca6b09892e220d78c : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dd6b3ee525c485abe912550552cd2ad : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9d9b04994ac41bcb2fba8d87fb4c85f : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730ffc17f253415eba0f8898a0529b99 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e86f651f9514bd88f23d6749e97e737 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24a1a0cbe0f8419ca1432ae06d6bbfc3 : SUCCESS in 22m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/55305425939e4edab7a1b7f483773497 : SUCCESS in 45m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f2145b0fcc64090bc63411fa78a9060 : SUCCESS in 59m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fb12ec6bea85415689189a8f304d13ad : SUCCESS in 48m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ae7f4f624b9443269ac93dc79bac2992 : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":20},{"id":"96299eb62de203b0b659a232593b3803a15d9854","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:09:19.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/595c854d29cf44c9a605a670ba2ebaa2 : TIMED_OUT in 3h 14m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab10b417adac43528e2c1c74ebd84bcb : TIMED_OUT in 3h 13m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dfb13f79e7c0411a86115ba02295c630 : TIMED_OUT in 3h 01m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ce80283c96e48c0b64f7902c689db5f : TIMED_OUT in 3h 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"db83daa84b4b3bcc8fc2ca38a57e8435fff4b215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 16:31:15.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"f4e0f0e500973c5124ee99964b62cc6f7b1b36b2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 16:31:30.000000000","message":"Patch Set 21:\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"4f7abc88784c04d69725ae7253f7c1baf88d5dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 17:33:04.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/61a8c715cff64929bf6abfbc3c672025 : SUCCESS in 19m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b220f9aca4694b0193d07497ebec20a4 : SUCCESS in 35m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ec18ef7d94904844bf7e840e8f4a76a5 : SUCCESS in 51m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0e40e82f24664926ad04b1d8deadd508 : SUCCESS in 51m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/513d5651ddd4407f91579bb25f70e797 : SUCCESS in 54m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/290cc528ec8b4f3aa63d7c5a3f302b76 : SUCCESS in 28m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/65cf54799cd446518d562b52ec12a79f : FAILURE in 21m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8939750e87e646d0b753dcb34c2fc55f : SUCCESS in 59m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d7fa7da1f9944f0a460ac3ed3c17fe1 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9dc1de2d57444e7ab1bba268f5d60b1 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80f433a1690c4ea7b92b67af02c7da9f : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3df510cebf7461886e8886b2c70c9f6 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4c6fd8b181477789f8bb959d8488f7 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd4b7f1506e44f779daa58fd5aa47421 : SUCCESS in 5m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f23ccc4068142ecb563093486a527e5 : SUCCESS in 23m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f80c61f9b024773bd790d3fb88b9752 : SUCCESS in 36m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c0edd45d718460bb6576fa7e42a92d5 : SUCCESS in 54m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37faccd80bff48bb809728dc12242843 : SUCCESS in 50m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1c283eec2c8048a09c3908fcbe9db20b : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":21},{"id":"9efbf6d8b50440e62abc149b22d5a773a84a50fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 18:29:21.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0390a1e581e14baa9619eb2571efe591 : SUCCESS in 1h 22m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0160a12e8b649db9bc4f9fe5f6015ef : SUCCESS in 1h 15m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/996a98031b68497abef97da74e6f4e89 : SUCCESS in 1h 50m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f80defa6d48d4dd9a3ed687ee9ccc639 : SUCCESS in 1h 07m 33s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"9c7ac22a6ea86f48e92b9e34ab5519c7adaddec5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-05-24 10:23:40.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"40c32255389ad1ecca1b48f23e2da1b570b322db","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-06-28 09:05:29.000000000","message":"Topic prepare-for-kolla-wallaby-release removed","accounts_in_message":[],"_revision_number":21},{"id":"53d10219ab93a40068b26110ed52d60dd1a83076","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-06-28 09:05:37.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"6368d2392c1fd2a0ac94ab593a02a94b94c4117d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 10:18:39.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/00b8b559a63d483cb78930ee2ae416d6 : SUCCESS in 1h 09m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5b7a07d9ac9f4829953695eeb6212b14 : SUCCESS in 57m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/77da4263e20748448ee8231c9a6d0cdd : SUCCESS in 1h 08m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/277fa650b5ee4dc1af84c403c3c651fc : FAILURE in 16m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80dd5d21bec440a987d6daa966661c08 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fbdede8cb024fe385471b69a0d43d19 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/489f5d341eb147f594419a1e97123b5a : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/778811341cba42ce98026cb92d7ab9ba : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f12e4800e545a7b67048e889026826 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9541544627cc4c5985280a289dcc80ce : SUCCESS in 7m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22c40fdd119b4d78b41485d7f17a699a : SUCCESS in 55m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8907ebe5e4f4839a0901a6e929bc76c : SUCCESS in 55m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0136782898b4418ba77c02018468c899 : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":22},{"id":"8a78e6db82d81b099b22be93b0713f3100b4923a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 14:35:15.000000000","message":"Patch Set 22:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":22},{"id":"9d5f4b7221bc520a720fa5850d17b2534aeb3e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-06 11:54:00.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"c2aed075ee13b6a1e85be1320f9f3ad1208ab394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 13:21:39.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4958e547d24a4a95893189e13bfb260d : SUCCESS in 23m 41s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9e7e4e2f65de435680c1b785c4940b25 : SUCCESS in 34m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3b6ddc7e989640459798e307ddcb85fa : SUCCESS in 1h 11m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e61dad3b1c1b48969f2b213685814b64 : SUCCESS in 1h 03m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a13965acf0214636a18b748639ce6dd9 : SUCCESS in 1h 22m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bb76a5f4a107458d9b30710ca02f9bd2 : SUCCESS in 29m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/db02d07a648b491b85e5692468d64fdb : SUCCESS in 19m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5e4940e3b38f41ab8cad4064cf00ebae : SUCCESS in 1h 02m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ebad0b3cb54b89b5e6986ee0fa56c5 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57c45bede07d47939df8482b5514e493 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/527d510632114fbaad7cb9b170e7c57b : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f17eef0ad3a46c0b464277e86d50e4b : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc7ba2134064a739ab685902a97159c : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4cc132855ef40b2894d20e272435bfe : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d8682fac83a451aaa7f1423e037efe8 : SUCCESS in 20m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05a8ef999dbd440ba3ce7c195a10ed7c : SUCCESS in 38m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6fa25dcb7ff24680970669abbd8b6eb1 : SUCCESS in 1h 02m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/924caaec4cd64ae5a6fd08dbc4d93996 : SUCCESS in 52m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/85d054fb87b84a158acaf6a0201d304c : SUCCESS in 56m 48s","accounts_in_message":[],"_revision_number":23},{"id":"a93a573c3bfe93c81d412634e71809ac40a1e9af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 16:43:01.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f883fe8e586649dd9758519981eda148 : SUCCESS in 2h 35m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/828369321cb74e22ae4463f43bc6091f : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8fe77f8da0934512bda655df7299d23a : SUCCESS in 6m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3fa67273f6bd47c5803d19a5a656861d : SUCCESS in 1h 47m 31s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"1ec2784ffdd09fc945eb33a831d31bb80c4c94fd","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 15:43:15.000000000","message":"Hashtag added: cleanup","accounts_in_message":[],"_revision_number":23},{"id":"bf6ed46b328a7f713e9af2308e41e0f316969995","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 16:04:04.000000000","message":"Patch Set 24: Patch Set 23 was rebased","accounts_in_message":[],"_revision_number":24},{"id":"96559ad90d9827a77768141c617d36ab1039f825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 17:21:45.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d9ef80eb416e476994717cf4452c5b7a : SUCCESS in 29m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d36f6883545f47cf9af3459a5f32d25a : SUCCESS in 43m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/20016470b6a948ce8480dd45a8923381 : SUCCESS in 1h 04m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9dc421a2095e4abd9702538ff8edde16 : SUCCESS in 1h 05m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f86f6b17c8e648469a29e3fa993d420c : SUCCESS in 1h 16m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ba43bc8eace444f1b2e59934ab5e2bcd : SUCCESS in 36m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/db2abb8663a34848b67c8e3dc84dbf6b : SUCCESS in 14m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/54b3e83ef46a44118abaaaf6d812305d : SUCCESS in 58m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a07c78ebf9483a9ba017640f4e2dc8 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4f2952c4e3349f9b5ffd4d825cdb197 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/012c336395d6414b87daa68301b8398c : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3357b8ae36a84fca8679a94cc072c4f7 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b9853a6eb341beaa12bb3d15379501 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a22c28ba3c74e3a8ece4fe4b05724ab : SUCCESS in 4m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/673a6094251a4b0e89562bdc0d8f63e8 : SUCCESS in 23m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/400ae78f55364e54aac21383a5db5d31 : SUCCESS in 27m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d93f5db24c741898ab4cc28d3b8acc6 : SUCCESS in 44m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0fdc72d1269646c194dbcf6dab14e9d1 : SUCCESS in 54m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/84e67bf832234446b676ac688a88d496 : SUCCESS in 53m 30s","accounts_in_message":[],"_revision_number":24},{"id":"968166f623de86316f053eccd2b70222fd8fe537","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 19:09:12.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/30b99cd309c24f28984d81a9f1c6df1a : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03a09f7d165943edb16ba77d00e823e2 : SUCCESS in 47m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eff6f4fb0f8647aa91ab6bbb50f06b8b : TIMED_OUT in 3h 03m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32966a57bbe44acf9645face1adc2077 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fcc5c59b3b841e395b9898c70346c2a : SUCCESS in 7m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c90096a458a343e3b1e1c62102484b45 : SUCCESS in 50m 08s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"dbb505675573a5481114ff47d1585f6cb5ec24c1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-29 11:46:21.000000000","message":"Patch Set 24: Code-Review+1","accounts_in_message":[],"_revision_number":24},{"id":"7fbd762b1e45ba215b77d688ebdc768341daa576","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-04 13:07:12.000000000","message":"Patch Set 24: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":24},{"id":"5888e2b0a52112b9d41d709595916b326212270c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-04 13:27:00.000000000","message":"Patch Set 24:\n\n(5 comments)","accounts_in_message":[],"_revision_number":24},{"id":"ecdd6ba17f89244171174558fe726df690d5ba08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-04 13:29:52.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"76958e22c30ecfafda7aa53be1727c4aa0e9b277","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-04 14:28:47.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"4ce5d03586fafd4f3553bca8db4c1523847ef4d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-04 14:41:35.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"445a8273000f0d4fa27cdb1695d1e788ef96c9b3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-04 14:44:39.000000000","message":"Patch Set 26: Code-Review+2\n\ntentative +2; I would prefer to do the lists differently but this patch already does a great job of making our approach much better","accounts_in_message":[],"_revision_number":26},{"id":"b74e4cac2118740e1077e583b0f80701d1f17b00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 15:57:26.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7553b715d0a047828d2790c195a21c95 : SUCCESS in 41m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a9a498242ac64b42bd258b6be05e9c09 : SUCCESS in 53m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ebb250e276764e3e955019427b950d67 : SUCCESS in 1h 10m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7faba9ce1bb64efc82dedd276221ebfa : SUCCESS in 1h 02m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f822a719a01749a0959f74ab6bdb627c : SUCCESS in 1h 14m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3be12fc0f3de46e7b63d3b3eeeea868e : SUCCESS in 22m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/acad313b417a469580df8ebf45aecf1a : SUCCESS in 27m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f757b00c0ba4893a97a072eba35b83c : SUCCESS in 55m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9d608b13b846c783fc1b98598dedf6 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/338aa95c2ad349b0b2ce26b31018982c : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a6b8ee8e1b14638bd6b99036623cd7a : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c0e067dc8f34a8b970b9161b532262a : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313526a00496405ba6b3d92f68b0ad71 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f66db11af9434bdead946e915e02a5eb : SUCCESS in 4m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7eb18ec1b417456abf66ae7a77f6879b : SUCCESS in 19m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/65429b3bc94b4cd0a4526619a013295e : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23b00992139e4128ab8ba5a6d8aafb52 : SUCCESS in 51m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d35a664e72d94c80ba51208f3973659b : SUCCESS in 51m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/87e259ac0791455f8076b1824aaa7cbf : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":26},{"id":"a662a6a7cddde1b02716a8aa8c61565f41ac35e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 17:27:47.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ab15f10110e4574aaa81161757572fa : SUCCESS in 1h 19m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4cc9b971725423ea5c630968346ad7f : SUCCESS in 1h 12m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/96422bebc3c34300b0658c92fff9237c : SUCCESS in 2h 38m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e47e1de02ad4fdf872ef3c8c0675209 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f158b54fd2914495b6a9e022fb3f5c7a : SUCCESS in 7m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/89385887002c4abf9cdb3eccc4dc1ab2 : SUCCESS in 1h 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"2b45827b4589ce6d362234f27f4a4a2418a1b51f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-06 10:15:57.000000000","message":"Patch Set 26: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"83c419e7ecca7bbe2921e01d9cce4e6c8db2f116","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-06 10:50:06.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"e37679a6636992218a5cdf3d478e023a4fdb1239","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-08-16 02:15:16.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"000f2bc6592822f3e65d4fc86153e7a7457c5f93","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-23 15:43:11.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"bcbdbf1560c6bb7c5171f609954644b95fcbb30e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-24 11:16:24.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"72f381a579269ef6883c08fe395e804888e246fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-24 11:16:39.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":28},{"id":"ef381167048b2fe2f04168a96f9bed92520b36bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 12:22:12.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/131fe9888db8491ba3b2d3213bcbb976 : FAILURE in 56m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/543363b0824e41b3bbf714989fff40b8 : SUCCESS in 41m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d50372c74dd246bfbeec1d1c99761659 : FAILURE in 53m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5dee1da507740edaa9c7173a0bcfef0 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f38659c5a8342f186f131abe499fb29 : SUCCESS in 6m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/27e544dafaea482e9374c6099fa69407 : FAILURE in 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"0129443d75467e336ebdcf75f986110f83bd8ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 12:42:25.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f0e3a899fe2d4a4297fe26a18df0a80a : SUCCESS in 28m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2b74f9a66e9c4182b76e960185c5cf30 : SUCCESS in 42m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ac585793ecc54c7a94e41d23054cd45e : SUCCESS in 1h 14m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2c7989ac71ce447cb713d400ea96783c : FAILURE in 36m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/618a8a7f804e49b192fd4e236114627e : SUCCESS in 1h 23m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/382f03c0aee341618f86868aff8ee11e : SUCCESS in 22m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4b259482a6a04249ad5ea5bc992a9608 : SUCCESS in 21m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6902ce463c8744829c4bd9f673c2efcc : FAILURE in 35m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067195d187464aa89b462b03f6477f1b : FAILURE in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e27e947a19144eddb1e32aab43bf85e1 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0470e0b09fe1420ba8ce193519f8d0b3 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ba0973bba4649e096e4145500f07597 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f4db332e5045e2acff000eda3553c9 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2da07f3c7614495c854048dfeb7181ca : SUCCESS in 7m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b0677b18d9d241109d473c3363ff06d0 : SUCCESS in 22m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a88e99a86c344003890bf192806adf9d : SUCCESS in 34m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcb6d8b272cc41479f13515a161696a7 : SUCCESS in 51m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/82f7e7fcf39b42eda7f44214086a0218 : SUCCESS in 51m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3ddf6060e5ba45ada6275f724608fe12 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":28},{"id":"4c0d20ca4c9afbf693e95503789f0be5c26a9fee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-24 12:46:56.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"499376f5fa2945db0d2987557ff3b3e77ff185c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 14:05:43.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/45b95104baaf4d4db352301b9118f6cc : SUCCESS in 29m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5a75d76389c345088c6b68c096c2c1e2 : SUCCESS in 40m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1fa208be4d4e44c9babaab88a188bcd7 : SUCCESS in 1h 10m 11s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/88dca413264f493682f6e9fd51259488 : SUCCESS in 1h 03m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1a4a7a6bac3845a88b27c66363867b40 : SUCCESS in 1h 13m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e81b5f5b314645d48905a7821a3cdd8b : SUCCESS in 38m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/75db9697deef4713b1afa688e1d17471 : SUCCESS in 17m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/48cc58640ba14b8a8e3bcc3eabf0bc64 : SUCCESS in 55m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89836b603c2d4f6ba8bd3eec3f00ac35 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4ce754fa0854bf0a7464c7552ff6f04 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2df0f79d442545c985b9c3530c0351fe : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7acd09b4dd243bcb3248dd3f6adb45c : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673d1d9b89134aa2b26ca0871a5d8682 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22c858f175b34fde8aa023510cacf426 : SUCCESS in 7m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/93260e349b6b40f4946df1467b69d612 : SUCCESS in 21m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f04b144334284ad19480b9122f56bf43 : SUCCESS in 38m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/400cdbac5f21421394bd5a85426b3c3e : SUCCESS in 52m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f282967cf64448d931a9fc42b24db11 : SUCCESS in 48m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/645eb58db42e41a38605c3a60587b3df : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":29},{"id":"54de7669f1cb3ef7f955576157968037c5b512cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 14:46:39.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/011c9c7c67834f69b5543ac0690a4721 : SUCCESS in 1h 32m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f53639a8a7b34755b4d3a35dbce5f650 : SUCCESS in 1h 08m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1046e2fce1944d4a576c0880c8dac64 : SUCCESS in 1h 57m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6bffd1b000d4e9480f44af0ee39c665 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc31b7f6d7b24c8fbfe3359491ee9ed6 : SUCCESS in 4m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c73e5bdce4544d89806dea6a9bf255a : FAILURE in 1h 04m 42s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"b966b06c91413ce4d638a0663a74dca125d42dda","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-25 08:23:51.000000000","message":"Patch Set 29: Code-Review+2","accounts_in_message":[],"_revision_number":29},{"id":"a1aec4aec28bfe1849f5af5715015845f9b53c2a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-25 15:51:35.000000000","message":"Patch Set 29: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"61fb47864efdc0c45dd405c401676f28f7c71c4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 15:51:57.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"2ff3e902df92b9e1fc679eb8ccb9812433faf75f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:18:27.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2ecfdbced2104362a647c79962ddef3b : SUCCESS in 41m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/80ee6c8c598c4b8eb65c22260c2c1808 : SUCCESS in 1h 10m 46s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4496157f26774cd1a3449c257ae3b6d4 : SUCCESS in 1h 25m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3e081d176a40e2905c6083999f92db : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/955ea586a7524482bc4fcbf7ef476a31 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/277198d2c6a24b3ebeba005329473d56 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3afc64bb55462c8441aa108dd162f1 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4cda33e7ef8491eb72e3bca4a6fe17d : SUCCESS in 5m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6285ec3baba5469fb0a1ef7b97e26f30 : SUCCESS in 30m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44a068c9f2014d14af2e14c41789ba2d : SUCCESS in 57m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b1d3f8e1c7240e881f42a830b75775c : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":29},{"id":"7148587aed89d48551724b2987fda528a24f4eba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:18:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":29},{"id":"f7ee7f08ad96c7f647fa33b55b43524d7652ffdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:22:17.000000000","message":"Patch Set 29:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fc5a676550407bbea11d8c339d2c40 : SUCCESS in 2m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94ca807822714013a10f8dca0d2d60b9 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"ffaf102fecc20510513de4898ce85b5b119e1795","revisions":{"54d5991db141b8f25072069edacfb27a9a05a908":{"kind":"REWORK","_number":1,"created":"2021-04-06 11:18:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/1"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:17:47.000000000","tz":120},"subject":"Debian/Ubuntu: install APT keys like APT expects them","message":"Debian/Ubuntu: install APT keys like APT expects them\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in /etc/apt/trusted.gpg.d/ directory.\nAnd this is what this patch does.\n\nThere is one list of APT files for both distributions.\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54d5991db141b8f25072069edacfb27a9a05a908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54d5991db141b8f25072069edacfb27a9a05a908"}]},"branch":"refs/heads/master"},"5e37e2ae635b0d4832a1ba12891530d43583f624":{"kind":"REWORK","_number":2,"created":"2021-04-06 14:26:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/2"}}},"commit":{"parents":[{"commit":"9f8fd2b92a98691681d4f09e607685fa7f96ec40","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f8fd2b92a98691681d4f09e607685fa7f96ec40"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 14:23:53.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for each\nrepo source file, point to proper key. And this is what this patch does.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e37e2ae635b0d4832a1ba12891530d43583f624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e37e2ae635b0d4832a1ba12891530d43583f624"}]},"branch":"refs/heads/master"},"9d665dcbf21feea1789eaf45d1993593cf8f078e":{"kind":"REWORK","_number":3,"created":"2021-04-06 14:33:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/3"}}},"commit":{"parents":[{"commit":"9f8fd2b92a98691681d4f09e607685fa7f96ec40","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f8fd2b92a98691681d4f09e607685fa7f96ec40"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 14:32:49.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for each\nrepo source file, point to proper key. And this is what this patch does.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d665dcbf21feea1789eaf45d1993593cf8f078e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d665dcbf21feea1789eaf45d1993593cf8f078e"}]},"branch":"refs/heads/master"},"2873c084ffbd580f81d620eaef008d1d027af287":{"kind":"REWORK","_number":4,"created":"2021-04-08 08:07:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/4"}}},"commit":{"parents":[{"commit":"b57ad170e9ae3301a1ba42e1415a45d320a60d9f","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b57ad170e9ae3301a1ba42e1415a45d320a60d9f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-08 08:02:11.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for each\nrepo source file, point to proper key. And this is what this patch does.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2873c084ffbd580f81d620eaef008d1d027af287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2873c084ffbd580f81d620eaef008d1d027af287"}]},"branch":"refs/heads/master"},"a504f40104a4ff10ffc46b12c0302e544fb188cc":{"kind":"REWORK","_number":5,"created":"2021-04-13 10:56:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/5"}}},"commit":{"parents":[{"commit":"b0ce7ca228c442f1db4edb22467da69dff8458e4","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0ce7ca228c442f1db4edb22467da69dff8458e4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:38:42.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for each\nrepo source file, point to proper key. And this is what this patch does.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a504f40104a4ff10ffc46b12c0302e544fb188cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a504f40104a4ff10ffc46b12c0302e544fb188cc"}]},"branch":"refs/heads/master"},"23f3f02c45c337b2605f41fd04dc616b2f8d9ce6":{"kind":"REWORK","_number":6,"created":"2021-04-14 17:02:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/6"}}},"commit":{"parents":[{"commit":"03fb814ef54b3d63846aaf39219c5f17a695bd2a","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03fb814ef54b3d63846aaf39219c5f17a695bd2a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-14 17:01:54.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23f3f02c45c337b2605f41fd04dc616b2f8d9ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23f3f02c45c337b2605f41fd04dc616b2f8d9ce6"}]},"branch":"refs/heads/master"},"d043385205a914d2d39f050ad9391e8dd8df7a51":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-04-19 17:19:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/7"}}},"commit":{"parents":[{"commit":"1535b9a4cfb3ede8aa57dc710e0bc8128b7bd987","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1535b9a4cfb3ede8aa57dc710e0bc8128b7bd987"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-19 17:18:51.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d043385205a914d2d39f050ad9391e8dd8df7a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d043385205a914d2d39f050ad9391e8dd8df7a51"}]},"branch":"refs/heads/master"},"2c7fb50e1b6cdf3f8aeee5976828e42e8c73a415":{"kind":"REWORK","_number":8,"created":"2021-04-20 12:55:33.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/8"}}},"commit":{"parents":[{"commit":"e5d634d55ea12625b186e318669f7fac17d873f5","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d634d55ea12625b186e318669f7fac17d873f5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-20 12:55:04.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c7fb50e1b6cdf3f8aeee5976828e42e8c73a415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c7fb50e1b6cdf3f8aeee5976828e42e8c73a415"}]},"branch":"refs/heads/master"},"7ffcf16cfa9f03d2e75749e18d6254baab27987d":{"kind":"REWORK","_number":9,"created":"2021-04-21 11:27:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/9"}}},"commit":{"parents":[{"commit":"e5d634d55ea12625b186e318669f7fac17d873f5","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d634d55ea12625b186e318669f7fac17d873f5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-21 11:27:02.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ffcf16cfa9f03d2e75749e18d6254baab27987d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ffcf16cfa9f03d2e75749e18d6254baab27987d"}]},"branch":"refs/heads/master"},"b12717eeb2d7b1f77f4f2cebdfa88f7e95b01f7d":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-04-21 11:28:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/10"}}},"commit":{"parents":[{"commit":"02ae29202b1efd47b99f9d683716466b0206ddae","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02ae29202b1efd47b99f9d683716466b0206ddae"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-21 11:28:11.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b12717eeb2d7b1f77f4f2cebdfa88f7e95b01f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b12717eeb2d7b1f77f4f2cebdfa88f7e95b01f7d"}]},"branch":"refs/heads/master","description":"Rebase"},"3dd995dc45243f435d9dd8a71b66777dce3dafb6":{"kind":"REWORK","_number":11,"created":"2021-04-21 22:04:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/11"}}},"commit":{"parents":[{"commit":"f7e748a09a3f9e08642e0a6682c6a1cd48ad3652","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7e748a09a3f9e08642e0a6682c6a1cd48ad3652"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-21 22:04:36.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3dd995dc45243f435d9dd8a71b66777dce3dafb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3dd995dc45243f435d9dd8a71b66777dce3dafb6"}]},"branch":"refs/heads/master"},"6d3bc9232bfdb6b504f149e95ac3884c547ffee1":{"kind":"REWORK","_number":12,"created":"2021-04-22 06:35:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/12"}}},"commit":{"parents":[{"commit":"f7e748a09a3f9e08642e0a6682c6a1cd48ad3652","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7e748a09a3f9e08642e0a6682c6a1cd48ad3652"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-22 06:35:15.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d3bc9232bfdb6b504f149e95ac3884c547ffee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d3bc9232bfdb6b504f149e95ac3884c547ffee1"}]},"branch":"refs/heads/master"},"57cf8424d2df6c05d7a66c3779ec33b9b3bfda7d":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-04-22 09:47:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/13"}}},"commit":{"parents":[{"commit":"115560c94f6d01a62a4ca07126bdf51233df8532","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115560c94f6d01a62a4ca07126bdf51233df8532"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-22 09:46:56.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57cf8424d2df6c05d7a66c3779ec33b9b3bfda7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57cf8424d2df6c05d7a66c3779ec33b9b3bfda7d"}]},"branch":"refs/heads/master"},"624d65e6e01594df4f039317d5e124c0e19be4dc":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-04-22 17:16:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/14"}}},"commit":{"parents":[{"commit":"ac8c82b43a696a5dbef61cf12391a10f9f8e07bb","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac8c82b43a696a5dbef61cf12391a10f9f8e07bb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-22 17:14:24.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/624d65e6e01594df4f039317d5e124c0e19be4dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/624d65e6e01594df4f039317d5e124c0e19be4dc"}]},"branch":"refs/heads/master"},"3c33d83bed7a761895f970ef1d2ce764eafffadf":{"kind":"REWORK","_number":15,"created":"2021-04-26 18:03:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/15"}}},"commit":{"parents":[{"commit":"e0f8cbc50cf7184930a501b0f5452a64f523cd6f","subject":"move to Debian \u0027bullseye\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0f8cbc50cf7184930a501b0f5452a64f523cd6f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-26 18:01:28.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c33d83bed7a761895f970ef1d2ce764eafffadf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c33d83bed7a761895f970ef1d2ce764eafffadf"}]},"branch":"refs/heads/master"},"4955003e52b05b753a1341c739e4319643b60ed3":{"kind":"REWORK","_number":16,"created":"2021-05-03 18:14:00.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/16"}}},"commit":{"parents":[{"commit":"444313975e6099055c5573363eea8af5a035fdbd","subject":"Install RabbitMQ repo key on Ubuntu and Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/444313975e6099055c5573363eea8af5a035fdbd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-03 18:11:42.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4955003e52b05b753a1341c739e4319643b60ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4955003e52b05b753a1341c739e4319643b60ed3"}]},"branch":"refs/heads/master"},"9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-05-05 14:43:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/17"}}},"commit":{"parents":[{"commit":"dccb1a50e8545900711e2d579c87958ac447569c","subject":"Merge \"Deprecate tempest and rally\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dccb1a50e8545900711e2d579c87958ac447569c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-05 14:43:43.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2"}]},"branch":"refs/heads/master","description":"Rebase"},"36e0c7e198654180e759cf07f041e838f8f035fd":{"kind":"REWORK","_number":18,"created":"2021-05-06 12:05:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/18"}}},"commit":{"parents":[{"commit":"68fc72cd5d5003025da53cd1160e9cd19367cb10","subject":"base: check for compatible distro using just sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68fc72cd5d5003025da53cd1160e9cd19367cb10"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-06 12:02:27.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e0c7e198654180e759cf07f041e838f8f035fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e0c7e198654180e759cf07f041e838f8f035fd"}]},"branch":"refs/heads/master"},"882307862cfb208b497751b6a1ab0d95100ff95a":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2021-05-17 07:00:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/19"}}},"commit":{"parents":[{"commit":"73c456f85b4e861e4be12282a67ca59f5f319357","subject":"Merge \"docs: update supported OS distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73c456f85b4e861e4be12282a67ca59f5f319357"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 06:57:18.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/882307862cfb208b497751b6a1ab0d95100ff95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/882307862cfb208b497751b6a1ab0d95100ff95a"}]},"branch":"refs/heads/master"},"11a60e438af93ecce75aa19eb2e722fa5ebe1809":{"kind":"REWORK","_number":20,"created":"2021-05-17 12:39:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/20"}}},"commit":{"parents":[{"commit":"0506ede84f97159a8156ad51aea395d0868e6141","subject":"Drop ppc64le support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0506ede84f97159a8156ad51aea395d0868e6141"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 12:39:17.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated already and in Debian \u0027bullseye\u0027 it even complain\nwhen we use it the way we did.\n\nProper way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nThere is one list of APT files for both distributions.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11a60e438af93ecce75aa19eb2e722fa5ebe1809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11a60e438af93ecce75aa19eb2e722fa5ebe1809"}]},"branch":"refs/heads/master"},"eb58d81cc7b146d6c7c7b707141c8f008edf94a5":{"kind":"REWORK","_number":21,"created":"2021-05-17 16:31:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/21"}}},"commit":{"parents":[{"commit":"0506ede84f97159a8156ad51aea395d0868e6141","subject":"Drop ppc64le support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0506ede84f97159a8156ad51aea395d0868e6141"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 16:30:55.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated and in Debian \u0027bullseye\u0027 it complain when we use\nit the way we did.\n\nThe right way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nBoth distros use the same lists now.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb58d81cc7b146d6c7c7b707141c8f008edf94a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb58d81cc7b146d6c7c7b707141c8f008edf94a5"}]},"branch":"refs/heads/master"},"1edc99a9067cb3e1cd9be46d9bae00143dad841c":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2021-06-28 09:05:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/22"}}},"commit":{"parents":[{"commit":"8e020c5f86e033d9baba2c80c39f800bbd3cf2de","subject":"Merge \"tox: Use allowlist_externals instead of whitelist_externals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e020c5f86e033d9baba2c80c39f800bbd3cf2de"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-06-28 09:05:37.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated and in Debian \u0027bullseye\u0027 it complain when we use\nit the way we did.\n\nThe right way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nBoth distros use the same lists now.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1edc99a9067cb3e1cd9be46d9bae00143dad841c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1edc99a9067cb3e1cd9be46d9bae00143dad841c"}]},"branch":"refs/heads/master","description":"Rebase"},"3f9dc572cbe0e583df809063187f54a710caa66b":{"kind":"REWORK","_number":23,"created":"2021-07-06 11:54:00.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/23"}}},"commit":{"parents":[{"commit":"5e0b238b01250392fa9f92c069a7253902d6b325","subject":"Merge \"Bump up cAdvisor\u0027s version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e0b238b01250392fa9f92c069a7253902d6b325"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-07-06 11:52:08.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated and in Debian \u0027bullseye\u0027 it complain when we use\nit the way we did.\n\nThe right way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nBoth distros use the same lists now.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f9dc572cbe0e583df809063187f54a710caa66b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f9dc572cbe0e583df809063187f54a710caa66b"}]},"branch":"refs/heads/master"},"92d5c2850da48f63faf709069f42ba34d0dd6fa8":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2021-07-28 16:04:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/24"}}},"commit":{"parents":[{"commit":"53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1","subject":"Add boto3 as s3 dependency for Glance container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-07-28 16:04:04.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated and in Debian \u0027bullseye\u0027 it complain when we use\nit the way we did.\n\nThe right way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nBoth distros use the same lists now.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92d5c2850da48f63faf709069f42ba34d0dd6fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92d5c2850da48f63faf709069f42ba34d0dd6fa8"}]},"branch":"refs/heads/master","description":"Rebase"},"25737eb38454ed9ae19b7a244ea2b841bc1acf3c":{"kind":"REWORK","_number":25,"created":"2021-08-04 13:29:52.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/25"}}},"commit":{"parents":[{"commit":"53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1","subject":"Add boto3 as s3 dependency for Glance container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-08-04 13:29:20.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated and in Debian \u0027bullseye\u0027 it complain when we use\nit the way we did.\n\nThe right way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nBoth distros use the same keys now.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25737eb38454ed9ae19b7a244ea2b841bc1acf3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25737eb38454ed9ae19b7a244ea2b841bc1acf3c"}]},"branch":"refs/heads/master"},"12c3754687aff3d9a70ba2290c91bbcfa0321762":{"kind":"REWORK","_number":26,"created":"2021-08-04 14:41:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/26"}}},"commit":{"parents":[{"commit":"ea6ceef686c018d810a2ff20ca4ba81aa77164e5","subject":"Merge \"drop leftovers of RHEL support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea6ceef686c018d810a2ff20ca4ba81aa77164e5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-08-04 14:40:48.000000000","tz":120},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated and in Debian \u0027bullseye\u0027 it complain when we use\nit the way we did.\n\nThe right way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nBoth distros use the same keys now.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12c3754687aff3d9a70ba2290c91bbcfa0321762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12c3754687aff3d9a70ba2290c91bbcfa0321762"}]},"branch":"refs/heads/master"},"38e7ec00b7e95513d8de79bc0528ff7b5a82217a":{"kind":"REWORK","_number":27,"created":"2021-08-24 11:16:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/27"}}},"commit":{"parents":[{"commit":"ea6ceef686c018d810a2ff20ca4ba81aa77164e5","subject":"Merge \"drop leftovers of RHEL support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea6ceef686c018d810a2ff20ca4ba81aa77164e5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-08-24 11:16:15.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated and in Debian \u0027bullseye\u0027 it complain when we use\nit the way we did.\n\nThe right way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nBoth distros use the same keys now.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38e7ec00b7e95513d8de79bc0528ff7b5a82217a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38e7ec00b7e95513d8de79bc0528ff7b5a82217a"}]},"branch":"refs/heads/master"},"ce127fefd5ff226dd48c309f6963a7d6e9e5e8db":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2021-08-24 11:16:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/28"}}},"commit":{"parents":[{"commit":"ee84d3b8b9c3ac68a6f33a55e17acd17c0850805","subject":"Merge \"Config: Set default home to /var/lib/{user}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee84d3b8b9c3ac68a6f33a55e17acd17c0850805"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-08-24 11:16:39.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated and in Debian \u0027bullseye\u0027 it complain when we use\nit the way we did.\n\nThe right way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nBoth distros use the same keys now.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce127fefd5ff226dd48c309f6963a7d6e9e5e8db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce127fefd5ff226dd48c309f6963a7d6e9e5e8db"}]},"branch":"refs/heads/master","description":"Rebase"},"ffaf102fecc20510513de4898ce85b5b119e1795":{"kind":"REWORK","_number":29,"created":"2021-08-24 12:46:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/784923/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/784923/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/784923/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/784923/29"}}},"commit":{"parents":[{"commit":"ee84d3b8b9c3ac68a6f33a55e17acd17c0850805","subject":"Merge \"Config: Set default home to /var/lib/{user}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee84d3b8b9c3ac68a6f33a55e17acd17c0850805"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:14:49.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-08-24 12:46:52.000000000","tz":0},"subject":"Debian/Ubuntu: handle APT keys in proper way","message":"Debian/Ubuntu: handle APT keys in proper way\n\napt-key is deprecated and in Debian \u0027bullseye\u0027 it complain when we use\nit the way we did.\n\nThe right way is to store key files in separate directory and then, for\neach repo source file, point to proper key. And this is what this patch\ndoes.\n\nBoth distros use the same keys now.\n\nMore info: https://michael-prokop.at/blog/2021/02/16/how-to-properly-use-3rd-party-debian-repository-signing-keys-with-apt/\n\nChange-Id: Ie19bf46236907d79cb42a5e94cd1067e952b8cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffaf102fecc20510513de4898ce85b5b119e1795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffaf102fecc20510513de4898ce85b5b119e1795"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
