)]}'
{"id":"openstack%2Fkolla~784927","triplet_id":"openstack%2Fkolla~master~I3fd97893e710b02371eed2a409c80783abf681a0","project":"openstack/kolla","branch":"master","hashtags":["debuntu-cleanups"],"change_id":"I3fd97893e710b02371eed2a409c80783abf681a0","subject":"rally: use Python 3 on Debian/Ubuntu","status":"MERGED","created":"2021-04-06 11:33:16.000000000","updated":"2021-04-12 03:52:46.000000000","submitted":"2021-04-06 18:12:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784927","meta_rev_id":"88e43a016fb6799a7660d277133c185c401c29bf","_number":784927,"virtual_id_number":784927,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 18:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-12 03:52:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-06 15:32:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-06 15:32:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-12 03:52:46.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 12:12:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-06 15:05:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 15:32:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"61632d09ab4e9c000bb45ed36b7909815bf11b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 11:33:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1f968b10c9f05f6c1bce57b33ec03038c016cce","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 11:33:32.000000000","message":"Hashtag added: debuntu-cleanups","accounts_in_message":[],"_revision_number":1},{"id":"c99aa785731398e5e99e29e911f11d4b3ca302d6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 11:33:50.000000000","message":"Patch Set 1: Review-Priority+1\n\nneeded for Bullseye move","accounts_in_message":[],"_revision_number":1},{"id":"f79fdb42f48d5582f1ef184b20a401b5f4bef137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 12:08:20.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"86b19ecd6c2ec494d78bd63c0761e5ad912020ee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-06 12:12:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9efbc080478ac940cfc1873c70b49a06910a526c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 13:15:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"947ed7e841ccf1b370086285eb541c0d05d354de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 15:05:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/201c8bb1c1ed4b148fc4c1932dce7dc1 : SUCCESS in 31m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b88832ac848f43928988d3ca3782f814 : SUCCESS in 46m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ce8bbce4a8784d8f9e0cf18ad3858c34 : SUCCESS in 1h 14m 27s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fee201638f644977b46d05d3526a21e6 : SUCCESS in 1h 10m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/622e2eb6ee0848289ad84d1df3cfcb44 : POST_FAILURE in 1h 46m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3d22580dd8e04e5b8e1dbeec2d4dc0cc : SUCCESS in 43m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5bff8f1f9ea24fc6a85222bda86053bf : SUCCESS in 17m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf43dac12a2e437a861a8d2b68ff9f05 : SUCCESS in 54m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38cc82f32a3d47808f2d64a2efc925d5 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7a2182db2014b79a19c119447b3a6d3 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bd111889cbc47dab2bf26a84022a65b : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1ccecf45eb342bca7fffd25b5aa668a : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b814430c8ce948e49238eef56f0e856e : SUCCESS in 6m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/639ee99ab4f947ae81f6b1f51fddbaf9 : SUCCESS in 19m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e77a26653244a7dae52a5906197e4ab : SUCCESS in 32m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/90c04a9273a44f2b835f3239554fcaf5 : SUCCESS in 57m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8660da127ce24f32bba59aa671f05afe : SUCCESS in 53m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/60e73a792f734577a4d0addd394fd053 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ff01c30622fd5cf689bb76f03ef1f6886ada18f2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-06 15:32:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"96baec08c57d519d38db48e72615b3d025e0f75c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-06 15:32:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8b9bc7d45b421d319a810f58ea2082d9e03effcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 16:55:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/56f4c6d7eeb44d60b3bd555c33b88c5d : SUCCESS in 28m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5fd2b174cc1f4ba797f47c246ac89133 : SUCCESS in 41m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/edc25159248a4baaa9b9ee54ac615802 : SUCCESS in 1h 06m 37s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fe5e5393d978404d86d4bd9a91d1fa1d : SUCCESS in 1h 02m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/63eca845e3c14fd9a426fb199f5ae0e8 : SUCCESS in 1h 15m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/43dff70a918b4d86bc3ffe7dc4e186e9 : SUCCESS in 50m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c2865a625f65458084a9f8e736dd7500 : SUCCESS in 22m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f5f93a3c26a4c988426f80b36759615 : FAILURE in 23m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b543b98b4c3451cb7a35eba671a9da0 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00d36c27e7fb4a61a3f8057767de935f : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32e6cd53e81843248807ef370ca0e453 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82ab57cf787c427596d99866e11f58f2 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58dcbfcd334d4efbbeedc0bc2564ff56 : SUCCESS in 7m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85db5c24044946f5be5684326010c947 : SUCCESS in 23m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8ebc3c1100748269a097988cf73ccde : SUCCESS in 47m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3ef0db900d740c2a328d804388c3edd : SUCCESS in 1h 05m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/904f88e83bc6445994241b6e9b1ebd2e : SUCCESS in 1h 02m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ad4e71b2e19443ca804b746efba881cb : SUCCESS in 1h 17m 19s","accounts_in_message":[],"_revision_number":3},{"id":"e7f738dc6cbf5b57e4a731cf182a0ded3e4330ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 16:55:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e43ebede44fe22790af2a7b74ac56165a126026d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 18:12:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c9b090852e4a496e9a595c737988511c : SUCCESS in 47m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/80734e101a024818b3582c35f50b5680 : SUCCESS in 1h 03m 32s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/36d9ba162e134f61a853209ba6283be8 : SUCCESS in 1h 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7683aacd0344e4899aa272d7c72af38 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11694276f6e741469ca435c3fce03480 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92f93224a70247e2b263d8e5a084825d : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91606145f56c43aaad49244f19a7c9ce : SUCCESS in 5m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c0fec842fba4f879cf9f5ad910b2edd : SUCCESS in 53m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff5deb9dcca5480ebff3631c797a830e : SUCCESS in 57m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ea5b0c1c088c4a09ad39510652cfefc5 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":3},{"id":"6a16bd30da77f2727cc8e6442e179a5b91b9a2a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 18:12:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d36718fb5705c331e77268fa85b327fc3680b62c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 18:13:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2813b926d51748f88426aa5aa19efa61 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3},{"id":"210a622bddccd9cc9135d01cccfea0925b3c9814","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:52:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"53c667364cad70c02dfd983316c9b769a664fc12","revisions":{"ca59e28b6789cb0f2d3b18edfa82821d14a1f1ef":{"kind":"REWORK","_number":1,"created":"2021-04-06 11:33:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/784927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/784927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/784927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/784927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/784927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/784927/1"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:32:17.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:33:09.000000000","tz":120},"subject":"rally: use Python 3 on Debian/Ubuntu","message":"rally: use Python 3 on Debian/Ubuntu\n\nChange-Id: I3fd97893e710b02371eed2a409c80783abf681a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca59e28b6789cb0f2d3b18edfa82821d14a1f1ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca59e28b6789cb0f2d3b18edfa82821d14a1f1ef"}]},"branch":"refs/heads/master"},"115ea243ce782f7e3ee790204f6ccf49cd03b519":{"kind":"NO_CHANGE","_number":2,"created":"2021-04-06 12:08:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/784927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/784927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/784927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/784927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/784927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/784927/2"}}},"commit":{"parents":[{"commit":"ac4e496eeb93943c52a64c110a52d433378aa88b","subject":"Merge \"octavia: Fix log dir permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e496eeb93943c52a64c110a52d433378aa88b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:32:17.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:50:40.000000000","tz":120},"subject":"rally: use Python 3 on Debian/Ubuntu","message":"rally: use Python 3 on Debian/Ubuntu\n\nChange-Id: I3fd97893e710b02371eed2a409c80783abf681a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115ea243ce782f7e3ee790204f6ccf49cd03b519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115ea243ce782f7e3ee790204f6ccf49cd03b519"}]},"branch":"refs/heads/master"},"53c667364cad70c02dfd983316c9b769a664fc12":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-06 13:15:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/784927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/784927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/784927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/784927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/784927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/784927/3"}}},"commit":{"parents":[{"commit":"0db9b010afb27d21a540aa9b15c733c5919825fe","subject":"Merge \"CI: add monasca-grafana to allowed-to-fail images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0db9b010afb27d21a540aa9b15c733c5919825fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 11:32:17.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-06 13:15:47.000000000","tz":0},"subject":"rally: use Python 3 on Debian/Ubuntu","message":"rally: use Python 3 on Debian/Ubuntu\n\nChange-Id: I3fd97893e710b02371eed2a409c80783abf681a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53c667364cad70c02dfd983316c9b769a664fc12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53c667364cad70c02dfd983316c9b769a664fc12"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
