)]}'
{"id":"openstack%2Fkolla~785222","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ica59f7407a8d73b6a194f09f44754b44d28af654","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Ica59f7407a8d73b6a194f09f44754b44d28af654","subject":"Bump OpenStack versions for Train","status":"MERGED","created":"2021-04-07 15:19:35.000000000","updated":"2021-04-07 18:57:44.000000000","submitted":"2021-04-07 18:56:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785222","meta_rev_id":"f8b51d72f904c53b6f2a7289af1d24cc814edd16","_number":785222,"virtual_id_number":785222,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 18:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-07 15:20:26.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-07 15:49:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-07 17:28:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 15:20:26.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-07 15:49:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-07 17:05:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb488394ac8d879c0233981c7a3df2754b29eb42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 15:19:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5694c0496660acae4fd4e7508134afdff427011","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-07 15:20:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ba261984cc28b81ff4b071678d5a51c22d7590be","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 15:49:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea7ae407410f50879ab77e8201d8d135c1959ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 17:05:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/72ee1513af3a40fdb5993aa9bcea4b8f : SUCCESS in 37m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/88dfdaa49d534f47a5b61b1615227fa2 : SUCCESS in 52m 53s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/500f1389cfa545b3be24f51563b795d0 : SUCCESS in 27m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1a2d06283ec94469a51f7e8fcb59adae : SUCCESS in 40m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e64bde9d82c34cd6b983c2300a29c0e5 : SUCCESS in 1h 16m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/21cf632112804b1992cd364b94795536 : SUCCESS in 1h 03m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/36d5641542cf472b885855b69be3b46d : SUCCESS in 1h 22m 49s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1c85ad5a9c7d4c3ea2aafdf581b1b862 : SUCCESS in 1h 02m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/eedc12572c1e45bf99e3326bdbbfd9e3 : SUCCESS in 1h 39m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/12330792972945a7839a7e2092938a2b : SUCCESS in 1h 01m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ac54d056bcd84239997ec2f24f451690 : SUCCESS in 24m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/02e6bb76ac8f45da97fb78435f90efe7 : SUCCESS in 19m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/21ae912159f54b10a94ffcd75724b0b8 : SUCCESS in 53m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24d688a795849fa81a3ac6db3944422 : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ca6c64626afb433c9380277a8b3d5d9c : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d675592a4124537867feb48523a35d1 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/941c87f747b84ace957410570346bf93 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40de402b3a14d28bda6b0cf0e214518 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/561b68ae80c44db69a49e512dff89346 : SUCCESS in 25m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cb22c63b1a904f14bca8cd25ad453abd : SUCCESS in 27m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/47d8f4bb5b5f4a74b2aa78985dbca685 : SUCCESS in 53m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ea3b1501bc1a4af3b8f2dc93e3b662d4 : SUCCESS in 56m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7bc8b3ba00ff4087887b2710fb9328ee : SUCCESS in 1h 01m 18s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"362d3e4899ba15f9156ac9584936a50726b151ce","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 17:28:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb8399b550de8cd186e949c216d80a59c5f672f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 17:28:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02aad7bddd702c932148ef3fc9148c6d6427ad61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 18:56:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/af7769c7f50645b48ed4393e89d1922d : SUCCESS in 42m 35s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/92262b11be15481f8d1e1e68d3e56f7f : SUCCESS in 36m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b791d3125c9a4d968c0224593a59e0a4 : SUCCESS in 1h 23m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4eb7c2c205f848949db09057c4df5731 : SUCCESS in 1h 01m 39s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/97552f8dbe814a1da3d592aff4a99fa5 : SUCCESS in 1h 23m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d434900cc6474c348bfa50c25d766e00 : SUCCESS in 49m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc1c6f4fcc14c8ab1591af0f16467e0 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6647c8c861cd40f9a84a38eda49e69bc : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7707b8d1b1f64a7c9a8c0361da531987 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2dab3f548bb04625b022af96813d7ff0 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e563fb7c15b402f80110c009e046e31 : SUCCESS in 6m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ce35fe7d0574b0dbdc4a641c1e28116 : SUCCESS in 26m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/06afbdfbe18b435b9206c93cfbda56df : SUCCESS in 1h 01m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/338d7d32f6df4ff6b1282bb5e97313b4 : SUCCESS in 1h 13m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"6a7d7af8c30160ffb88e19b06979c9a14c6962d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 18:56:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f8b51d72f904c53b6f2a7289af1d24cc814edd16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 18:57:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b87e65f8c54d4409950bb9cde979246f : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3251922dd517d114cf0a9e9a4a9fbac8d1bdb9fd","revisions":{"3251922dd517d114cf0a9e9a4a9fbac8d1bdb9fd":{"kind":"REWORK","_number":1,"created":"2021-04-07 15:19:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/785222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785222/1"}}},"commit":{"parents":[{"commit":"29f875b6eb6ca09f901a61e4687a17aa07b3909e","subject":"Merge \"Drop lower-constraints\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29f875b6eb6ca09f901a61e4687a17aa07b3909e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 15:19:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 15:19:22.000000000","tz":60},"subject":"Bump OpenStack versions for Train","message":"Bump OpenStack versions for Train\n\nChange-Id: Ica59f7407a8d73b6a194f09f44754b44d28af654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3251922dd517d114cf0a9e9a4a9fbac8d1bdb9fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3251922dd517d114cf0a9e9a4a9fbac8d1bdb9fd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
