)]}'
{"id":"openstack%2Fkolla~785255","triplet_id":"openstack%2Fkolla~master~I8dd4c8cae0e9d4a06693f61ce317d16a064003e0","project":"openstack/kolla","branch":"master","topic":"masakari-monitors-root","hashtags":[],"change_id":"I8dd4c8cae0e9d4a06693f61ce317d16a064003e0","subject":"[WIP] Run Masakari Monitors as root","status":"ABANDONED","created":"2021-04-07 19:16:53.000000000","updated":"2021-04-08 12:06:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e2e6c66b9d08242d80b21e4654b07d1607bf96f6","_number":785255,"virtual_id_number":785255,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-07 20:17:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 20:17:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b3f355f34c29ccf6f21efec5ad6e1056eca8ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-07 19:16:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5e070d559ad19d3c80883e735ae8e2bd069b182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 20:17:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/befbbc61d70c42568ba0dd66b1d8d14d : SUCCESS in 25m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/004bc14ce23742f9bdeb75a55b7662f7 : SUCCESS in 47m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/40cbfa19b1784e7694de102eadf916ef : SUCCESS in 45m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/46e92270c5954409ac59890c6139105b : SUCCESS in 19m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3cc8ab1bee734b6c9ed279a8968ebce4 : FAILURE in 26m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb867f4349cc47c99fc1f027524a3769 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03d7e6e749ec4fa5863ae94830de54a6 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85f6024fffc34c869c02ff7d60ed50bc : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/761d538400bd4c63a9f496b6f0d28613 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5794991a378457bb638644d7098d323 : SUCCESS in 5m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/10e1e5d681cf4c56b01ad3b347206b42 : SUCCESS in 24m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b601d2b3a92a4bc88662177e739605e5 : SUCCESS in 38m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6cb36a62a8a484a910748b9b53127dd : SUCCESS in 54m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/876a38746962428db0adff8e0fad5c05 : SUCCESS in 56m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/375079d4cd89495dac141a1500350656 : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e2e6c66b9d08242d80b21e4654b07d1607bf96f6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 12:06:21.000000000","message":"Abandoned\n\nit was not required (as monitors use the privsep)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e80b54ae511f1bfa6cff51664177aaff9d07cdd8","revisions":{"e80b54ae511f1bfa6cff51664177aaff9d07cdd8":{"kind":"REWORK","_number":1,"created":"2021-04-07 19:16:53.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/55/785255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/785255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/785255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/785255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/785255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/785255/1"}}},"commit":{"parents":[{"commit":"eb83693ea114152b2ff16c2e01be5f82736bbb19","subject":"rabbitmq: use Focal repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb83693ea114152b2ff16c2e01be5f82736bbb19"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-07 19:16:28.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-07 19:16:28.000000000","tz":0},"subject":"[WIP] Run Masakari Monitors as root","message":"[WIP] Run Masakari Monitors as root\n\nChange-Id: I8dd4c8cae0e9d4a06693f61ce317d16a064003e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e80b54ae511f1bfa6cff51664177aaff9d07cdd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e80b54ae511f1bfa6cff51664177aaff9d07cdd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
