)]}'
{"id":"openstack%2Fkolla~785329","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I95aa64f0b292cd020574dbedd320ef90529cc7a4","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I95aa64f0b292cd020574dbedd320ef90529cc7a4","subject":"base: configure curl before first use","status":"MERGED","created":"2021-04-08 08:40:44.000000000","updated":"2021-04-12 03:50:42.000000000","submitted":"2021-04-08 11:18:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785329","cherry_pick_of_change":785328,"cherry_pick_of_patch_set":1,"meta_rev_id":"490ce530ed35d8b17ce831c9787b593a6a461c09","_number":785329,"virtual_id_number":785329,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 11:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-08 08:41:04.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-08 08:41:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-08 08:41:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 08:40:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-08 10:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e67df320da67656044161cc662a504131504354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 08:40:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"2486db725d999e93f678e14a29c4b4e00ddb8bc3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-08 08:41:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dddeea40bbd03a08723ca89f3e3bab4008be8d5c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 08:41:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4420f53e489dd09d668cbbccab76918c2a02e992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:04:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/006a7ca502b44aad9b1c17d6f2f499e3 : SUCCESS in 33m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cd3aa65ec2154e73a81e6a5239b5a1a6 : SUCCESS in 36m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/63fb3b1d84d345d7b431e874bdcd802b : SUCCESS in 55m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6c11b1518e6c45eaa6e6d070aca10955 : SUCCESS in 57m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/853bdcc9e28645f0addf3a06bdcc667e : SUCCESS in 1h 05m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/190122a57a5f440bb32efe8b0322f1e2 : SUCCESS in 32m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/22c4590c049041ebbc593ad1576fdd7f : SUCCESS in 31m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2785ce3cd4484f0c9a91b7fe9d8bd537 : SUCCESS in 56m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8a180a92fe74ed0814ea952f3c85047 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5575e7d611e442eb091078f0e7f21d2 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f15d2e4b97145599a546ec484827953 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8965b283aa94568b44d9c1d08a5875c : SUCCESS in 7m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d75e3673caa413ea937936b34df0547 : SUCCESS in 42m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/73b39192f2c04af6acebe6c6610e8878 : SUCCESS in 41m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8fde39d35bbe47248ba0faebdfd9d9ca : SUCCESS in 1h 00m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/393e9fc25ee94183b0d48744128cce78 : SUCCESS in 59m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ad25f58abc754d8884196875185f2ce6 : SUCCESS in 1h 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"946e664637eed80d9bd65b8aa28033fa869cbb7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:04:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"600ec011d74d9e1105cf55fdbf1a099e2270d9bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:18:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/052b80b032834202be6a09cb37611eba : SUCCESS in 29m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/97b52adc403a4fcfbea7bc4201549d49 : SUCCESS in 58m 16s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c0dc3ec1a9e644dfadb88501eafc60a0 : SUCCESS in 1h 09m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b44b621ddafa49c3b407f6f50ee21e0c : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9443d6bca538409e8d9517b660fb1b36 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8382626a594e41e7851394cbb94386b6 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba98635fa9a647d3abd8814aa95c1023 : SUCCESS in 5m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/33d952565be64cdf927c3a3d081420af : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/63cf75477c16492b8dca1a1219f8b925 : SUCCESS in 53m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dd66f7d66f534474bbcb9272e810499b : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1b4fda3c9fa0cb2442666534d46fcc7fc115262f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:18:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cdf258ffd96ba43aa18592721bb7ed2608533d59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:19:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4138cca15e294e4e82fbc6759fbf3ec2 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1},{"id":"490ce530ed35d8b17ce831c9787b593a6a461c09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:50:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00339ecc48e301727a648d9e9b0713bac13fefa8","revisions":{"00339ecc48e301727a648d9e9b0713bac13fefa8":{"kind":"REWORK","_number":1,"created":"2021-04-08 08:40:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/29/785329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/785329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/785329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/785329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/785329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/785329/1"}}},"commit":{"parents":[{"commit":"097ff32cdb068ec9319438e5c1895680fd3e3f45","subject":"Merge \"rabbitmq: bump Erlang requirements to v23 in Debian\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/097ff32cdb068ec9319438e5c1895680fd3e3f45"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-07 09:55:23.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 08:40:44.000000000","tz":0},"subject":"base: configure curl before first use","message":"base: configure curl before first use\n\nWe want to have all uses of curl behave the same so we ship config file\nfor it in all images.\n\nSo do it at start of image so \u0027base\u0027 makes use of config too.\n\nChange-Id: I95aa64f0b292cd020574dbedd320ef90529cc7a4\n(cherry picked from commit 4328a5b3cc0e90e32952ecfeab2b63407ee8e22a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00339ecc48e301727a648d9e9b0713bac13fefa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00339ecc48e301727a648d9e9b0713bac13fefa8"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
