)]}'
{"id":"openstack%2Fkolla~785330","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I95aa64f0b292cd020574dbedd320ef90529cc7a4","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I95aa64f0b292cd020574dbedd320ef90529cc7a4","subject":"base: configure curl before first use","status":"MERGED","created":"2021-04-08 08:42:00.000000000","updated":"2021-04-08 12:01:54.000000000","submitted":"2021-04-08 12:00:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785330","cherry_pick_of_change":785329,"cherry_pick_of_patch_set":1,"meta_rev_id":"55f054020f186f1a4d97871c160ba40d3001a66c","_number":785330,"virtual_id_number":785330,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 12:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-08 08:44:34.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-08 08:44:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-08 08:44:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 08:42:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-08 10:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62dfa77488e315c46a65f57926c48fb181bddd13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 08:42:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"3313156746fb8f1a0c13fba961cda767c66d5d22","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-08 08:44:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af1b2c85a2fdf5ceca468f93456d37f9c4431939","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 08:44:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"113a332ae3bce327c91220d887d97b9b86fa059c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:21:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/01be3fe413514b8ca58c2da53160c1aa : SUCCESS in 40m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/619782041632417ca37764bf5fd858ba : SUCCESS in 46m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/470321ac45984d90b307e46b8fe6d45e : SUCCESS in 40m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/12daba414dcf4baf91d2934914588b30 : SUCCESS in 36m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dc57c7d03ae94226a40abf0dd185c1a3 : SUCCESS in 1h 22m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/487aa42c200e4652b8a73d87e4da0d45 : SUCCESS in 59m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ab707670fed540bcaf33a9d26ffaf64f : SUCCESS in 1h 08m 39s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/97441df3f9714e2f8317c6fd3edbb6a3 : SUCCESS in 55m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fe3edd8815ed4e3e9e7085531f42d3b9 : SUCCESS in 1h 37m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e9e12ac4deca4295a01527efbe8a904e : SUCCESS in 53m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e0468fe9fe674a4bacd0cf6cc7c6a223 : SUCCESS in 36m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/025be88cf4334256911ff8eb6c2836fe : SUCCESS in 34m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ccf69cfea0d5458ea2de568bb087c186 : SUCCESS in 56m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c5c82d3c443448bb31cfb871ab49291 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/358d47bb45c349a0a38f4fff6ee1c500 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b333e611e7a84c2a866c9d40b37347e1 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/354ea95491ab46c49c44517a96c74f14 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341ebebffa714342a1fbeb9a4f24f254 : SUCCESS in 6m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/38c64ce8804d4e80bc403f7f1081c39e : SUCCESS in 32m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05e2f5997e24415c877ed68bb8eb1671 : SUCCESS in 33m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/71e91e9f495f462d9614c825beeb93c8 : SUCCESS in 48m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/acd1d60ace2f4c3f82d55cf39e5fe202 : SUCCESS in 50m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5dc9b0bd1f3b4ea49eeb055632573789 : SUCCESS in 1h 16m 11s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"470f28b9c84e468a6a73964027ab8ef7107dcb52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:21:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"377568d8aad75d2caf3f34fce8f9a777d14ee3fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:00:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4a564a14a35249c5b8c1364848c2236d : SUCCESS in 56m 54s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/53e218dee8df43e7afbcdbb5f5ad61dd : SUCCESS in 35m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9b0a49a172a541a18b0c7007ed493845 : SUCCESS in 1h 27m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d647dcff570e4ac99adf4ae752f38137 : SUCCESS in 58m 13s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/843e5cf31bcd48dcbeb16883bda46daa : SUCCESS in 1h 29m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/832a9454085e4dcaafe010d0e81b623b : SUCCESS in 51m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c459dcd51b04833a6ae7eb17354c3ba : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/051aac654e824374842f5136bc3434b9 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cacd418de2d4bb58f3803b10f74ee4d : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/617044bed8cb4f779e299bb5dc703153 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e137260b494ed9bc94c02c0ec9216f : SUCCESS in 6m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d93b97e5bc0842858327a7db6c86843a : SUCCESS in 27m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0c6861c280e4c1a84c5e9ad6ab08901 : SUCCESS in 52m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/159188b3a3d74cd8b695f3c4dc810446 : SUCCESS in 1h 04m 09s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"8237e026d86bb5d29577fd30d22c88e11aabd27d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:00:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"55f054020f186f1a4d97871c160ba40d3001a66c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:01:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6efaa071a6e44af4a1e968db4cfb58ea : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56a570ddcdc287644e31c950f7591d355f61d6f4","revisions":{"56a570ddcdc287644e31c950f7591d355f61d6f4":{"kind":"REWORK","_number":1,"created":"2021-04-08 08:42:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/785330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/785330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/785330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/785330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/785330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/785330/1"}}},"commit":{"parents":[{"commit":"3251922dd517d114cf0a9e9a4a9fbac8d1bdb9fd","subject":"Bump OpenStack versions for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3251922dd517d114cf0a9e9a4a9fbac8d1bdb9fd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-07 09:55:23.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 08:42:00.000000000","tz":0},"subject":"base: configure curl before first use","message":"base: configure curl before first use\n\nWe want to have all uses of curl behave the same so we ship config file\nfor it in all images.\n\nSo do it at start of image so \u0027base\u0027 makes use of config too.\n\nChange-Id: I95aa64f0b292cd020574dbedd320ef90529cc7a4\n(cherry picked from commit 4328a5b3cc0e90e32952ecfeab2b63407ee8e22a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56a570ddcdc287644e31c950f7591d355f61d6f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56a570ddcdc287644e31c950f7591d355f61d6f4"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
