)]}'
{"id":"openstack%2Fkolla~785401","triplet_id":"openstack%2Fkolla~master~I9e8fb8fab1adba03aba4ed9ce42e4d4666bc764b","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I9e8fb8fab1adba03aba4ed9ce42e4d4666bc764b","subject":"Modify healthcheck_mariadb","status":"MERGED","created":"2021-04-08 11:54:13.000000000","updated":"2021-05-06 15:19:13.000000000","submitted":"2021-05-06 15:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785401","meta_rev_id":"07926eab7edcac44d837390dd4755d5aaea50e00","_number":785401,"virtual_id_number":785401,"owner":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 15:16:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-05-06 13:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-27 07:57:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-05-06 13:59:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 12:45:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 09:50:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-04-14 14:35:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-14 14:35:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-26 08:51:13.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"c8ee81aa65315714b43eea8e8f7d89ea7b53860b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-08 11:54:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"522fcbd3fb528b782f8472eb0a67effd14d7b4b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:45:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fcafec8097b64d338285c66f9f91918f : SUCCESS in 21m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fc097538c9b2445ba2479e7d9f58d970 : SUCCESS in 37m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5ff0c39ab274489fab555e7e91aef0ac : SUCCESS in 32m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fc0a4efabceb4103b6a6b94bda868aa2 : SUCCESS in 18m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/70673e4613164bbcb74ba98167cf7e05 : SUCCESS in 49m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e86abc569e64ff7a61aa768c28dc71d : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/192ac47a222f438980e580440cb81a37 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e95a2ce94c4477e8bbe5fdfaadc849a : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60f7b65e57fc47d6ad7819379f57474c : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898c127b29a64953853b3ad51a426e94 : SUCCESS in 6m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b39a6e4f9ba4815b84b583c3ff1cad0 : SUCCESS in 21m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2537b3075294826a0345aac902beb07 : SUCCESS in 43m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1e424aa69254f9fae2b34fcd9ec42c6 : SUCCESS in 44m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a90ffe760bb1479bbc447357dd57d36c : SUCCESS in 44m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e7f90ce3eed140a98154858f59e52af6 : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":1},{"id":"64cbe8817ebe16156f8ee59770a2bfa8adf46c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-09 07:52:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95505af072cba9efd6e73763e0de92f8e9749a53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 09:41:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a256213be3864d4a95537d795195c7da : FAILURE in 57m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0450fbb148a42bf9531c7f6ac1d78eb : SUCCESS in 1h 07m 28s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ef6e833c97f4c40aa4ebc2d0d9acd34 : SUCCESS in 1h 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46a3af454ee914e3cd5a007af5b28a23a0b807ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 09:50:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e1b8a3bc1a7a481e989a45a2e17902f2 : SUCCESS in 28m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1a50a57774c241909927f22a21a7d04d : SUCCESS in 1h 24m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a60269c095f04e88b5adf60f0fcf1bd9 : SUCCESS in 34m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/31d8f572a26c4e55b27d9b1e6f573cf9 : SUCCESS in 20m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/33f5788c844a4b8d80fb50f9eb210c0f : SUCCESS in 58m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fe91d2568da4f0580a353a8e921c76d : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d19d1bf7c859424e9221ea9849a11027 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/340c0a97c2624367906a39e67d69dc94 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cde0af6c1b214d9fa13f2869ac4cd4d9 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4a3c61b3dff49cdb5385c815c71f97b : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68d2a5dd57594fa0a68a64895c88ca0a : SUCCESS in 25m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/de6fe8fece254f24bbbcabaa2ce5058a : SUCCESS in 44m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de383e4624494bce929a15c5626c4f2b : SUCCESS in 55m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/951c9ed5dc334b11b56cce1b32e7fe56 : SUCCESS in 40m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e5437dd1b4304d468d731c5b7d1104d8 : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d478c4c7a345e5a0d4a7dcbf06faaf2d17ad4241","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-12 09:50:10.000000000","message":"Patch Set 2:\n\nNeeds updating - mariadb image is getting removed: https://review.opendev.org/c/openstack/kolla/+/785838","accounts_in_message":[],"_revision_number":2},{"id":"2046ef8123ed8c029449b1c902e5ca099916c989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-12 10:02:25.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ba62963057b9d2ee5cf8763a76af316dcc0d226f","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-12 10:03:44.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Needs updating - mariadb image is getting removed: https://review.opendev.org/c/openstack/kolla/+/785838\n\ndone","accounts_in_message":[],"_revision_number":3},{"id":"cfc87e8378c930807de756a71809de6c788e2378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 11:10:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/aa0bce44634445278c2b9078d5dfbeba : SUCCESS in 30m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4794cc0e02314d43b343a47903b7fa12 : SUCCESS in 40m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe9c9e8b37e64c5eb4d497c235d9b227 : SUCCESS in 36m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6352791e3b294de0bcc479f0ff26ddd7 : SUCCESS in 21m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/52c95c728c8a418a97eed18858d43694 : SUCCESS in 53m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c338892cfb5d4e67939dfc2ebad63199 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f771b3d7fefe42ee88be344263d8f7f3 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af76d9feedb5454fb3a5eea1f038f225 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b156e93ec394c119ee26db6fc9f2cdc : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e2983940274cbbb5e646527d10bca8 : SUCCESS in 7m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4680b26b6d0043a5aa205cb9035fabd9 : SUCCESS in 21m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/366a629e4d38413d94e4887b19e40e21 : SUCCESS in 39m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1604a97b5e504aa3969d1f1f1c6bad74 : SUCCESS in 49m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/00b5e2e3956b4544945b8f01af7e98a1 : SUCCESS in 50m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b29b62e74dbf46c49b6f6d18105643cb : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":3},{"id":"1e49f1db0bbce504c0d851e296c9da3d3891677e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 11:53:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2471636140b743a4a3d0996c0383d80c : FAILURE in 50m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f63c123292947d98bcbadf60fa352a2 : SUCCESS in 43m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cea68941aa3642a8916ddce4a1c82510 : POST_FAILURE in 1h 50m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f0ea5ecc32412cba5115fbe37f90c76352fa179","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-14 14:35:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bf3615bae06734bbfe921ca1b08ba4c15dd9773d","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-04-26 08:51:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0c6987c810f8a1f2c4450d63748f022d1e1069fc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-26 08:56:12.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bdec28c96d96d71cdde680dec3bf05d4a7863d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-26 15:08:43.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"abf5571042909ce34e809f354401e7905d100319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:31:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d1b94b07086e42eba0fd9724bdec89ed : SUCCESS in 1h 14m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1c4e66f7e5ad44caada53612f9134c45 : SUCCESS in 42m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4d550c8a894b459882e01aefc1941d13 : SUCCESS in 51m 44s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8d8f539da6dc4ff3b9df60e8dcaba5b4 : SUCCESS in 1h 19m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d2b72c4454dc4a3cb49667cfd6564652 : SUCCESS in 1h 20m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fc525bc386ce4fe4b94057d5eae74a2e : SUCCESS in 52m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/078aad734a77408badb5739dd5fbfbd8 : FAILURE in 28m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d487b70adce149cc951c1c958d3b58ae : SUCCESS in 1h 10m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee733024dbd4b4e86ed1fe6c2f0825e : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecf44112288c49939c7782cb2d5c06af : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09123d51bac949ad9b2c6eecbb2db97b : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bfc4a5dbf3f4c11b77193d5f1e0015b : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b6574be7004eb6b8c1b2c56920ee50 : SUCCESS in 6m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f030d8b3fe3a4f63a1b6fd86e2724cdf : SUCCESS in 22m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7644481d06f74ca0b414862ec88f8fc0 : SUCCESS in 33m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9671bb4eb6a74bf68233089ec3dcbbd8 : SUCCESS in 54m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a663378751c7463a852057f0abdd6e63 : SUCCESS in 56m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e538ec7eea8d48e085d62fed6017606d : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":4},{"id":"dcf904aab552a5129f20b3b956ed9fa0645f0ba3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:12:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/17a4848814a04d9785990045f148378f : SUCCESS in 2h 00m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b5c4dde574a46eebc51bb9a0f4b9137 : SUCCESS in 1h 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"213a3d18d5bc58701518ca1ff47049bea6ac99f5","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-27 01:11:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"356a2de53aaea1ab7c3dc717a8c967bda7d43791","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-27 07:57:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0b6a0af55d4956c456cd2025a442089f96a89b15","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-06 13:59:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c5e4b912458ceaf7b886269f416b02246615325e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 14:00:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"979ad92e1a8e3953b1a29ee7a1b0efd83501938e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:16:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2dc3a0333ac54f3eb4f15df9b903fd36 : SUCCESS in 40m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4c534c6162c94a4cab90b0accde4d337 : SUCCESS in 1h 04m 03s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3be9be55d68345ceb86c36aec7bdbb30 : SUCCESS in 1h 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b00780c21b34c909fdfff77de9ed82f : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09ef6cde8ff044e6bb502c46085cb092 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04d1a0ec207f4431af8b82ff95dc3801 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb2b5ee94a2429cbab2b56dae108bdc : SUCCESS in 7m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb921380bd5d4670acab7a8e51f2fccd : SUCCESS in 37m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6a96218c9b5542e585839c8720bddeba : SUCCESS in 58m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b0788e8617e04cb388d28ecc65ef5a61 : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":4},{"id":"a1835603e4ab9db59adca294c63955c05a7548a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:16:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"07926eab7edcac44d837390dd4755d5aaea50e00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:19:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb7edb7d5a3455491e324b573b7c89e : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"407d774ffb36b8acfe0898b769150c76b741b82d","revisions":{"aed29b854819d5df8608cd04297b28373a695e07":{"kind":"REWORK","_number":1,"created":"2021-04-08 11:54:13.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/01/785401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/785401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/785401/1"}}},"commit":{"parents":[{"commit":"ab8ee75bda2df6a54bab63d90dd0a6542085704f","subject":"Merge \"kolla-toolbox: Upgrade openvswitch collection to newer version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab8ee75bda2df6a54bab63d90dd0a6542085704f"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2021-04-08 11:42:46.000000000","tz":0},"committer":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2021-04-08 11:42:46.000000000","tz":0},"subject":"Modify healthcheck_mariadb","message":"Modify healthcheck_mariadb\n\nUse the healthcheck_mariadb script to check the status of mariadb\nRelated patch of kolla-ansible:\n  https://review.opendev.org/c/openstack/kolla-ansible/+/780404\n\nChange-Id: I9e8fb8fab1adba03aba4ed9ce42e4d4666bc764b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aed29b854819d5df8608cd04297b28373a695e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aed29b854819d5df8608cd04297b28373a695e07"}]},"branch":"refs/heads/master"},"3514a4d3739832e08d7789f2e81858a7b6fab012":{"kind":"REWORK","_number":2,"created":"2021-04-09 07:52:41.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/01/785401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/785401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/785401/2"}}},"commit":{"parents":[{"commit":"ab8ee75bda2df6a54bab63d90dd0a6542085704f","subject":"Merge \"kolla-toolbox: Upgrade openvswitch collection to newer version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab8ee75bda2df6a54bab63d90dd0a6542085704f"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2021-04-08 11:42:46.000000000","tz":0},"committer":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2021-04-09 07:48:18.000000000","tz":0},"subject":"Modify healthcheck_mariadb","message":"Modify healthcheck_mariadb\n\nUse the healthcheck_mariadb script to check the status of mariadb\nRelated patch of kolla-ansible:\n  https://review.opendev.org/c/openstack/kolla-ansible/+/780404\n\nChange-Id: I9e8fb8fab1adba03aba4ed9ce42e4d4666bc764b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3514a4d3739832e08d7789f2e81858a7b6fab012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3514a4d3739832e08d7789f2e81858a7b6fab012"}]},"branch":"refs/heads/master"},"3efeeee523ca108f16a8b8ab9ddbf564c34f4649":{"kind":"REWORK","_number":3,"created":"2021-04-12 10:02:25.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/01/785401/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/785401/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/785401/3"}}},"commit":{"parents":[{"commit":"ab8ee75bda2df6a54bab63d90dd0a6542085704f","subject":"Merge \"kolla-toolbox: Upgrade openvswitch collection to newer version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab8ee75bda2df6a54bab63d90dd0a6542085704f"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2021-04-08 11:42:46.000000000","tz":0},"committer":{"name":"Lin PeiWen","email":"591171850@qq.com","date":"2021-04-12 10:02:20.000000000","tz":0},"subject":"Modify healthcheck_mariadb","message":"Modify healthcheck_mariadb\n\nUse the healthcheck_mariadb script to check the status of mariadb\nRelated patch of kolla-ansible:\n  https://review.opendev.org/c/openstack/kolla-ansible/+/780404\n\nChange-Id: I9e8fb8fab1adba03aba4ed9ce42e4d4666bc764b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3efeeee523ca108f16a8b8ab9ddbf564c34f4649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3efeeee523ca108f16a8b8ab9ddbf564c34f4649"}]},"branch":"refs/heads/master"},"407d774ffb36b8acfe0898b769150c76b741b82d":{"kind":"REWORK","_number":4,"created":"2021-04-26 15:08:43.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/01/785401/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/785401/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/785401/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/785401/4"}}},"commit":{"parents":[{"commit":"ab8ee75bda2df6a54bab63d90dd0a6542085704f","subject":"Merge \"kolla-toolbox: Upgrade openvswitch collection to newer version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab8ee75bda2df6a54bab63d90dd0a6542085704f"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2021-04-08 11:42:46.000000000","tz":0},"committer":{"name":"Lin PeiWen","email":"591171850@qq.com","date":"2021-04-26 15:08:30.000000000","tz":0},"subject":"Modify healthcheck_mariadb","message":"Modify healthcheck_mariadb\n\nUse the healthcheck_mariadb script to check the status of mariadb\nRelated patch of kolla-ansible:\n  https://review.opendev.org/c/openstack/kolla-ansible/+/780404\n\nChange-Id: I9e8fb8fab1adba03aba4ed9ce42e4d4666bc764b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/407d774ffb36b8acfe0898b769150c76b741b82d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/407d774ffb36b8acfe0898b769150c76b741b82d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
