)]}'
{"id":"openstack%2Fkolla~785500","triplet_id":"openstack%2Fkolla~master~Ib32cfe4bacbfb2ccd526b62846b8d89c06506d02","project":"openstack/kolla","branch":"master","topic":"prepare-for-kolla-wallaby-release","hashtags":[],"change_id":"Ib32cfe4bacbfb2ccd526b62846b8d89c06506d02","subject":"[release] Use Wallaby sources by default","status":"MERGED","created":"2021-04-08 19:17:36.000000000","updated":"2021-04-21 18:38:55.000000000","submitted":"2021-04-20 08:39:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785500-prepare-for-kolla-wallaby-release","meta_rev_id":"539e3c0ffb64bada5e57de354d4cea25847d8645","_number":785500,"virtual_id_number":785500,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-20 08:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-20 07:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-19 20:04:37.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-04-20 07:34:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-21 18:38:55.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 20:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-08 20:08:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-16 13:05:53.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-19 19:51:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-19 19:51:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"92d44577b98c91317fe580d613c8e5a2bea1227f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-08 19:17:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e967cd45222fc43d49952ed076d16f575a5d942","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 20:06:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1e4321f8ea240f7906742451c043245 : FAILURE in 42m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d14045d3ec524330a23d9272abd7c44b : FAILURE in 40m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f3fb73dbded1442facff4b8133823c0e : FAILURE in 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7b9c008047e33dda2aec87c95e3e1520369976b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 20:08:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a3fdeded8d0f4f159b930023e7e64602 : SUCCESS in 35m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4cd7176901e54be5b79766f16b7930c1 : FAILURE in 35m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f9dd8f3aa72243c69ce9baa49d64316b : FAILURE in 30m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1a3c21d1f70e4fa5989b4668455674fd : SUCCESS in 22m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/561d6b3128ea471c8e4d58a5e53dfe83 : FAILURE in 13m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca28a76fee440b1b08500f4221c1416 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ebc925f09b54345a9437c9a9ca1403a : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/602a538e8eac418f9cbb250b7b11ef7f : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f62e7ec5fb964dfb898884d87e922c21 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/183f8f49d1ac4970811cd2cc69c96b56 : SUCCESS in 6m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/306d6bdac9714a44af8a753f12dd7c4f : SUCCESS in 21m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4b14f7d9503c46288708766e741cf593 : FAILURE in 33m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e6aa7540fcf44c7a8ac62ccef0fd2bc : FAILURE in 17m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/893d7c7e4f694fcf9cf524d733727a83 : SUCCESS in 48m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ae7be638d02444b8642ebc29c58b191 : FAILURE in 19m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7b4a64fc9ecd217b9f07361b43790d40e231f129","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-09 09:12:10.000000000","message":"Patch Set 1:\n\nblocked by vmware-nsx","accounts_in_message":[],"_revision_number":1},{"id":"237ec06e2baac73dc7f025b8f251a94c204c4cc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-09 16:04:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"906af9efab9132de4d47cc349256d1dff651e53a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:32:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4407a0da3d9b422ea4942e2a89058927 : SUCCESS in 21m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7888550431fe4b458e603160ba2ad8db : SUCCESS in 43m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/982905914fd04a0ab17d6ca2caf109e5 : SUCCESS in 35m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b3fd7538f13a4637b4a9be485ed84a88 : SUCCESS in 21m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9323d4dfcd2243879ff61e57f6ad462b : SUCCESS in 58m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66829ece7146451ea74eed122a48327e : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cebb5a2890e64d8d917360235bc0f972 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15cf0194ee8849cfbaaa2e02670be8f7 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04ffdefa1acf482d99c2c6efa1c2386e : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4302fc071d4c4dbac04aa14c67a4d3 : SUCCESS in 6m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e616f4d6c7234c479b16ba12131f098a : SUCCESS in 25m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3517006564ca472ca1dab7cf23837339 : SUCCESS in 36m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/47dbd42444c34fd8af06c67e7c11e5ae : SUCCESS in 1h 13m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0ff35a9d33894dccbf0c67857a005607 : SUCCESS in 52m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d9e4a8085fd94d11aad913df695eb343 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4d2336bac39b3b2eab1b24f937e0743fbedd0ea2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 18:20:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b76719319f6b4c248ae0855c3e2ce1f2 : FAILURE in 1h 37m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d99bd845510446dd8a7b6a4ba14d245f : SUCCESS in 1h 30m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/608c651ef998414aa9592ab9a3b3d3fc : SUCCESS in 1h 53m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"584f9c8f58e6f6c0e9dec2192902d2236290ea57","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-16 13:05:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"534c8f2aa3b9ee2474174ac7508fccd59b5c98e0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-16 13:21:58.000000000","message":"Patch Set 2: Workflow-1\n\nlet\u0027s do it on wallaby I guess (once we branch it)","accounts_in_message":[],"_revision_number":2},{"id":"a4162a08e1e5bd52a405effa906b8683fe641548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-19 19:50:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7d7a1ef639ebcffab11905cbe9272d3241e564db","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-19 19:51:45.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"938ff45938bca4714071328f0670fa0101324642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-19 19:53:19.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f9d4f89522e80c548e6b7d0552962d7e617bc515","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-19 20:04:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"05ff38ac5567bffaa16bcc8b3e6dc82e3c8a82dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 21:04:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/87f2400ef7884548b925b3cbb941c6db : SUCCESS in 27m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e55d7607d86f4eceb5643e43d246f3ed : SUCCESS in 47m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/da3a1d7ffdd14a1bb1ee14eba99d3e36 : SUCCESS in 32m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/96204dea22604b03bb8a3bdb86b533bd : SUCCESS in 22m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c439cc612b24076b320ba1dc1c82b52 : SUCCESS in 58m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/558b44518e864f59a43e565019d9762a : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e04f81236a014494abf6ef8f4e7a63ec : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b80d95ff2bb74cef8f6728acfd5494b0 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/029875832d664e2c9efdf7b992b7e07c : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b789a128dfd42af97f1ddbd1df806be : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdaea7ccfb82412bb5a62d435095c472 : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c34a989210a4db099cdeed705952af1 : SUCCESS in 26m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dab9212d64884f6792b8048ed040c55c : SUCCESS in 30m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df3ffff7a63e472a86d70272191edd99 : FAILURE in 6m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d97bba62da44cff90e24291fd6c4d18 : SUCCESS in 52m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b3be89dd1e840dda275cb45694bb171 : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":4},{"id":"bc8908bb034c13f7aa6d8d11b955cb3a6e3ba5f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 21:56:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/03392ad125e24a2485a48ac69ac58e8f : RETRY_LIMIT in 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d1285b9e69046aeb90095da2df57ebb : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b3835c7dfd0a4828bf6601feb4b78fa1 : SUCCESS in 1h 45m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"59051c47db2d86312ed11abd0c4ee0ddde33442f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-20 05:11:48.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ac11543bd4e9106f5372fbf48bfb7b048fdfc451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 06:32:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5f1e4046ba69479397043348d5c92979 : SUCCESS in 26m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/53ea36dff0de4ae3a4b99db1dcd39948 : SUCCESS in 45m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/36e3420db035456b807ca0dd24ecc67e : SUCCESS in 39m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2444e76aec1948b08f7ee02f524d3916 : SUCCESS in 23m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8f681637c534435ca636633208029cfc : SUCCESS in 1h 02m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a71b3dec0749a6aad99d53ac5faa2c : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acab1d7668e54740b01c0e1adaf3560e : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f35c21d421a411490233ab013691c58 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4378c58c426414885aea8baa7ec57a1 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aaf311367624164841c9dc8300c4a44 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/645e6adae7a04cd1ad43692091641d4c : SUCCESS in 4m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b40b31c0b7447d2862347a955c7f222 : SUCCESS in 22m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7d0f1667f9548eaaffd80b91db9205b : SUCCESS in 40m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6888af55a075449a9e2030bcca5915a3 : SUCCESS in 1h 01m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/20d86ef492a94b319e1909ea0ebd94fc : SUCCESS in 55m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a6b3e2e3427c408db714775cd3bbd241 : SUCCESS in 1h 13m 04s","accounts_in_message":[],"_revision_number":4},{"id":"37822562972b552d67954586fd679bcc5c28ec1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 07:16:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/836a13b6f0ed4eb3abb198ca060b52d1 : RETRY_LIMIT in 1m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed881e6997ee45c1ad91340945e7e145 : SUCCESS in 1h 14m 09s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8496eb80cd034dca8024184c36e946cd : FAILURE in 2h 00m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8693a3de1dd93c6f6076757ec1a575d18274c592","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-20 07:34:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"82f7df35f367ba5ae8c495bdb9e88d9379cf1738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 07:34:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f7c09ff8393bc942069f0fe81899108b943205ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 08:39:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0c127297ba3a4e0d88b9a64aa0eee72d : SUCCESS in 45m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8fdc35feabe4a27be274e88f168dcb2 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d062a7780df546d6bba30824d75099c5 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbf3d1386a874aacb405a0b543ec9b32 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2321e60858e6401683629bc6127d829d : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f87af1258d46414993726c307a7d2b97 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac5377eb3cda45d798cfc5a21b28a2c5 : SUCCESS in 37m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7fe02d2b45d94b69898a80d10fed13e4 : SUCCESS in 58m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/223607b2082043e094539b5898ae3522 : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":4},{"id":"cc21b8e912c27278f52a92180419cf142dd4cb10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 08:39:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8b61b52ee8824e8408d229c83ad2cae2ae9a8a91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 08:42:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba45eea7c86406191e05e1448715401 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/121844783d904b07bb3fc4f7a223b647 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4},{"id":"8618db9835c6eb58a597edd5f84fdf9799dc009f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-20 08:47:20.000000000","message":"Patch Set 4:\n\nI thought we agreed to add docs on how to use these?","accounts_in_message":[],"_revision_number":4},{"id":"e28f7d0ee500ad7fb9fb73f4652ec7cf7a6598f4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 10:43:24.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e I thought we agreed to add docs on how to use these?\n\nIn reno and contrib, no?","accounts_in_message":[],"_revision_number":4},{"id":"cec0fe3d10c036300e898c204912ebd1fe2ec58b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-20 12:21:37.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e \u003e Patch Set 4:\n\u003e \u003e \n\u003e \u003e I thought we agreed to add docs on how to use these?\n\u003e \n\u003e In reno and contrib, no?\n\nLink in docs to specific contrib examples was my suggestion.","accounts_in_message":[],"_revision_number":4},{"id":"3860c9c1903da4d55938925d745d61f7e33ed7b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-21 18:38:55.000000000","message":"Patch Set 4:\n\nfor posterity: followed up in https://review.opendev.org/c/openstack/kolla/+/787198","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c96a4316b324d2105326544b5026e5d825ceaf1b","revisions":{"0da3409cdf31c70b2235424ed38a7684f4aff0c6":{"kind":"REWORK","_number":1,"created":"2021-04-08 19:17:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/785500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/785500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/785500/1"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 19:16:12.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 19:16:12.000000000","tz":0},"subject":"[release] Build Wallaby release images by default","message":"[release] Build Wallaby release images by default\n\nChange-Id: Ib32cfe4bacbfb2ccd526b62846b8d89c06506d02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0da3409cdf31c70b2235424ed38a7684f4aff0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0da3409cdf31c70b2235424ed38a7684f4aff0c6"}]},"branch":"refs/heads/master"},"bb1f942760967031cccbc8f4506331028deb3689":{"kind":"REWORK","_number":2,"created":"2021-04-09 16:04:39.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/785500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/785500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/785500/2"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 19:16:12.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-09 16:03:53.000000000","tz":0},"subject":"[WIP] [release] Build Wallaby release images by default","message":"[WIP] [release] Build Wallaby release images by default\n\nWIP because I had to remove non-released neutron plugins.\nDiscussion pending on how to approach them if they don\u0027t release\nat all.\n\nChange-Id: Ib32cfe4bacbfb2ccd526b62846b8d89c06506d02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb1f942760967031cccbc8f4506331028deb3689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb1f942760967031cccbc8f4506331028deb3689"}]},"branch":"refs/heads/master"},"9c945095ce587926f89f50ffb944601f624a7b20":{"kind":"REWORK","_number":3,"created":"2021-04-19 19:50:29.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/785500/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/785500/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/785500/3"}}},"commit":{"parents":[{"commit":"d8449975646ba0a19a8f71078c74b2d4c08dffe5","subject":"Tidy up Wallaby renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8449975646ba0a19a8f71078c74b2d4c08dffe5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 19:16:12.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-19 19:50:20.000000000","tz":0},"subject":"[release] Build Wallaby release images by default","message":"[release] Build Wallaby release images by default\n\nRemoves 3 Neutron plugins from the default config.\n\nChange-Id: Ib32cfe4bacbfb2ccd526b62846b8d89c06506d02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c945095ce587926f89f50ffb944601f624a7b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c945095ce587926f89f50ffb944601f624a7b20"}]},"branch":"refs/heads/master"},"c96a4316b324d2105326544b5026e5d825ceaf1b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-04-19 19:53:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/785500/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/785500/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/785500/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/785500/4"}}},"commit":{"parents":[{"commit":"d8449975646ba0a19a8f71078c74b2d4c08dffe5","subject":"Tidy up Wallaby renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8449975646ba0a19a8f71078c74b2d4c08dffe5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-08 19:16:12.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-19 19:52:44.000000000","tz":0},"subject":"[release] Use Wallaby sources by default","message":"[release] Use Wallaby sources by default\n\nRemoves 3 Neutron plugins from the default config to build.\n\nChange-Id: Ib32cfe4bacbfb2ccd526b62846b8d89c06506d02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
